)]}'
{"id":"openstack%2Fpuppet-neutron~830044","triplet_id":"openstack%2Fpuppet-neutron~master~I7f9267eef43a08f7cca12f1062ab532d96f1075f","project":"openstack/puppet-neutron","branch":"master","topic":"user-params","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-02-22 17:15:51.000000000","reason":"Removed by Takashi Kajinami by clicking the attention icon"}},"hashtags":[],"change_id":"I7f9267eef43a08f7cca12f1062ab532d96f1075f","subject":"Avoid hard-coding OS user/group in each manifest","status":"MERGED","created":"2022-02-20 16:16:19.000000000","updated":"2022-02-23 07:58:42.000000000","submitted":"2022-02-23 07:58:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830044-user-params","meta_rev_id":"5a3dcbf385e584a663350b9f775726aa18fdc1d2","_number":830044,"virtual_id_number":830044,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-23 07:58:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-02-21 09:22:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-02-22 14:55:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-02-22 14:55:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-20 18:27:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-21 09:22:22.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-02-22 14:55:45.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"bb9976dfefc7da01fb380085adece39664c08211","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-20 16:16:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b2d4be0b2c481f03796df86dc03a9dc5e79f0907","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-20 18:27:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1851948cb3d41ca84ef0809642e4c55 : SUCCESS in 3m 18s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70610c6851e844d6b61d9247239add26 : SUCCESS in 3m 20s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6a041dec7b043c991f6e0150c7a7b98 : SUCCESS in 2m 56s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/027ae9f156554fed87266335fc4891a0 : SUCCESS in 16m 17s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/e70fc859a5a94a048667c7f0694e6fa8 : SUCCESS in 17m 39s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7124259544cf448baa4db396c51cef16 : SUCCESS in 13m 51s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/621a798671ae4b5da099551debc4d592 : SUCCESS in 17m 53s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c9653d0a3ea4ef6a577dbfc80a46dd7 : SUCCESS in 15m 03s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ea3fd46f9e84cd3ab2d77f90a4719e9 : FAILURE in 40m 44s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2289073ac354441ebfb41dbc2de54b2f : FAILURE in 35m 37s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5922ca88aacd499abe166bf79ee1114a : FAILURE in 34m 46s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d940d578ef2941b8840bc52fa99c97c9 : FAILURE in 34m 46s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8c6143f7e6574ae3bb3d772722d14b12 : SUCCESS in 21m 44s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d190f766add14afbb5e2097b2e95430c : SUCCESS in 1h 08m 40s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6adb92fe0b6f4137ac8a3e02ed61c5db : SUCCESS in 1h 33m 42s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/782a8cde052745588c0a4060e7315ac7 : SUCCESS in 1h 09m 51s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2603831f1fde4593aa988bf0b99ec2ba : SUCCESS in 1h 19m 16s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/25a3752f2191491a95be52a29f23f6f8 : SUCCESS in 17m 57s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/12227e19ac8e40328808cdface70184a : SUCCESS in 46m 29s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bbb10f4502cb4c5e9b3c2bd4a27c3c2c : SUCCESS in 1h 08m 23s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/629b3ac09b6642ed983f58e83598f429 : SUCCESS in 58m 48s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/200da0d3bfce4a1e9637159af11b8376 : SUCCESS in 46m 23s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe0df0841c5348b1a065d8d790f2ddff : SUCCESS in 19m 35s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/0f4929011f57419baec2d11e9e28f136 : SUCCESS in 24m 14s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/96f5d10712a641c0825090e8037861d4 : FAILURE in 9m 45s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/648826e9607e4277be244eff8f356d6a : SUCCESS in 19m 09s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1acac0338bc947d2a8fa1d57f5abe0e4 : SUCCESS in 17m 17s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/479d77c233a346189a31063a36427ba0 : SUCCESS in 1h 58m 07s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/fe75275e34af4375b2f5d294f02b7c7a : SUCCESS in 36m 40s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/cf8b1ce7e45544159d38e181adacb39f : SUCCESS in 56m 52s","accounts_in_message":[],"_revision_number":1},{"id":"6fcb732e6b05c0642badb5f98e878fec721901c3","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-02-21 09:22:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4efef096ae7c23323d877eb4f99d4434e461a233","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-02-22 14:55:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8107cc007e10402ef11d13f05300f239d15f483b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 14:57:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"be256892a568230f3f4e2edd577d861a3cabba8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 17:14:51.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/172afebfd1d64458b59fe669d027f578 : SUCCESS in 4m 17s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3a0625ea26c4829aa3205d485888e5e : SUCCESS in 4m 29s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a5d4cd1caa3243caaf2824c0249975f1 : SUCCESS in 4m 37s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/1517d49737c24a1284110603e8a7a30a : SUCCESS in 18m 19s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/dcce9d728d674ed8b30254fbd46b8b37 : SUCCESS in 21m 46s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92312db434ab4b4fb283ba2b78fb4ec9 : SUCCESS in 19m 21s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/375a93618a9648d9abb587086ea006cc : SUCCESS in 21m 38s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/dbac67639bb04db68e72d9b2a39a2946 : FAILURE in 1h 24m 23s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/61e2416e40b7483ca06eb3ef467248e6 : TIMED_OUT in 2h 06m 44s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8a904c5b47e44b0cb5f11b87c27e924f : SUCCESS in 1h 30m 51s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7e0f5f85b49f4246844114d13398ea7c : SUCCESS in 1h 34m 47s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/6422cbc2fce84c70a925ebb0c9e55560 : SUCCESS in 23m 36s\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/dc58889c160c4faab450141e7c1f39d5 : SUCCESS in 2h 07m 02s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/dd7f1f9991174d1384c3631ad1348aa8 : SUCCESS in 44m 25s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1cb1ad663ab14d50b019df7feaf5ae4a : SUCCESS in 1h 01m 39s","accounts_in_message":[],"_revision_number":1},{"id":"a345a4ed3f5617c179e3152e0a2f41ace3db11c2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-23 03:46:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0766a95d5afd8a94901d6644609c7523456a7a85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 05:41:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c538c18dbc154dadb853518c63c57a06 : SUCCESS in 4m 25s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/06adf2aaad334f02958a221bc45742e7 : SUCCESS in 3m 48s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e77d8e2117847ef8dbc1cb31d099e56 : SUCCESS in 3m 54s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/987897bec65b4c25a21cae0bc89afd94 : SUCCESS in 19m 34s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/c83a7fa51e21424a950eca11c46eeb7f : SUCCESS in 19m 49s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10e859e966564e789e5642d92ac00a5a : SUCCESS in 13m 05s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8d2cfb32c5443a5b51df51f11018bee : SUCCESS in 17m 21s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8abbb5008bd14d069a0bd489a6b48328 : SUCCESS in 17m 03s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a643323c9da49259496b6030c4e4337 : FAILURE in 31m 20s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a15508d4a190465cb9b3d13a5bd08ff0 : FAILURE in 32m 42s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49e7beb287a34cbea022c445b253ab9b : FAILURE in 33m 27s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/763f478eda654506bd30e69d5c8e285c : FAILURE in 29m 43s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/49832b9b3212460fb328bbbcd83e0e2a : SUCCESS in 24m 39s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/202a7798a1914e768472eee57c221530 : SUCCESS in 56m 20s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/80e787868c42401997bcc8178e21fec9 : SUCCESS in 1h 21m 07s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/73dc28cbc026472bb4707c2a89d0373d : SUCCESS in 1h 06m 47s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/707c50d728054d4e8a090181bc242a86 : SUCCESS in 56m 52s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0590615c60714c5ea4a7bc7edcb8e4c9 : SUCCESS in 18m 25s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b9ec5231ca4a4ef890411808e191cc58 : SUCCESS in 43m 21s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e4a894b2dec946a6808b4b86f2d96134 : SUCCESS in 59m 38s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bbeb0825ee3148159fb1f88f5c759411 : SUCCESS in 49m 14s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/82fb43c021414002bfae8d60eb33dca9 : SUCCESS in 48m 54s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62d2be42181047efa31098c7604a9d81 : SUCCESS in 16m 57s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/497a98bc6b134e8ebb7407ee2351b1d8 : SUCCESS in 22m 17s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/8da77d1499f54db28946d3fce390f678 : FAILURE in 8m 01s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b1a1957b8034924b60f4344502125e1 : SUCCESS in 14m 49s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44be82d8844b4de7b265ed2371950bcd : SUCCESS in 17m 46s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/b72f54d445cd4bd0a49cedd1a0873dea : SUCCESS in 1h 52m 39s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/bdc54fcdef7f4aaab4c770aa02fb30f8 : SUCCESS in 35m 25s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d836bf149b8845b688f9c57a42592df3 : SUCCESS in 1h 05m 00s","accounts_in_message":[],"_revision_number":1},{"id":"c5cfc9ea4d47feac85c5d0f5e7427e42a1a1970a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 05:43:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b7880e49f1bc04c25abd4be0a500330e375d9992","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 07:58:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/715060cc7502462398ce30bca2deb9f7 : SUCCESS in 3m 38s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0af1cb59210446228e1508c09f76c7bb : SUCCESS in 3m 37s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f37d82bd6df0477a8b3177cdf73399fc : SUCCESS in 3m 34s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/8e5b58c9a0104646a76d624eef1c7026 : SUCCESS in 19m 09s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/c77f457bd5a14f87b417f95eed08b124 : SUCCESS in 17m 17s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5e8a6fa99ae488e9d2981c903208374 : SUCCESS in 15m 44s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/952ed57efbf14349a569043f71eea3ea : SUCCESS in 17m 51s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a96c26286cb440778116b42d07955357 : SUCCESS in 54m 38s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4eebe2ad20fc42da97775c42dc48765f : SUCCESS in 1h 15m 59s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/843c987743574d1683c78cd7328a41ad : SUCCESS in 1h 01m 05s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a8f6956eb299402995b5ca7260ecbe68 : SUCCESS in 55m 13s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/e5b8325d3c744036bb080819e4026842 : SUCCESS in 21m 40s\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/825e844f321543f5b9d61ddd9a736977 : SUCCESS in 1h 42m 30s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/f8ee7bb67c00486a9ed4468e4baaa5b0 : SUCCESS in 35m 35s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c6d44ead47e74c7c9fb068fe04bf3ea7 : SUCCESS in 56m 12s","accounts_in_message":[],"_revision_number":1},{"id":"5a3dcbf385e584a663350b9f775726aa18fdc1d2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 07:58:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"46b552b1dc380513b17b67c2d9de2d4969eb0bfb","revisions":{"46b552b1dc380513b17b67c2d9de2d4969eb0bfb":{"kind":"REWORK","_number":1,"created":"2022-02-20 16:16:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/44/830044/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/44/830044/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/44/830044/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/44/830044/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/44/830044/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/44/830044/1"}}},"commit":{"parents":[{"commit":"9664dbf90759db6839fe7b90767108f03fd5cca7","subject":"Merge \"Add HashClient retry options for the dogpile.cache backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/9664dbf90759db6839fe7b90767108f03fd5cca7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-20 16:16:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-20 16:16:08.000000000","tz":540},"subject":"Avoid hard-coding OS user/group in each manifest","message":"Avoid hard-coding OS user/group in each manifest\n\nand replace hard-codes by definition in params.pp .\n\nChange-Id: I7f9267eef43a08f7cca12f1062ab532d96f1075f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/46b552b1dc380513b17b67c2d9de2d4969eb0bfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/46b552b1dc380513b17b67c2d9de2d4969eb0bfb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
