)]}'
{"id":"openstack%2Fpuppet-neutron~834680","triplet_id":"openstack%2Fpuppet-neutron~master~I8fd20c07986a4ac571b458cea1acc170af558ddc","project":"openstack/puppet-neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-03-25 19:09:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8fd20c07986a4ac571b458cea1acc170af558ddc","subject":"dhcp-agent: Support more dnsmasq_* parameters","status":"MERGED","created":"2022-03-22 12:50:04.000000000","updated":"2022-03-25 19:10:56.000000000","submitted":"2022-03-25 19:09:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"834680","meta_rev_id":"3ee0d981c50890a791d877815e04d08ca9233d25","_number":834680,"virtual_id_number":834680,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-25 19:09:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-03-23 21:18:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-03-25 16:44:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-03-25 16:44:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-22 15:57:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-23 21:18:11.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-03-25 16:44:41.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"63355cd8d0cd4a0170ce1c791a423339f3d4aaa1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-22 12:50:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab97ccd6a4335b8cef6f98f1d898b6767e372e35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-22 13:19:31.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"2635ee70e0f8a3d1ccc1298714473efbef64c4bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 15:57:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b924d4975b3140e993b45dd20e53b91b : SUCCESS in 5m 11s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7da5b95ed3b04829b9523bac11a86824 : SUCCESS in 5m 32s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b2014f2830d4cf3a804bc13d307faf7 : SUCCESS in 4m 52s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/337cbceb297e401c9af7b6ab1a382bf6 : SUCCESS in 21m 03s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/8f14da2c78044b54998770df66349517 : SUCCESS in 20m 06s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8cc92e1c27d44fdcadc7b991e69bf838 : SUCCESS in 16m 11s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0510c8123557470f96b7cbdd9c959eae : SUCCESS in 17m 22s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f360f246eee34285920d024a17105802 : SUCCESS in 16m 06s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea19a3c905b34b90ade0ea8773fe4c4f : FAILURE in 48m 59s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/247f0c627acf4f25925faf68cea360fd : FAILURE in 51m 03s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/279449dab1d1434d9156e3000e4f0e00 : FAILURE in 55m 55s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2374238e29ea457aac26fa91f06c283f : FAILURE in 41m 50s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e5512651321e4700a6972fa9dd09f2c6 : SUCCESS in 23m 51s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/28a6cffa632e4a5eb0665c81bbbfd7c1 : SUCCESS in 1h 10m 08s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b7017d6843a94486b02a168b6830a795 : SUCCESS in 1h 21m 45s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a866afb9ff324c4cb9660a4d82199207 : SUCCESS in 1h 10m 58s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/47e8766693d3470c83e19e7687318cd1 : SUCCESS in 1h 01m 12s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e73b606441e14d7686728817233479c2 : SUCCESS in 23m 54s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/578bed95d5a54bd280b0b850646340d5 : SUCCESS in 48m 19s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b34fc67614f74283bcfabe6e6b02a6e5 : SUCCESS in 1h 07m 55s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4d3ee44bedf8476e8a4b94937a88fa0c : SUCCESS in 1h 35m 26s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5f9b1ea0a5344036bf2170924a6abc40 : SUCCESS in 58m 32s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8779f12b0f9249e1bca59f0041bd0f4d : SUCCESS in 16m 04s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/055ff8b2466b466890143305da0aa3ac : SUCCESS in 27m 01s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/f36a105ae0374d378a37028b74e08688 : FAILURE in 12m 32s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1028d5efa7d24558a55695d8844ddd72 : SUCCESS in 17m 34s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3aad2aff073c4eb7b55fe032f8c3174d : SUCCESS in 24m 42s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/c237909181c44967b0f482c3be9eff63 : SUCCESS in 2h 15m 34s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/1001bfe61ad64954859c40358c464a07 : SUCCESS in 47m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e182f06cf4cb4b5ea72abe9b2d4d16f7 : SUCCESS in 5m 36s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8ff7518f94a645419d1b767ca2bd8731 : SUCCESS in 1h 04m 18s","accounts_in_message":[],"_revision_number":2},{"id":"0981f2881d91cd4e03806546ba0e701e3fe6b6a7","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-03-23 21:18:11.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4bbc08604b402b3fb5931da7657e255ff404dbeb","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-03-25 16:44:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4db9b525d80c07b353d122bb7feb02e9a9c60230","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 16:46:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4755947ef0078bce8aa38441c8b129a67bf566b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 19:09:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/901b8eb133a6463cafb1133a9de65ef4 : SUCCESS in 5m 42s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/147812a8129b4cdc9c4a8a9a6ed4040b : SUCCESS in 5m 24s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7bf4b049e9ad44f4baa5ee727436ee09 : SUCCESS in 5m 32s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/17eff6f0dfe74e79a9af74fa30a9e6ee : SUCCESS in 20m 23s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/5f4393235d7d46a3a47015274f12ddd0 : SUCCESS in 23m 28s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9bd58781c9b44496ae2eded09f19a8de : SUCCESS in 19m 33s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9668cb67b33648088433152b56b621ff : SUCCESS in 19m 28s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fe1f04188e4047f29c64bcfe26ea9c61 : SUCCESS in 1h 03m 19s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/17292916f5344d3e9955fd43697d9203 : SUCCESS in 1h 10m 54s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/878ff50a29ca445198fa3be4b845532a : SUCCESS in 1h 14m 52s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f290419d03ae43438bdcfb233d105951 : SUCCESS in 1h 05m 07s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/1dec75883f1d4355ae3c12d2909850c6 : SUCCESS in 25m 54s\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/ff84c5cb0c1741d5b8567b7ef8df32f1 : SUCCESS in 2h 05m 13s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/206c7c335e1e470789d88580a4ecf134 : SUCCESS in 44m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a787cfe3f4784988af9fe4d662868922 : SUCCESS in 5m 32s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4c120b83f7f0451d92a89026e532cccb : SUCCESS in 1h 03m 03s","accounts_in_message":[],"_revision_number":2},{"id":"3eea05844f92260d8500e3db7554d3e95dbd1354","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 19:09:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3ee0d981c50890a791d877815e04d08ca9233d25","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 19:10:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54dbb971c34f467a8349bd71d139c41f : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c081d81d7f69fe2e6e64939c2aee6580e845cf5c","revisions":{"bb2c9a6859d295390aa65f7d99bdb249853999ab":{"kind":"REWORK","_number":1,"created":"2022-03-22 12:50:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/834680/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/80/834680/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/80/834680/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/80/834680/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/80/834680/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/80/834680/1"}}},"commit":{"parents":[{"commit":"cf8278f69fc6388a9c910e4475379ae7a597e614","subject":"Fix the ignored unit test case for dhcp-agent configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/cf8278f69fc6388a9c910e4475379ae7a597e614"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-22 12:40:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-22 12:49:51.000000000","tz":540},"subject":"dhcp-agent: Support more dnsmasq_* parameters","message":"dhcp-agent: Support more dnsmasq_* parameters\n\nChange-Id: I8fd20c07986a4ac571b458cea1acc170af558ddc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/bb2c9a6859d295390aa65f7d99bdb249853999ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/bb2c9a6859d295390aa65f7d99bdb249853999ab"}]},"branch":"refs/heads/master"},"c081d81d7f69fe2e6e64939c2aee6580e845cf5c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-03-22 13:19:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/834680/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/80/834680/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/80/834680/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/80/834680/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/80/834680/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/80/834680/2"}}},"commit":{"parents":[{"commit":"3ca0ed61c465403a6e6cbcbead0bd76059ee2b1d","subject":"Fix the ignored unit test case for dhcp-agent configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3ca0ed61c465403a6e6cbcbead0bd76059ee2b1d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-22 12:40:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-22 13:19:15.000000000","tz":540},"subject":"dhcp-agent: Support more dnsmasq_* parameters","message":"dhcp-agent: Support more dnsmasq_* parameters\n\nChange-Id: I8fd20c07986a4ac571b458cea1acc170af558ddc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c081d81d7f69fe2e6e64939c2aee6580e845cf5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c081d81d7f69fe2e6e64939c2aee6580e845cf5c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
