)]}'
{"id":"openstack%2Fpuppet-neutron~835385","triplet_id":"openstack%2Fpuppet-neutron~master~I37ee7ce506539d9361febebb0a0d7ad973513c6f","project":"openstack/puppet-neutron","branch":"master","topic":"quota-cleanup","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-04-05 16:35:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I37ee7ce506539d9361febebb0a0d7ad973513c6f","subject":"Remove deprecated parameters of neutron::quota","status":"MERGED","created":"2022-03-27 15:18:42.000000000","updated":"2022-04-05 16:35:45.000000000","submitted":"2022-04-05 16:35:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"835385-quota-cleanup","meta_rev_id":"19dbf816cdbc7ede01a0076b55313dc5e704d984","_number":835385,"virtual_id_number":835385,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-05 16:35:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-03-28 20:19:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-04-05 14:00:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-04-05 14:00:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-27 17:09:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-28 20:19:11.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-04-05 14:00:24.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"9803a5f0aa9bb9e6fbd4894b1d0cb5072f0fcbbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-27 15:18:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"56f730f91cb125fbc1d8660ba7a84e3be2904f9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-27 17:09:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ad53cc232fd4f6196dcfbd4f1e3b04e : SUCCESS in 5m 57s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0bda0e87173a472ba371d3343f132a3a : SUCCESS in 4m 23s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d76409069dc4af98b889010b83ef588 : SUCCESS in 4m 57s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/a06a57b67d774feeabe475a2a33cff7b : SUCCESS in 18m 51s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/b136ee2387b847949a2aacf3c6bf6da1 : SUCCESS in 18m 28s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/579fe5f65c254a46bffcb3a075d13420 : SUCCESS in 15m 27s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99591e2a7e744dd2bbbf05b5968671ef : SUCCESS in 15m 47s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98c8394bd3014c4888bde54d10e6d9cc : SUCCESS in 14m 31s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad9d8c380e66420f83ccf4cf26af4ef3 : FAILURE in 32m 33s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b47383dfefe84addb4388e63794059c0 : FAILURE in 40m 05s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9554ba8c281475699c5916db97d11c9 : FAILURE in 40m 58s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a792b2f5e37c4500a2e8b1dde1b51ade : FAILURE in 41m 45s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b1431d6cd54547c5bc4fb44f30871d32 : SUCCESS in 24m 35s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/31f543cfdf994b218f6cc252fae9eff8 : SUCCESS in 56m 57s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/062253c8c8e54a8ca6b79b9d9e84d3c5 : FAILURE in 1h 15m 18s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9c322d84d3884f4c815ba162c7a4d65c : SUCCESS in 1h 13m 45s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3ada21795263421da10caf2f0929ffc8 : SUCCESS in 58m 14s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/856427506cb64dc09dd18afed011e460 : SUCCESS in 21m 48s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7e0306adceec433a9ea5561d6c98a407 : SUCCESS in 49m 08s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c97a24c823e84e4fa5ac01b22cd9d49d : SUCCESS in 1h 09m 19s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d67c24c55bf144bca91b8e0a8b4fc8d9 : SUCCESS in 53m 32s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/72ac216607d54bceb602b6c3874a6900 : SUCCESS in 49m 25s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8efdcc167092487290f66db90835d9bf : SUCCESS in 17m 45s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/9fc9e20f98bb44edb46393f325675cf9 : SUCCESS in 27m 19s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/69dbf70d2a6e4ccb95b914b285b578a7 : FAILURE in 9m 20s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6fb7307f9a1479b97f61748ac0634ef : SUCCESS in 14m 54s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57a5fe88bba147a5a10c769d4da47804 : SUCCESS in 18m 33s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/90f18a7086304e01971718bd5bfc0a43 : SUCCESS in 1h 48m 43s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/15691581597a45c8b64f7749a0ba410f : SUCCESS in 36m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ae8b285df3240bba8ab7c307488045d : SUCCESS in 5m 06s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a163ce6371a54af6bfd97c12ef69b211 : SUCCESS in 55m 15s","accounts_in_message":[],"_revision_number":1},{"id":"2a9136212fe017178a235b2e5f8fcc3c9e260b62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-28 00:16:20.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"7133e821aa619d6dfa7e97260100b94cbab4350d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 02:57:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31edf608062c47bfb28880472bc858eb : SUCCESS in 4m 34s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2a29b15c7904967a050306be4b26eaa : SUCCESS in 5m 47s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc68a811daf84db7a04ff7d66c9731bd : SUCCESS in 4m 29s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/f8807eb4583542b8acadbee99865e360 : SUCCESS in 19m 24s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/49768c9c1a5243fd8ac7a04139b2017a : SUCCESS in 21m 27s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/169a93a74c4d4eeb942e8783b3101a53 : SUCCESS in 18m 18s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed69b8dc25904e719ff796d3da1e62ee : SUCCESS in 18m 45s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cee6393ad6ed465ea71cfea2ef2dd6d2 : SUCCESS in 17m 23s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa730f9fafd641f889de75ae9c0a00df : FAILURE in 37m 35s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7fa6999917a4ba295114acb7b86a946 : FAILURE in 43m 32s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f34d58d47cca40219b261cc63e70f8c0 : FAILURE in 45m 08s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8632f39682dc4a52a470c52443161a05 : FAILURE in 42m 47s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4b890ca62329489d80c525625e82523f : SUCCESS in 24m 41s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7f9418b1a614438ca31c150ac8df9a9d : SUCCESS in 1h 08m 00s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4a758b5a127f4380bec0b19c32ac0454 : FAILURE in 1h 25m 33s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fe61b658f5434498aa444d43e5cf1bf9 : SUCCESS in 1h 12m 32s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0a470f546864439db7a73f50b9ce55a0 : SUCCESS in 1h 03m 49s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/79017ebff52248d2a3a2a59103c88e1f : SUCCESS in 17m 45s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/da9b7e75c37f4dc8a2220910e43c46ef : SUCCESS in 55m 28s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d232668b87af463b9b852a436578b971 : SUCCESS in 1h 01m 06s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0995818f3332427d97bb9d711ad12607 : SUCCESS in 48m 54s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bf18df880a1b4dd9b8239816b2f6aec9 : SUCCESS in 42m 36s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ec72ff048504b2a941ec115ee3dcbcb : SUCCESS in 17m 53s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/9033695679b5485f95cdea5932c77160 : SUCCESS in 26m 55s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/2cb377e2589240bd8f6211099c0b85dd : FAILURE in 11m 39s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/38693e45530b45bb97dcc5b86492b45f : SUCCESS in 17m 04s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29d71d9192494e8c8030217b4c59c4b7 : SUCCESS in 16m 55s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/16ae61de4c8c4cf297c671df50a9c55d : SUCCESS in 1h 42m 39s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/6b1b1716e709403280bd94718043241d : SUCCESS in 37m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c8e389761604bb09c3154c6b4ca6e6b : SUCCESS in 4m 41s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d0ad91a30e9a4635a658e0df37d8bef4 : SUCCESS in 55m 14s","accounts_in_message":[],"_revision_number":2},{"id":"a0ef105624e9fc501b12c053022dc88d1fc950f1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-28 15:34:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"84231f30e51c5ef218835887e6b36b4c573142d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 18:55:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/451683100a814237b71bcef5a4172f47 : SUCCESS in 5m 04s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9d8bcf66c1441029ec5bb6b17a56a8d : SUCCESS in 4m 58s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f48c8706fdf4a1ea1b968f1d35142a7 : SUCCESS in 5m 15s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d7f54f3659234e55aca4ff95f99df7ba : SUCCESS in 21m 44s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/8f8c157017474082aa22c2c17dd332e9 : SUCCESS in 20m 38s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1567bad1f3642b1b1dc88c397aa1fd3 : SUCCESS in 20m 24s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b7964e4a7ba42b7b64d5624f94d3f36 : SUCCESS in 18m 48s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c0ce4404d1b4b03ae9807705405730c : SUCCESS in 18m 34s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/621722535833409487e643eb67fe4a4d : FAILURE in 32m 01s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/133a6e36f49a4399afca2054a4ecac7a : FAILURE in 40m 21s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b1a54cc65dd4dcb8fa6f85a87a68e2f : FAILURE in 44m 25s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7a677fd8ddd4665a2042d9d7567f869 : FAILURE in 37m 22s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/440e868c742543e0822ff99e341de0b2 : SUCCESS in 25m 04s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c2b2b843edd4464ab9415b4d91a46e3c : SUCCESS in 1h 12m 58s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b0f84360bd3c47689be82fbcec52ca96 : SUCCESS in 1h 19m 11s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5320df6cec224cef9e8828a55d9d39d9 : SUCCESS in 1h 18m 05s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/25b14e1a1770454dbbea7ff01bf9f556 : SUCCESS in 59m 00s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8bbb79515335442e9ebe5d28f9981e08 : SUCCESS in 21m 10s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dcf3e37503444fa3a58ff2b5945902f4 : SUCCESS in 1h 14m 36s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/23df54c5ed2f4eb08a1c41a96a44663c : SUCCESS in 1h 12m 57s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/10363927e09245e08a68a5b891b725be : SUCCESS in 1h 16m 50s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5f6863fa980f45ae915f61e595b67ced : SUCCESS in 1h 12m 44s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d7f2cb4566494c179adb504e43fb94f0 : SUCCESS in 17m 19s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/bd31db450c504e30941f22391d9f7ba9 : SUCCESS in 25m 50s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/a26bd91b718940fe9a703509c0047baf : FAILURE in 15m 34s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21a5bd5a48e84ac1b8069763d986fbc0 : SUCCESS in 18m 43s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a8d2e24f41f46c9b606b60bf3671f04 : SUCCESS in 16m 52s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/b991c6927ccb4afb8e1b0ca0c1ddb281 : SUCCESS in 2h 49m 12s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/24a8be944afa472c97e4d536246df339 : SUCCESS in 45m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc44835626b5454c918d2ce590dd35c1 : SUCCESS in 5m 18s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8f0ee10c14f5473488ceed2050231401 : SUCCESS in 1h 14m 41s","accounts_in_message":[],"_revision_number":2},{"id":"de139b634398ad2d25a4cfbfce1186d62d453e9b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-03-28 20:19:11.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"05578d16d6659e92dcb54ec95e3884eba45783b9","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-04-05 14:00:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fa43e81925ef2f970b527bcb50be0f2e4434ab96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 14:02:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"be9f05b589cfd1876322955ec693126d06dba3e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 16:35:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d229fab96f84575803bb32dbe81e472 : SUCCESS in 4m 01s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87583be352df4e1aa52824b126e3ab36 : SUCCESS in 4m 02s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/504e5443ba154f50a14dc325e3c5e784 : SUCCESS in 4m 28s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/84237d41540e430ea711ed05c648dea2 : SUCCESS in 20m 47s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/c53087b311a1432e966f62424e9205c5 : SUCCESS in 24m 35s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf5b5178771b44a7bdf62f015ace3382 : SUCCESS in 17m 57s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3088bc1dfc24207863f9e872c88149e : SUCCESS in 16m 51s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ced5f4f75e36413497fd7184ac0178fb : SUCCESS in 1h 00m 49s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/16dc9a669e704b92945aa65881347c09 : SUCCESS in 1h 17m 12s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/07fe99d980a54b02a116869b72dfff1a : SUCCESS in 1h 09m 33s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/edb385ab7cf44ea1948901e08e90bf5e : SUCCESS in 1h 03m 08s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/96f782aa0ff14dccb0af6f3839b0ce1d : SUCCESS in 25m 10s\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/aa11a7686de345a5b8e00662222b381c : SUCCESS in 2h 10m 26s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/9bf7e6689b8b4a3588d3fc29ba56c224 : SUCCESS in 42m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01cfbf2c3b244ea98aa03b669da6c36c : SUCCESS in 4m 34s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8e6e3b2f002847e38665292f4955d1cf : SUCCESS in 1h 05m 45s","accounts_in_message":[],"_revision_number":2},{"id":"19dbf816cdbc7ede01a0076b55313dc5e704d984","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 16:35:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d68de64dd72c4882a8efc685c02a9cd1c3188d1b","revisions":{"b27663c9dfcfd0db2cb948c10a7df0984d16f721":{"kind":"REWORK","_number":1,"created":"2022-03-27 15:18:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/85/835385/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/85/835385/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/85/835385/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/85/835385/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/85/835385/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/85/835385/1"}}},"commit":{"parents":[{"commit":"989282da3b079569faced0d4dce0854edccbb162","subject":"Prepare Yoga RC1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/989282da3b079569faced0d4dce0854edccbb162"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-27 15:18:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-27 15:18:02.000000000","tz":540},"subject":"Remove deprecated parameters of neutron::quota","message":"Remove deprecated parameters of neutron::quota\n\n... because these three parameters were deprecated during the Xena\ncycle.\n\nChange-Id: I37ee7ce506539d9361febebb0a0d7ad973513c6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b27663c9dfcfd0db2cb948c10a7df0984d16f721"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b27663c9dfcfd0db2cb948c10a7df0984d16f721"}]},"branch":"refs/heads/master"},"d68de64dd72c4882a8efc685c02a9cd1c3188d1b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-03-28 00:16:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/85/835385/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/85/835385/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/85/835385/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/85/835385/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/85/835385/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/85/835385/2"}}},"commit":{"parents":[{"commit":"c081d81d7f69fe2e6e64939c2aee6580e845cf5c","subject":"dhcp-agent: Support more dnsmasq_* parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c081d81d7f69fe2e6e64939c2aee6580e845cf5c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-27 15:18:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-28 00:16:20.000000000","tz":0},"subject":"Remove deprecated parameters of neutron::quota","message":"Remove deprecated parameters of neutron::quota\n\n... because these three parameters were deprecated during the Xena\ncycle.\n\nChange-Id: I37ee7ce506539d9361febebb0a0d7ad973513c6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/d68de64dd72c4882a8efc685c02a9cd1c3188d1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/d68de64dd72c4882a8efc685c02a9cd1c3188d1b"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
