)]}'
{"id":"openstack%2Fpuppet-neutron~837217","triplet_id":"openstack%2Fpuppet-neutron~master~I15b0add80546a3110c9a8fe0cdfd5831b91ff186","project":"openstack/puppet-neutron","branch":"master","topic":"deprecate-vpp","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-04-10 14:56:46.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I15b0add80546a3110c9a8fe0cdfd5831b91ff186","subject":"Deprecate support for networking-vpp","status":"ABANDONED","created":"2022-04-10 11:24:15.000000000","updated":"2022-04-10 14:56:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b559d7ca5c3a81ec1255f66c8602cc1eb95a8aa3","_number":837217,"virtual_id_number":837217,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-04-10 13:17:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-10 13:17:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2c1d8f1c21d9a7065ec39cadca62ee139386fef4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-04-10 11:24:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f8767dc181f80650c14836c2e749d0789623ec4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-10 13:17:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0fe09d197a0c44429db86c3a3074f85d : SUCCESS in 2m 58s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eeec29fc6a0042b9b6ee884147953df8 : SUCCESS in 2m 50s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f170ff5350f24509bb99e3bb9c20555c : SUCCESS in 3m 01s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/df685d668e644114b160092157c09d0a : SUCCESS in 16m 16s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/c39a45e073544e27a774b3cae2ae7f88 : SUCCESS in 16m 34s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7fece54695fe436bb7759c099e5719e6 : SUCCESS in 15m 38s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f05fe1067c7041a086b8744eede0af95 : SUCCESS in 14m 50s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66eb89ec69994ab989ad5277d82c0373 : SUCCESS in 14m 14s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d96bb41f8e4f4dd18e14cdf10464dcb6 : FAILURE in 27m 13s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02537002b3f24de9bdae823ae1c77d3d : FAILURE in 37m 59s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f27063ed8fb4dfe8462e078437b78c7 : FAILURE in 39m 26s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b06739777664c24bf02fe7c86de42a9 : FAILURE in 30m 22s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3a1ab33d86334a2898df6bc040c74817 : SUCCESS in 21m 40s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b6e457be56ab4d66acb281a2157a5d6a : SUCCESS in 55m 59s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0b410ad2dea7463e8aa184082f89b554 : SUCCESS in 59m 10s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/04c1109622eb4f8e9df8fb1255ec1b94 : SUCCESS in 59m 03s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d6650c02e05d49a990ff71f8003c6593 : FAILURE in 56m 08s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e4005b3b668448d8979f2947c741c72c : SUCCESS in 18m 24s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3f5ebe314be84d878aea9606070ecfea : SUCCESS in 48m 09s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ee0da1e48d99481096c76f38ab9f4a17 : SUCCESS in 1h 00m 31s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9730f1e1bc5a421ba2e0bbb8420dc038 : SUCCESS in 57m 08s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b6f16324f2db40fa831fda7638f5415e : SUCCESS in 52m 23s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/83e08a66a61f4bde89b317babe6076ed : SUCCESS in 14m 49s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/a7975a3af09f41bc894849e45c4b5e0d : SUCCESS in 21m 10s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/0dbe5344fa1448c4ba0e3f30bd461be8 : FAILURE in 7m 58s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af2aafe6fc474871b5183270c06fcae4 : SUCCESS in 15m 35s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/30314de5991b4eb0bdba3b5616250960 : SUCCESS in 14m 28s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/65137abe6c6e44fdb1da862f101a327e : SUCCESS in 1h 48m 31s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/a32b1140d0a14d54a91ebcb89d4d3f79 : SUCCESS in 35m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c01d28227b74546bb9fcde8c82fcee5 : SUCCESS in 3m 15s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1bc2217599bb49e9b0e92614d7caf9f4 : SUCCESS in 55m 23s","accounts_in_message":[],"_revision_number":1},{"id":"b559d7ca5c3a81ec1255f66c8602cc1eb95a8aa3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-04-10 14:56:46.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9277d588ab2110b0dffa08afc69a4ebad444372c","revisions":{"9277d588ab2110b0dffa08afc69a4ebad444372c":{"kind":"REWORK","_number":1,"created":"2022-04-10 11:24:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/17/837217/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/17/837217/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/17/837217/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/17/837217/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/17/837217/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/17/837217/1"}}},"commit":{"parents":[{"commit":"38f1a2e2507dc2ccd90a642e38d37bb358a390b2","subject":"Merge \"Do not start/restart neutron-destroy-patch-ports.service\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/38f1a2e2507dc2ccd90a642e38d37bb358a390b2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-10 11:20:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-10 11:20:56.000000000","tz":540},"subject":"Deprecate support for networking-vpp","message":"Deprecate support for networking-vpp\n\nThe networking-vpp plugin[1] has not been updated since end of 2020.\nThe plugin doesn\u0027t support Python 3.8 (default in Ubuntu 20.04) and 3.9\n(default in CentOS 9).\n\n[1] https://opendev.org/x/networking-vpp\n\nChange-Id: I15b0add80546a3110c9a8fe0cdfd5831b91ff186\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/9277d588ab2110b0dffa08afc69a4ebad444372c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/9277d588ab2110b0dffa08afc69a4ebad444372c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
