)]}'
{"id":"openstack%2Fpuppet-neutron~840465","triplet_id":"openstack%2Fpuppet-neutron~master~Iaadc6313d50d283dece737290604b2253578dd3d","project":"openstack/puppet-neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-05-11 12:40:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iaadc6313d50d283dece737290604b2253578dd3d","subject":"Remove clean up of [DEFAULT] ovs_integration_bridge","status":"MERGED","created":"2022-05-04 13:39:53.000000000","updated":"2022-05-11 12:40:49.000000000","submitted":"2022-05-11 12:40:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"840465","meta_rev_id":"19246e6d3513a09ef7fb6744a26f352ed1d8ea2b","_number":840465,"virtual_id_number":840465,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-11 12:40:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-05-06 20:54:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-05-11 09:09:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-05-11 09:09:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-04 15:41:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-06 20:54:56.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-05-11 09:09:08.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"43a953062fb460990502e294b37d7c83d463880c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-04 13:39:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6484da426e669d610c007a46d91e67b8c40c7607","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 15:41:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/225f6814c88c4bca9cc81768d39f3304 : SUCCESS in 2m 59s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45c2b50b676142fa8d770abf4e0ca4b5 : SUCCESS in 2m 53s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8db8716ab54e48178631e77b71f95484 : SUCCESS in 2m 48s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/10fef80ed0cc4bd7b6e64e71538f5d1e : SUCCESS in 19m 27s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6864ce699ade40d586d26cf0816a5a7c : SUCCESS in 15m 49s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94e4295baa5d48f8bfa2b08543cf9552 : SUCCESS in 15m 43s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d88c5c2faed40d0b4d10f8d2e085fa3 : SUCCESS in 14m 52s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10ddc08a5551469eab3e941ec2b1b42b : FAILURE in 35m 57s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/517ba12f587848abb6da7a09fd099484 : FAILURE in 36m 51s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8afa09bc6364ca4b38af15efd890ba3 : FAILURE in 32m 50s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8f21e5a44194db4ae833491eafad895 : FAILURE in 34m 22s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/afa5c6afa6c94d75a94698f8a8240832 : FAILURE in 12m 56s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/936d520fe99648e696c5763f22cc47d8 : FAILURE in 10m 32s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3e4d6aa7570542b1b9d21f31e52a88e8 : FAILURE in 10m 59s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b7c73fc45a76403ab3b1c0100307f863 : FAILURE in 10m 30s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc705620dd7e40cb9afec025dff459cf : FAILURE in 11m 35s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7be81b20296c4d26b76c03a8b3046a30 : SUCCESS in 17m 44s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/dd6b8dac043142399374f68908af5ca9 : SUCCESS in 21m 34s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/dd28741d074e496c86b8db1f156e8090 : FAILURE in 7m 43s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1706cc6bd4e4b8f95de1531e0a3d93e : SUCCESS in 14m 32s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9240664ad9894c29a1df614195da54ac : SUCCESS in 13m 10s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/02da4079c748469c87280743f20baf50 : SUCCESS in 1h 56m 50s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/0c099275412047039a17fe326eada9a8 : SUCCESS in 43m 22s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/174211c373c84202bd499023fa348895 : SUCCESS in 1h 01m 56s","accounts_in_message":[],"_revision_number":1},{"id":"721b5743ccaf8496118b23c7ed37fd7427b940a0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-06 02:24:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"94117de61bdf327adecc66579df018f75fd061b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 04:47:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d7677e9b92da4d11a21828c822256e35 : SUCCESS in 6m 56s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bac5650595fb4b90b244867de553fd26 : SUCCESS in 3m 53s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49ea5dfae18b4fc6aeaf669a255ce8de : SUCCESS in 5m 42s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/6409e0feeb79436da842d443b16db5a4 : SUCCESS in 20m 09s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b401266e6844f5f971f582b10cf9537 : SUCCESS in 18m 11s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/374fbdb354f74c57ab38b38d860063f8 : SUCCESS in 20m 53s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b056a734b24143fbba2050ed58b9b918 : SUCCESS in 18m 14s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab1936af9a2442b5bb0f73dc4e166cb7 : FAILURE in 33m 40s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3cdeb97603cb4ca0832d4c8c2a46a6e1 : FAILURE in 41m 26s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0aea5ce284c9484399937de4f9df2a91 : FAILURE in 43m 10s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2713cbc0c0d142a8afd86431bde7bd59 : FAILURE in 42m 10s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c770efeb5fbd4ebea4b336a093ab76ec : SUCCESS in 21m 33s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d84884ffd4944e148a458e124f492469 : SUCCESS in 54m 00s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/394666e78cb34e97a5d8ae031d37ae59 : SUCCESS in 1h 08m 44s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/66ba52ca23dd42e49ee81979afc535f6 : SUCCESS in 1h 02m 25s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/01c18daceeaa407db0b6c30dd659fb85 : SUCCESS in 54m 44s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/196360be2d7b42a0a73349295e1b8f01 : SUCCESS in 19m 02s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/33d9325201704799b51639d9bf58bfc1 : SUCCESS in 26m 48s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/2059cd5186804f03829e74f7f8fa999e : FAILURE in 9m 34s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b9b5f428f9646aa8297d69169feab2e : SUCCESS in 17m 21s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e350c6878da64629b4215797ae780042 : SUCCESS in 19m 50s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/cafc46075b3043caafb28c2b89c27477 : SUCCESS in 2h 06m 42s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/4e78b1ce4bb3484d8f564390b58875f9 : SUCCESS in 41m 12s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9fb0a125267c4cdabc841d15c8d1f6f5 : SUCCESS in 1h 04m 29s","accounts_in_message":[],"_revision_number":1},{"id":"e4cdd44e2182fd43f546feecb990cafa6ac8a3ac","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-05-06 20:54:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"708fbc6cc705a0853b376f3b67b1c0ad2160f018","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-05-11 09:09:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"12593d15c2b85fc0883449c147f49ba80d15b5fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 09:11:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"02942bce70b63d97ff7b27f3fc99ab4c8c0ad329","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 12:40:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc03ef2bda6b4ec7bc8eb0daad6370c7 : SUCCESS in 3m 19s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e90de836cb1542fd9baea5917f9b1bb5 : SUCCESS in 3m 18s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e3f7260cb9148b4937ba84e297be391 : SUCCESS in 4m 00s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/898d2761d3fe4a27ba65ca59b43ed11e : SUCCESS in 19m 55s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f2d96565c33416fb6593cadbc1ed19f : SUCCESS in 18m 04s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6d6fc5ac06c42589555b3fdb51ab31f : SUCCESS in 18m 25s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/201d5f5a73094e64bbc58b9705193104 : SUCCESS in 51m 46s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/017e0e9e7f684c31a908ef55dcae7033 : SUCCESS in 1h 07m 55s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e54a2726473b49f89e9c243c59903519 : SUCCESS in 1h 04m 00s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/49c6fd1fb57a4c3990cff35a21c55a11 : SUCCESS in 54m 59s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/ffe920ef13e942f8aaaca8d5a226ba7f : SUCCESS in 25m 57s\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/6795c3d00ab049618b55146cd2ae042d : SUCCESS in 2h 36m 07s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/99aa1e0b7e5343489d315b3002cd7e17 : SUCCESS in 59m 13s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/43ffe3a8c9d64b3aaccead3940fcbeb0 : SUCCESS in 1h 23m 58s","accounts_in_message":[],"_revision_number":1},{"id":"19246e6d3513a09ef7fb6744a26f352ed1d8ea2b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 12:40:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"612d367242ba619ff5320734bae5066acafba699","revisions":{"612d367242ba619ff5320734bae5066acafba699":{"kind":"REWORK","_number":1,"created":"2022-05-04 13:39:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/65/840465/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/65/840465/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/840465/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/840465/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/840465/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/65/840465/1"}}},"commit":{"parents":[{"commit":"df0d55cf784bd50b9f858a28ba35134197c21305","subject":"Merge \"OVN: Clean up deprecated parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/df0d55cf784bd50b9f858a28ba35134197c21305"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-04 13:36:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-04 13:38:20.000000000","tz":540},"subject":"Remove clean up of [DEFAULT] ovs_integration_bridge","message":"Remove clean up of [DEFAULT] ovs_integration_bridge\n\nThis logic was added during Victoria cycle[1], and has been kept for\n4 cycles. So we can safely assume the old parmaeter is removed during\nupgrade.\n\n[1] 802da636e95a44663c272834cb43e4a4faf3c42d\n\nChange-Id: Iaadc6313d50d283dece737290604b2253578dd3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/612d367242ba619ff5320734bae5066acafba699"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/612d367242ba619ff5320734bae5066acafba699"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
