)]}'
{"id":"openstack%2Fpuppet-neutron~840926","triplet_id":"openstack%2Fpuppet-neutron~master~I4b3a6ecf627c6d7203d51f13d69a9400085f8df6","project":"openstack/puppet-neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-05-11 15:12:38.000000000","reason":"Takashi Kajinami replied on the change"}},"hashtags":[],"change_id":"I4b3a6ecf627c6d7203d51f13d69a9400085f8df6","subject":"Deprecate the mech_driver defined resource type","status":"MERGED","created":"2022-05-06 15:37:22.000000000","updated":"2022-05-11 19:38:13.000000000","submitted":"2022-05-11 19:37:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"840926","meta_rev_id":"26d44a3549e0676602f8ffa35ebdc465831335f7","_number":840926,"virtual_id_number":840926,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-11 19:37:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-05-06 20:55:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-05-11 09:07:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-05-11 09:07:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-06 17:45:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-06 20:55:19.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-05-11 09:07:45.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"a460d05c0a6615c897a59332d7f0c420292ca092","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-06 15:37:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a13fe72d87827399ef5ae4c94449d03be8ba5b2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 17:45:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e82edb7225cb401eadb047d24a958bcc : SUCCESS in 3m 27s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fadc4d2e43054948acac45969a8ed449 : SUCCESS in 4m 00s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d59750d2ec94eb085778b39f5e1e2eb : SUCCESS in 3m 04s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/1557627e9a734da0868aba5bb804bc6f : SUCCESS in 21m 00s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/294207841e814adf814ddac428be8e60 : SUCCESS in 14m 23s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a1febba010046029923a3083a87bb13 : SUCCESS in 16m 47s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b22d0869ce946dcba034a7c6fae967e : SUCCESS in 15m 54s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ca1a9a2a39941839107930aeb903546 : FAILURE in 30m 53s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fcc3a6887fad45008e62b499e21e0229 : FAILURE in 33m 58s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b69055aaedc4da38f9e6ef53dac7b9e : FAILURE in 44m 21s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d45bd3211760497784e262aa5940bc40 : FAILURE in 39m 44s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/82c221a9fbba45a89d1fb22a84640325 : SUCCESS in 20m 40s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1b446795a628439da96079aa26a3d3f8 : SUCCESS in 58m 01s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a6415773664d4189b20fd6ccd0728e76 : SUCCESS in 1h 03m 20s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d98297de72d3490d9475f11b38fad0af : SUCCESS in 58m 30s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/44514b5467874b3fb5e7c3162692bce2 : SUCCESS in 50m 32s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b14a8aaca694800880637afdbfeab4b : SUCCESS in 17m 24s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/41be80310b0148da84cfbad60fade500 : SUCCESS in 21m 21s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/1e5b9450e547458585639afeb4d1822f : FAILURE in 10m 03s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8dbad051634e421b89950fea9ba08d8c : SUCCESS in 17m 03s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5645f18f67354a5a8094615d6222b434 : SUCCESS in 17m 59s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/8e7df92d0c664e34b9857cb043e41177 : SUCCESS in 2h 04m 10s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/3db517f7897f4a89bc918e9936f4c05e : SUCCESS in 42m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ed82957561c46aa958adf93b9fd3d49 : SUCCESS in 3m 25s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b750da4c357644448d72e8e1cbb475e0 : SUCCESS in 1h 04m 18s","accounts_in_message":[],"_revision_number":1},{"id":"82f6843ead95e168dcdf19cfc559419dddf83c00","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-05-06 20:55:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"02a12f4f8e8999487859afc66876ebc12754bb4b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-05-11 09:07:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ba96402a2d3649a1a127636d473bbb2cbe52dcc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 09:09:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"57f27b92c260488d40a1d49acff09a77221b253f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 11:44:23.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c7f820e90cd43598f235b7ad3d1d134 : SUCCESS in 3m 12s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/153b26dcb9a14770a6787ea6f29bc3c7 : SUCCESS in 3m 33s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b107bb7569a4742987ef5027117bf40 : SUCCESS in 3m 23s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/d1b6112d5c9c496296a9d2c2751c2882 : SUCCESS in 16m 35s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5735f7c149945f48841fdbe0a468faa : SUCCESS in 17m 11s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc93fcd5dd0245b491ec70ad900a3f83 : SUCCESS in 17m 44s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1eb6209a72c740298cfe0d7f6dd3c7c8 : SUCCESS in 54m 56s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3dfc051799194cbc8f691bec74c2e332 : SUCCESS in 1h 09m 01s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7602f1bdfc3d41228ee76e613d63932b : SUCCESS in 1h 01m 08s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ae7a48a4fa6048f390bd1c2c993de13f : FAILURE in 29m 31s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/7de0db6fdf47433d853b43e775843e23 : SUCCESS in 24m 04s\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/c832f887e8554169bc5df52e9ced5a0e : SUCCESS in 2h 06m 20s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/f1ccfe8eb3724ebdb6303905aef85a47 : SUCCESS in 41m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f36a6ea748334d19a5176f09ee183cee : SUCCESS in 3m 34s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a54ce2aab3a646fdb0df9b3f9ccf6b33 : SUCCESS in 1h 04m 28s","accounts_in_message":[],"_revision_number":1},{"id":"d7d38e051af748d5c9afcd4fbe2fe301a90efc16","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-11 15:12:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5dbfae71e07e205a70d13297b5e70ce8b5aca4db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 17:28:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/173d3118135c49a29fdb893622403958 : SUCCESS in 3m 38s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85ece53a78cb43d5a81f89ecd90c8d27 : SUCCESS in 3m 53s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0eb434b147634197808cf49c502b9097 : SUCCESS in 3m 48s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/8e79c05b579b427abbbf8e7b217d3fb7 : SUCCESS in 20m 45s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5d2f35c8b8f47ff8f3da219807e7a1d : SUCCESS in 14m 59s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5f63c37d0e24ab484a14a2d0f6a63f4 : SUCCESS in 15m 15s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bac57deb5ff24d10bbb3d4bc6fde7505 : SUCCESS in 17m 30s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5394c7b59aa4bba8d7aa0e056b2815c : FAILURE in 25m 32s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b169ea6d42e42b78dbe10648be1dd7b : FAILURE in 46m 42s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fbec81ad15374ea5935baa9151df0252 : FAILURE in 36m 03s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73b3c5d66434488388b373e28b04abef : FAILURE in 28m 12s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/239de1d3dff64c0c928f611b0f0187c9 : SUCCESS in 20m 08s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d312ce6d5ba446c4b6c80c931d66fc38 : SUCCESS in 1h 02m 20s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/670c61f66c794ed9b8d53eca8337e8cd : SUCCESS in 1h 08m 41s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/952b3dc3c1b64968a775afe4de19d3f1 : SUCCESS in 1h 00m 33s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/659f7f097e0a47dca91a5e1021b7a907 : SUCCESS in 52m 17s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9e5908cf0a7465189846652bfe918a1 : SUCCESS in 18m 00s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/bcd1c7005e0242d49da3c7fd155e1091 : SUCCESS in 24m 03s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/942bfe1dfe9543bd92c6d68c1b87619a : FAILURE in 11m 10s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1904d435f624f7f8f9b0b65792b84da : SUCCESS in 20m 13s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8bed91e12ff542969c029d038c9c81c7 : SUCCESS in 19m 01s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/44995b9a61fd4f56ba532a65f682194e : SUCCESS in 2h 12m 45s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/cdcd53f1f83941f2841ef78295738210 : SUCCESS in 43m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a5994ea44a6472b8cb3c3fd21d6a613 : SUCCESS in 4m 33s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/70275297f3214e33b85b4f5f22139d0c : SUCCESS in 1h 06m 36s","accounts_in_message":[],"_revision_number":1},{"id":"8a97ef0128012851ebb432a6220847bccc76d200","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 17:30:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"743e9ee1cb8c7cec9a3d35c4605d105a920f7dba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 19:37:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d2b86d29924448a80545bc115b39cdd : SUCCESS in 3m 19s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/afa3ec3944284570a6e4ff8e5a87ca39 : SUCCESS in 3m 51s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78f297ff9c884e68ad4cde5b982d6d23 : SUCCESS in 3m 42s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/127eae10736f435e982a21a72f5a9cc9 : SUCCESS in 22m 40s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/19435ce52da147c9ac2985a3bb3b656e : SUCCESS in 15m 28s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c92b6c86f644be49e3524846c88e416 : SUCCESS in 17m 17s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7c3e46105b55458a9ffdbebcd3a5f799 : SUCCESS in 54m 01s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4446580c7e7b4c2c93aa93862f9be5fb : SUCCESS in 1h 06m 39s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/eb7f47759aea4a139f0a32086e0a9fe5 : SUCCESS in 1h 02m 53s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/60e3b56168684bc38918a204a8c42ec2 : SUCCESS in 50m 56s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/8c9342893e0e4c38bc4c01ad8ab2c282 : SUCCESS in 25m 43s\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/7fc4b096c54045c88ab91ec122a13bd7 : SUCCESS in 2h 02m 30s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/0e21df412cbe47d6bee4e142ab77b29b : SUCCESS in 38m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc39967a7e0a427a827ad9b06a7c4e43 : SUCCESS in 4m 01s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2fd47c29407345c681f2e26c2ef9ce38 : SUCCESS in 1h 01m 44s","accounts_in_message":[],"_revision_number":1},{"id":"d702b824822cc2f7e80a650ef84811ed3666522e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 19:37:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"26d44a3549e0676602f8ffa35ebdc465831335f7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 19:38:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f13cff521774c2db9a11131ee931d90 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"069b2226a9d521865725d84fe950558cb7c06036","revisions":{"069b2226a9d521865725d84fe950558cb7c06036":{"kind":"REWORK","_number":1,"created":"2022-05-06 15:37:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/840926/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/26/840926/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/26/840926/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/26/840926/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/26/840926/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/26/840926/1"}}},"commit":{"parents":[{"commit":"df0d55cf784bd50b9f858a28ba35134197c21305","subject":"Merge \"OVN: Clean up deprecated parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/df0d55cf784bd50b9f858a28ba35134197c21305"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-06 15:28:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-06 15:37:11.000000000","tz":540},"subject":"Deprecate the mech_driver defined resource type","message":"Deprecate the mech_driver defined resource type\n\nThis defined type manages two configurations for SRIOV but both of\nthese are required for ancient releases of Neutron and are invalid in\nrecent releases.\n\n1) The ml2_conf_sriov.ini file was removed during Ocata cycle by [1],\n   and now no additional file is required by neutron-server.\n\n2) The supported_pci_vendor_devs was deprecated during Newton cycle[2]\n   and was removed during Ocata cycle[1].\n\n[1] https://review.opendev.org/c/openstack/neutron/+/413439\n[2] https://review.opendev.org/c/openstack/neutron/+/391911\n\nThis deprecates the mech_driver defined resource type because now it\nhas no actual logics. We\u0027ve implemented separate classes for individual\nmechanism driver and this common implementation is no longer required.\n\nChange-Id: I4b3a6ecf627c6d7203d51f13d69a9400085f8df6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/069b2226a9d521865725d84fe950558cb7c06036"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/069b2226a9d521865725d84fe950558cb7c06036"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
