)]}'
{"id":"openstack%2Fpuppet-neutron~841005","triplet_id":"openstack%2Fpuppet-neutron~master~I7f95546d447e057de09a2e444a2957c805a47edb","project":"openstack/puppet-neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-05-11 11:44:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7f95546d447e057de09a2e444a2957c805a47edb","subject":"Remove allow_overlapping_ips","status":"MERGED","created":"2022-05-07 06:32:49.000000000","updated":"2022-05-11 11:45:27.000000000","submitted":"2022-05-11 11:44:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"841005","meta_rev_id":"233d0a19b857b4d1850b4e03df3d597c3f80bb1e","_number":841005,"virtual_id_number":841005,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-11 11:44:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-05-09 22:05:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-05-11 09:03:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-05-11 09:03:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-07 08:49:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-09 22:05:06.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-05-11 09:03:17.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"12a5c39e6a17bf42157aff7e91ebd9ba6e459bf1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-07 06:32:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6442104804c6fdc2f0984ca7673342011dcd04e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-07 08:49:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2cae34a48bd9489bbac05bfe0a01565a : SUCCESS in 3m 39s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc39d637b6104b8eae0555f6cc94032f : SUCCESS in 2m 40s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21bf5bae1aef48c6bc4a8d44c4337384 : SUCCESS in 2m 31s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/bcbef22a2137413694d0cb8317881c5e : SUCCESS in 16m 41s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7eb78c631bd4a7bbf15c60fecaa47b9 : SUCCESS in 12m 52s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1a89d3d0d224fbd89a0f455eb3f256d : SUCCESS in 13m 17s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3bfd5beaedd4bf6a61dfbc8d0140195 : SUCCESS in 11m 38s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/400f16493d40464ab5017cb0ff65b768 : FAILURE in 33m 48s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/765f463baaa542bd96c4ff242f1e9bbc : FAILURE in 39m 21s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6710f6cd233647a8932049a5c26efcd5 : FAILURE in 36m 07s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed8c2dd847814c4a829a16f14be358c7 : FAILURE in 38m 19s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/435ae7e307d54d51bbd53907eb25a258 : SUCCESS in 24m 28s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ca88e3057a4d4e4d821b5cce507a3efa : SUCCESS in 54m 23s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/60b0508a3eba4a629ddcc1f3f6805322 : SUCCESS in 1h 03m 32s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/177fe14b991148cdbd438b16c53eeb23 : SUCCESS in 55m 10s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8ce64c274a6f4bd9ac1ebab8d80c1021 : SUCCESS in 51m 03s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eacfba02f133469da717bc566f36bdff : SUCCESS in 15m 31s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/651fa307151c4252957d3387d722f9d4 : SUCCESS in 22m 22s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/23dcfb43f53d45a0ae359cd0c70cf49f : FAILURE in 10m 14s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7620a21b0dfe4b408d81268b1da7329d : SUCCESS in 15m 46s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b1a896682e34de4ab0eec219582aa74 : SUCCESS in 17m 53s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/7f3a4ffe172a47798380bef0afda514d : SUCCESS in 2h 10m 51s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/0514fd7653f34eac86df0266d005dcaa : SUCCESS in 53m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bab000ee625f406e917c36d6929f4cb4 : SUCCESS in 3m 16s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/30b418d6ee54446ea81595ba2e4c5081 : FAILURE in 51m 24s","accounts_in_message":[],"_revision_number":1},{"id":"1a2317abc56085cabbd865d6dced76eb11e7e4fb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-08 23:07:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"887c633c36a0f96ad2995542c5ce44f3a87a8823","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 01:02:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11051ba246014840834d88f2882aa4b9 : SUCCESS in 2m 58s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/97a5e9e64dfe4b06b78dc4394971dd19 : SUCCESS in 2m 52s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91fce0def1d740a2908a3ce3287f20fd : SUCCESS in 2m 55s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/cd6438df78ab466892ed47739b5a11e5 : SUCCESS in 15m 22s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f8f60b35b8247bcbea7c1c05b515ac6 : SUCCESS in 12m 53s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cbcc45d9c95c4ac6991aaac6ae1d735b : SUCCESS in 12m 42s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95afdd10f21f436b8a9fb82f26d887a5 : SUCCESS in 12m 31s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5cf9373d47fc42db8712e64cccb77aee : FAILURE in 25m 55s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86144686fc6546a587f8496010ceb16e : FAILURE in 28m 58s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eaa6674ebd444c8891e7d45bc2e32c2a : FAILURE in 27m 47s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3432766105d146169f10185d3cf3cbd2 : FAILURE in 30m 22s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4c805882971248ef8c065a16e81255da : SUCCESS in 15m 07s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4190d88ca85349b6b4f766b6cbce0320 : SUCCESS in 51m 43s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fda51f001e144169bb04bff291a26eb6 : SUCCESS in 57m 00s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/af85405ba64f4b2d897ad02936509463 : SUCCESS in 47m 02s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/82ff80896e73410e965de560ca7b47d7 : SUCCESS in 43m 08s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3906d155846945c1b64c331143347ed7 : SUCCESS in 14m 09s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/a8ba7512f95e4a14a5391a4ae90e9cc9 : SUCCESS in 18m 33s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/443206799ece40279ffd7872dcfc43d0 : FAILURE in 9m 10s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67843596240a4dc489ddbabae7542de4 : SUCCESS in 16m 19s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0277cdc72c5e408cb9ab8d52e852804e : SUCCESS in 13m 53s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/4f1bd41232e2434eb78e03d75580ad6e : SUCCESS in 1h 52m 34s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/8e46a0749c3b42a3bf15592a458dad41 : SUCCESS in 40m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e54b9f0f8944f57b12e69daefe18146 : SUCCESS in 3m 30s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f5205d43c56a44ceb07c46495fcd4286 : SUCCESS in 1h 08m 02s","accounts_in_message":[],"_revision_number":1},{"id":"5fed8a1730af770babc1143d4e627128be179d94","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-05-09 22:05:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6faab3a920a0d36d4364cbe7e2baaf76f60412cf","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-05-11 09:03:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"de2028e6556b395796c6f3c83298e17361e4b5ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 09:05:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0549fc9dec2d977a9a20b4aee5deff88fb90d889","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 11:44:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46a8e262365143039fab30676bd9132d : SUCCESS in 3m 21s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91f0a223db9d4b0e85e542288c5d8cfb : SUCCESS in 3m 02s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ebc2f49c8768480aa31093601bb86012 : SUCCESS in 4m 16s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/ae15e981cb0b4e09ba4a2799423d7b4a : SUCCESS in 22m 16s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4cddc138b2704273aacf9fbb8fd675fa : SUCCESS in 21m 13s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a616e7f981c4c648b163cbfb29046d1 : SUCCESS in 17m 17s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/94eda4620f9145e6bead3ce04c72e745 : SUCCESS in 53m 31s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f424d8e8fa664898ae74d00b859ddc8b : SUCCESS in 1h 02m 15s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/543726e334f7495ab582ea71b1b4d60c : SUCCESS in 1h 02m 02s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/422ce6776ef74afe8f4f31b2b33b70d6 : SUCCESS in 53m 36s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/c42dfbe03e3744bd9d0f7fd63c67f403 : SUCCESS in 23m 28s\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/ea0ef6d22d8742f493f1964a865b8d38 : SUCCESS in 2h 33m 37s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/be35d48dd7e84981a6048ad7143b04a9 : SUCCESS in 48m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cdbf804c0e6a46398963f62c1c44c708 : SUCCESS in 3m 36s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e79fc370f73f4666a9d13fd35873af6d : SUCCESS in 1h 16m 19s","accounts_in_message":[],"_revision_number":1},{"id":"216fb7e9ffeb59a97a63300fe662789f38cce639","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 11:44:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"233d0a19b857b4d1850b4e03df3d597c3f80bb1e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 11:45:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/094bd26c0dce4f8eaa8268b1ca9b922a : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0815ace85b3caf73f2ffd0b9ba30c630eaf3a950","revisions":{"0815ace85b3caf73f2ffd0b9ba30c630eaf3a950":{"kind":"REWORK","_number":1,"created":"2022-05-07 06:32:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/841005/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/05/841005/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/05/841005/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/05/841005/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/05/841005/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/05/841005/1"}}},"commit":{"parents":[{"commit":"df0d55cf784bd50b9f858a28ba35134197c21305","subject":"Merge \"OVN: Clean up deprecated parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/df0d55cf784bd50b9f858a28ba35134197c21305"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-07 06:32:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-07 06:32:17.000000000","tz":540},"subject":"Remove allow_overlapping_ips","message":"Remove allow_overlapping_ips\n\nThe parameter was already removed from Neutron.\n\nDepends-on: https://review.opendev.org/837286\nChange-Id: I7f95546d447e057de09a2e444a2957c805a47edb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/0815ace85b3caf73f2ffd0b9ba30c630eaf3a950"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/0815ace85b3caf73f2ffd0b9ba30c630eaf3a950"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
