)]}'
{"id":"openstack%2Fpuppet-neutron~845308","triplet_id":"openstack%2Fpuppet-neutron~master~I101db1ffbe7c183cd34d8a0162c305e503e55d43","project":"openstack/puppet-neutron","branch":"master","topic":"macvtap","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-06-13 21:59:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I101db1ffbe7c183cd34d8a0162c305e503e55d43","subject":"Add macvtap mechanism driver support","status":"MERGED","created":"2022-06-10 00:19:27.000000000","updated":"2022-06-13 22:00:59.000000000","submitted":"2022-06-13 21:59:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"845308-macvtap","meta_rev_id":"ee88f8c64c3520c617600333ff678ad93be6519e","_number":845308,"virtual_id_number":845308,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-13 21:59:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-06-13 13:53:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-06-13 19:20:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-06-13 19:20:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-10 05:20:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-13 13:53:46.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-06-13 19:20:18.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"7d2154134f88b8d0bcdda65239dc43d0484ee679","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-10 00:19:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c308c9add2c80283e2e92208003fb7380f5d543","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-10 00:46:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fcd58f3e77fad2a062637fc27f6a56ff47c74ad5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-10 01:18:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"34524aeff6e6309a3281921048b305e3dfd83eae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-10 01:27:56.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0927eb4f2e00f06a91c26ba5f205754eaf7e14ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-10 02:49:04.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"75cf400db314771478124458eb6b21d9e2dd3bce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 05:20:25.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d7ca4703e64404c94a104d791229c05 : SUCCESS in 3m 04s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f971db34c094df489b35e570ed155b3 : SUCCESS in 2m 53s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8fa4b739be44ebca9c45d5759982c78 : SUCCESS in 3m 26s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/ed47ec7cfc774a75810275bbea312b6c : FAILURE in 15m 52s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/871d9451223f4d1898f6e644f84d9c7b : FAILURE in 16m 07s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e1ee0c5adffc46efa4cb5bcdebb62f76 : FAILURE in 12m 14s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ac85d09acab46fd885aa71ab616a59b : FAILURE in 11m 18s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0d534a4ae6448ba914a1c8fb8e99d6c : FAILURE in 22m 47s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/180e612b3ed743f39d554397d94e51c3 : FAILURE in 40m 22s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/806de0e8c10e41dba7a7f6dbf516cd9f : FAILURE in 38m 24s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/236972dbd7354e5194310c76c08bc94b : FAILURE in 24m 43s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2367ade774f2434ebc8ef4a6b93eefc3 : SUCCESS in 22m 33s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/eaa8fd717a0d45f1b71e49f5bf2a0e70 : SUCCESS in 56m 47s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f22de34fad0a4983a30bf332efbde74e : SUCCESS in 1h 08m 12s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1483cc73bacd483ba306e4bb25590d01 : SUCCESS in 1h 05m 19s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dab99b0bd89d4d30b62b0decbec48b89 : SUCCESS in 54m 46s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d0b36d6c7d9450294a6a339a54f505a : SUCCESS in 20m 07s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/34f116d6ce3e4a68990ff49dae5e6a6e : FAILURE in 11m 33s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/500b76f10d1a4b0aa7f102c0db0933dd : SUCCESS in 17m 12s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb8136d5f6364054ae4e5b6ede8d0b0f : SUCCESS in 15m 47s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/6447a3e208194d759453a90ffb1d8155 : SUCCESS in 2h 25m 04s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/9b5db04e42164d45a8413f6ae982dcc1 : SUCCESS in 1h 00m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09eccde400f84abebbd203f9895410e6 : SUCCESS in 3m 25s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7b171d88641748cda49697decf40d9ae : SUCCESS in 1h 14m 29s","accounts_in_message":[],"_revision_number":5},{"id":"dcb8571cade26fd180391de61bec3038b05bc8f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-10 05:26:33.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"4e263f9583424f7b71446ffd0d41f4dc06784967","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 08:13:34.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1ef8f6a53824a8fa1e9dee483b622a7 : SUCCESS in 3m 41s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b428f0232aa4aa08903199a621ab9d8 : SUCCESS in 3m 01s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/658bf8865479450993aa5f6db99540d9 : SUCCESS in 2m 43s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/66a8e3aefbbe4f36ad0bed66c9de65c9 : SUCCESS in 15m 43s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c7a0a615867400aaa0c41ccb609b92f : SUCCESS in 14m 19s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65e5ac2007f344e79827e2ff30ee4148 : SUCCESS in 13m 59s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91757b1b9bc54f7fbce38964904d1370 : SUCCESS in 12m 45s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b1b1b90ee8746b79de1c26538eb8172 : FAILURE in 21m 38s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c25772080e44e22acadeac5948c505c : FAILURE in 41m 39s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5bda35f84380437ba95c0fc6d374afda : FAILURE in 36m 19s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c46ebd5ff6814974847cd3e1dead586d : FAILURE in 21m 04s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/616bf72284be436d92d75a1344c74b09 : SUCCESS in 19m 53s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b85c019390894d97b8e7ad94eaa92982 : SUCCESS in 55m 03s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2eaa4ccd60e14ba6b2147c952b043f78 : SUCCESS in 1h 06m 30s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a6a1aec5e182409db37355f01260f648 : SUCCESS in 1h 02m 29s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9c5313c217344eef9841a1a31880d88c : SUCCESS in 51m 10s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04ba54d1996f4ede819f0ebddf6513bb : SUCCESS in 14m 14s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/aa886081f4874053b21e9c19863263d6 : FAILURE in 7m 57s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7306a11cfb264448af6e77c63f06ee09 : SUCCESS in 13m 53s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c294dc16a594e359e87136faea82fef : SUCCESS in 14m 04s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/07bafce941bb45609f90c8574ddcedf6 : SUCCESS in 2h 41m 49s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/e7ac78ff349c4e1c818f7ad6c562b40b : SUCCESS in 48m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a00ef278b88d488d9c9c632a61cb4ff1 : SUCCESS in 2m 55s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/fc033d068f504ff8bbc20413b70dd294 : SUCCESS in 1h 31m 54s","accounts_in_message":[],"_revision_number":6},{"id":"d885ec3644b38a94f4c7c54fa27b9e89a3575082","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-06-13 13:53:46.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"333cb2654c6fdbb29b8f41c14f2983b20037381d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-06-13 19:20:18.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"556a85b9a67227db034415557f76c595e78f6cf2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 19:21:50.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"8e28c4cac9266fabf5e6e57915d2ded425040e0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 21:59:52.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c21cb3b980c94c06b2d11a381dc934a7 : SUCCESS in 3m 05s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c3406a901094eb7a8fa8bc303aaf432 : SUCCESS in 3m 06s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2220fa62f5e74100b0d1ae9196390547 : SUCCESS in 3m 57s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/fe44a0bed85c435a861bd72e2f12c9b0 : SUCCESS in 18m 30s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a9ae4e6d3204fbbb25c4866bd9b091f : SUCCESS in 14m 19s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c93c3c2ea974400885400e48ac80cf5a : SUCCESS in 16m 37s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a6dadb0ea5cf440d8566532e301d3e32 : SUCCESS in 1h 12m 21s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1a6e74581b564cad9cea2ba08eb30735 : SUCCESS in 1h 16m 26s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/21c8ed4cb11745c4abedea6031af6a81 : SUCCESS in 1h 04m 11s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/edbbe85e67b247eab13bf1fdb546157e : SUCCESS in 57m 16s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/7aaca1749c104bd8af5e40dee2329ae9 : SUCCESS in 21m 30s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/8e5c9c522e4241ebbf8ff1ba95345516 : SUCCESS in 2h 33m 44s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/e04539b06e76493db0cc522e0a71ce45 : SUCCESS in 52m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36e3416ce4c546f69cd1d753a6cebf28 : SUCCESS in 3m 40s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/10f81c1e5f62477da00f5c70001b86c1 : SUCCESS in 1h 20m 09s","accounts_in_message":[],"_revision_number":6},{"id":"0173d616dccf135118bf90ea99e79fd461eb670e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 21:59:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"ee88f8c64c3520c617600333ff678ad93be6519e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 22:00:59.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/efbd0fcf799d4d82aaee2b56022949f6 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"fecb262ede1de69323a7f4b730a0961af28aa801","revisions":{"6b0757afc3fcbe8dad8fd946cbde4c7359be2652":{"kind":"REWORK","_number":1,"created":"2022-06-10 00:19:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/08/845308/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/08/845308/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/08/845308/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/08/845308/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/08/845308/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/08/845308/1"}}},"commit":{"parents":[{"commit":"c86f68df9661d7bc872742050db58a6bc2e3c525","subject":"Merge \"Remove support for CentOS 8 Stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c86f68df9661d7bc872742050db58a6bc2e3c525"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-10 00:17:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-10 00:19:07.000000000","tz":540},"subject":"Add macvtap mechanism driver support","message":"Add macvtap mechanism driver support\n\nThis change adds support for macvtap mechanism driver, which is part\nof the ml2 plugin.\n\nChange-Id: I101db1ffbe7c183cd34d8a0162c305e503e55d43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6b0757afc3fcbe8dad8fd946cbde4c7359be2652"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6b0757afc3fcbe8dad8fd946cbde4c7359be2652"}]},"branch":"refs/heads/master"},"b4efac4f8d44afa82f8f6ceff0922970de76391c":{"kind":"REWORK","_number":2,"created":"2022-06-10 00:46:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/08/845308/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/08/845308/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/08/845308/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/08/845308/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/08/845308/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/08/845308/2"}}},"commit":{"parents":[{"commit":"c86f68df9661d7bc872742050db58a6bc2e3c525","subject":"Merge \"Remove support for CentOS 8 Stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c86f68df9661d7bc872742050db58a6bc2e3c525"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-10 00:17:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-10 00:46:15.000000000","tz":540},"subject":"Add macvtap mechanism driver support","message":"Add macvtap mechanism driver support\n\nThis change adds support for macvtap mechanism driver, which is part\nof the ml2 plugin.\n\nChange-Id: I101db1ffbe7c183cd34d8a0162c305e503e55d43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b4efac4f8d44afa82f8f6ceff0922970de76391c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b4efac4f8d44afa82f8f6ceff0922970de76391c"}]},"branch":"refs/heads/master"},"426580e840c16856c65238069ac47e5e5e499a43":{"kind":"REWORK","_number":3,"created":"2022-06-10 01:18:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/08/845308/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/08/845308/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/08/845308/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/08/845308/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/08/845308/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/08/845308/3"}}},"commit":{"parents":[{"commit":"c86f68df9661d7bc872742050db58a6bc2e3c525","subject":"Merge \"Remove support for CentOS 8 Stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c86f68df9661d7bc872742050db58a6bc2e3c525"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-10 00:17:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-10 01:18:45.000000000","tz":540},"subject":"Add macvtap mechanism driver support","message":"Add macvtap mechanism driver support\n\nThis change adds support for macvtap mechanism driver, which is part\nof the ml2 plugin.\n\nChange-Id: I101db1ffbe7c183cd34d8a0162c305e503e55d43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/426580e840c16856c65238069ac47e5e5e499a43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/426580e840c16856c65238069ac47e5e5e499a43"}]},"branch":"refs/heads/master"},"411cc9dbad7e1fffa38af476048c51f919944e03":{"kind":"REWORK","_number":4,"created":"2022-06-10 01:27:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/08/845308/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/08/845308/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/08/845308/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/08/845308/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/08/845308/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/08/845308/4"}}},"commit":{"parents":[{"commit":"c86f68df9661d7bc872742050db58a6bc2e3c525","subject":"Merge \"Remove support for CentOS 8 Stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c86f68df9661d7bc872742050db58a6bc2e3c525"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-10 00:17:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-10 01:27:46.000000000","tz":540},"subject":"Add macvtap mechanism driver support","message":"Add macvtap mechanism driver support\n\nThis change adds support for macvtap mechanism driver, which is part\nof the ml2 plugin.\n\nChange-Id: I101db1ffbe7c183cd34d8a0162c305e503e55d43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/411cc9dbad7e1fffa38af476048c51f919944e03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/411cc9dbad7e1fffa38af476048c51f919944e03"}]},"branch":"refs/heads/master"},"b9406fbe33d95f513fd514475706fd8d42dbd2a3":{"kind":"REWORK","_number":5,"created":"2022-06-10 02:49:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/08/845308/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/08/845308/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/08/845308/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/08/845308/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/08/845308/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/08/845308/5"}}},"commit":{"parents":[{"commit":"c86f68df9661d7bc872742050db58a6bc2e3c525","subject":"Merge \"Remove support for CentOS 8 Stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c86f68df9661d7bc872742050db58a6bc2e3c525"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-10 00:17:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-10 02:48:54.000000000","tz":540},"subject":"Add macvtap mechanism driver support","message":"Add macvtap mechanism driver support\n\nThis change adds support for macvtap mechanism driver, which is part\nof the ml2 plugin.\n\nChange-Id: I101db1ffbe7c183cd34d8a0162c305e503e55d43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b9406fbe33d95f513fd514475706fd8d42dbd2a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b9406fbe33d95f513fd514475706fd8d42dbd2a3"}]},"branch":"refs/heads/master"},"fecb262ede1de69323a7f4b730a0961af28aa801":{"kind":"REWORK","_number":6,"created":"2022-06-10 05:26:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/08/845308/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/08/845308/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/08/845308/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/08/845308/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/08/845308/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/08/845308/6"}}},"commit":{"parents":[{"commit":"c86f68df9661d7bc872742050db58a6bc2e3c525","subject":"Merge \"Remove support for CentOS 8 Stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c86f68df9661d7bc872742050db58a6bc2e3c525"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-10 00:17:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-10 05:26:23.000000000","tz":540},"subject":"Add macvtap mechanism driver support","message":"Add macvtap mechanism driver support\n\nThis change adds support for macvtap mechanism driver, which is part\nof the ml2 plugin.\n\nChange-Id: I101db1ffbe7c183cd34d8a0162c305e503e55d43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/fecb262ede1de69323a7f4b730a0961af28aa801"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/fecb262ede1de69323a7f4b730a0961af28aa801"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
