)]}'
{"id":"openstack%2Fpuppet-neutron~845309","triplet_id":"openstack%2Fpuppet-neutron~master~Ib34c67905db31e32653314c8873a938dcbd04ed3","project":"openstack/puppet-neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-06-14 10:59:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib34c67905db31e32653314c8873a938dcbd04ed3","subject":"Deprecate support for linuxbridge plugin config","status":"MERGED","created":"2022-06-10 00:26:19.000000000","updated":"2022-06-14 11:00:53.000000000","submitted":"2022-06-14 10:59:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"845309","meta_rev_id":"3c8c4d841ec7581c0fc1e1c83d3f85874542fede","_number":845309,"virtual_id_number":845309,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-14 10:59:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-06-13 13:53:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-06-14 08:07:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-06-14 08:07:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-10 02:52:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-13 13:53:57.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-06-14 08:07:45.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"03cd343ab3a04960524a5663f60b4fac54bae272","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-10 00:26:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9e2e0712b88b3266bfe406258f95025c4347c3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-10 00:35:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"eb675ec3ec4482f5a904c2e8a820356ce19279a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 02:52:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ba95c9988464dd897938db5d73b50c7 : SUCCESS in 3m 40s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/465651c4d6164b08aa8b47729ffd16e6 : SUCCESS in 3m 01s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9b8da0df1fd4f019d7c6e73f26e690b : SUCCESS in 3m 32s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/3a1ff54edeec44b5ad7db2ce04cd7657 : SUCCESS in 17m 07s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e08c023e66c4cbda508d8e9b9635fce : SUCCESS in 14m 57s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e394bc63d1e4b7285f5919687e09c92 : SUCCESS in 14m 05s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb2ccd263891407ba6683e404a84ba6f : SUCCESS in 13m 23s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf01dd41a1dd4bf8a860ec0df240afa6 : FAILURE in 22m 57s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/618914b985a94e7b870377868f0618b7 : FAILURE in 42m 18s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aae1474a626948078c446176c9fda3e7 : FAILURE in 37m 24s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eaff2fb76a56422f80bb62a5aa7351c2 : FAILURE in 22m 55s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fe56731da7ce4b24bb74fcd3209dddbb : SUCCESS in 21m 54s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d65988c4b2cf4834921591ee31f01162 : SUCCESS in 57m 34s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/23f029a1281842e786e40ba258436dd8 : SUCCESS in 1h 06m 31s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6a013af2a38440c98df4dbed0772e626 : SUCCESS in 1h 02m 04s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/850586563d214d69b28bea0a65d045ba : SUCCESS in 52m 21s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14c2d8324c7945b188eaec95c377acb1 : SUCCESS in 18m 01s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/39c8bb1d8c634e45b8efc3021d1d4e1a : FAILURE in 11m 05s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad2310fff5004deda138779e1ec0a4d5 : SUCCESS in 17m 46s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c7600fe046f419b9c811fa01960631c : SUCCESS in 17m 28s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/b5b21709bd524015a149af321a258200 : SUCCESS in 2h 13m 19s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/af4f4562796c4a72b3ed60bfa7b5258c : SUCCESS in 43m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ede7cb16afb74a57856378a10c126f06 : SUCCESS in 4m 01s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e1a449c7417140458fb204b0799d80f0 : SUCCESS in 1h 08m 59s","accounts_in_message":[],"_revision_number":2},{"id":"1eb2c5e3fb1021509eb9c9ef59dbab162eaa8a0b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-06-13 13:53:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"680d276ec75808bbb81f9417220ac7ea56e4b4ea","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-06-14 08:07:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"62eb1072ba4c11c57b034102ecbd3c643a5419a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-14 08:09:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3580fe7fab889f37b09a6c4d74ab7a689e85392b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-14 10:59:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82b285148dd24a159f9cac8e8a993690 : SUCCESS in 3m 45s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c16249b288e443edb5e57da44273947d : SUCCESS in 3m 51s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a46ce69ddd544a2bb4f138fd29ca4a7 : SUCCESS in 3m 37s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/00a0005c61fd455b98d97b3206ed6a53 : SUCCESS in 17m 05s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6bef66d6118e4ea6a53c642a128b5571 : SUCCESS in 15m 49s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9bde86a90074c68a5252a47999f9655 : SUCCESS in 15m 09s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c00d27b11b374388b06d453fb832c614 : SUCCESS in 59m 48s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0443852cc526415ab02f5ab324416d90 : SUCCESS in 1h 08m 11s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc3b51b874014e869a01082d208b237e : SUCCESS in 1h 12m 00s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a4db31cb801c43e3ac4786910c625059 : SUCCESS in 55m 37s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/770793a2fdaa47e293600537599aeddc : SUCCESS in 23m 15s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/13744a3edadb4e97912b37e8d658625a : SUCCESS in 2h 44m 45s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/004a1d203c5f471282645e32c71dd868 : SUCCESS in 57m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24b84f296a1b4d5f89b05be2f2f8ef4b : SUCCESS in 3m 38s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6f57634a9cd74de2ab8fe21bb05fedc7 : SUCCESS in 1h 21m 58s","accounts_in_message":[],"_revision_number":2},{"id":"593de71866e523f7df6763ac72e86011825757d0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-14 10:59:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3c8c4d841ec7581c0fc1e1c83d3f85874542fede","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-14 11:00:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6e66ca4259f4347a7734b6c6c711886 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"959c1cefeea69b65c67670b0ad44376843c9836f","revisions":{"941255739a8326b8a7387527645b802ef80fae57":{"kind":"REWORK","_number":1,"created":"2022-06-10 00:26:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/09/845309/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/09/845309/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/09/845309/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/09/845309/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/09/845309/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/09/845309/1"}}},"commit":{"parents":[{"commit":"c86f68df9661d7bc872742050db58a6bc2e3c525","subject":"Merge \"Remove support for CentOS 8 Stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c86f68df9661d7bc872742050db58a6bc2e3c525"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-10 00:25:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-10 00:25:24.000000000","tz":540},"subject":"Deprecate support for linuxbridge plugin config","message":"Deprecate support for linuxbridge plugin config\n\nThe file no longer exists since linuxbridge was integrated to the ml2\nplugin as a mechanism driver.\n\nChange-Id: Ib34c67905db31e32653314c8873a938dcbd04ed3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/941255739a8326b8a7387527645b802ef80fae57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/941255739a8326b8a7387527645b802ef80fae57"}]},"branch":"refs/heads/master"},"959c1cefeea69b65c67670b0ad44376843c9836f":{"kind":"REWORK","_number":2,"created":"2022-06-10 00:35:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/09/845309/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/09/845309/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/09/845309/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/09/845309/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/09/845309/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/09/845309/2"}}},"commit":{"parents":[{"commit":"c86f68df9661d7bc872742050db58a6bc2e3c525","subject":"Merge \"Remove support for CentOS 8 Stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c86f68df9661d7bc872742050db58a6bc2e3c525"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-10 00:25:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-10 00:35:17.000000000","tz":540},"subject":"Deprecate support for linuxbridge plugin config","message":"Deprecate support for linuxbridge plugin config\n\nThe file no longer exists since linuxbridge was integrated to the ml2\nplugin as a mechanism driver.\n\nChange-Id: Ib34c67905db31e32653314c8873a938dcbd04ed3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/959c1cefeea69b65c67670b0ad44376843c9836f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/959c1cefeea69b65c67670b0ad44376843c9836f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
