)]}'
{"id":"openstack%2Fpuppet-neutron~847266","triplet_id":"openstack%2Fpuppet-neutron~stable%2Fxena~I49888ab1c9f7c82d79e685b2e30e93f84257709e","project":"openstack/puppet-neutron","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-06-28 15:14:59.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I49888ab1c9f7c82d79e685b2e30e93f84257709e","subject":"ml2+ovs: Accept a string value for permitted_ethertypes","status":"MERGED","created":"2022-06-23 23:30:46.000000000","updated":"2022-06-28 18:10:44.000000000","submitted":"2022-06-28 18:10:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"847266","cherry_pick_of_change":846565,"cherry_pick_of_patch_set":1,"meta_rev_id":"a882b70ea206653378c57207c2c7f01cd8959588","_number":847266,"virtual_id_number":847266,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-28 18:10:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-06-24 13:32:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-06-28 13:36:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-06-28 13:36:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-24 01:14:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-24 13:32:03.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-06-28 13:36:20.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"eb116d2af53240153c198a01dd28a8982413ef8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-23 23:30:46.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"698192cb683a29dd7ae30ebff365b726b1168fbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 01:14:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f583ce843149431da1ba9040560e07ee : SUCCESS in 3m 09s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e66a475045274923bc357b4832e3aaa0 : SUCCESS in 3m 11s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae013785e38b4d3599b01a1e1825d0fe : SUCCESS in 3m 21s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/6b5ab07ca8ff4dbfa7a4b18e107973bc : SUCCESS in 17m 41s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/8e81252f53d54e0ca7f6ef2fecb004b4 : SUCCESS in 18m 12s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a1ed58ad5be40a1b4174e4a57a759bc : SUCCESS in 13m 45s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f76352ec12e44c68b439a45040f9218a : SUCCESS in 14m 29s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fcafab4e775241d38aaf5f93824c6afb : SUCCESS in 13m 52s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a75c3c6d6a848da99387fc6a1b3bc1d : FAILURE in 31m 49s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05f85b053f9541268bc58c95c44f0b3d : FAILURE in 59m 10s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba199f454b0c4e31a4930d6835742b84 : SUCCESS in 46m 20s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/129e6a9b3a654b34966953a9de61f7b1 : SUCCESS in 38m 04s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/24230169da17429b82e005007841e658 : SUCCESS in 24m 07s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/86aabddd47734999978ed0a65477aaa0 : SUCCESS in 1h 26m 34s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1da58ae3bfb64d9783b7c310c3a7584e : SUCCESS in 1h 37m 09s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fda78929e3a34cf6995985a169880dde : SUCCESS in 1h 22m 48s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/136ca9598f94451d88e2ea5d48459164 : SUCCESS in 1h 06m 28s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36228bbccddb47d49c073e6f1b02135a : SUCCESS in 15m 13s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/cc5ab9b745144a47b68ca519baacf4c3 : SUCCESS in 22m 37s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1ccf038510c4f28a5a07c3c735d6904 : SUCCESS in 16m 32s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb049747ae8f43558ec1713cb8250c5e : SUCCESS in 16m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7b2154480f0606872b114bba78463068590d6ff2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-06-24 13:32:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a0ca25f250fb7f95e47daeb3ab1e5a17b695ae17","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-06-28 13:36:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6729858dce5a834a9e89bf7a90aa7a653d9c0a15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 13:39:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"60dc57dd3f797d1442612a296076bf89eb0a91a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 15:14:16.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f719e9cb44c4afb9f2414b799e04f31 : SUCCESS in 3m 17s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7209947f3ad4682bc6a0253311babe7 : SUCCESS in 3m 18s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/100cf4636491403ab6273b75d7e0aec8 : SUCCESS in 3m 17s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/0acc037e33074993b00f8044549d4ab1 : SUCCESS in 18m 42s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/1079b2eb85414aefaef8fc2c1ee10687 : SUCCESS in 17m 46s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03dfeaa1a46b40b0bb65e7c3f2bf447e : SUCCESS in 14m 25s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3781b74d46a45e5b69afe6df4d0344c : SUCCESS in 14m 41s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/affde720bf414502b2f91acc72687853 : SUCCESS in 1h 09m 00s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/60750c67f414461aaffc5192650d027e : FAILURE in 1h 34m 07s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a63f421e40084d8c8fa903ead64c2a7c : SUCCESS in 1h 16m 56s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ccc242ebb98e4cbbae30665f52a3e8d5 : SUCCESS in 55m 20s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/32e62f691c534223a815e08e7fe3b4cc : SUCCESS in 24m 31s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6afde1a8fdb41e9a152d215ce4b2422 : SUCCESS in 18m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a6f089ae359753f970869e044315743daabae2f3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-28 15:14:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b8f0de533e9b888c6f58a305ed5c0606aa112801","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 16:52:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9820c32ec67049bcac447765a4250f83 : SUCCESS in 3m 18s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bcde6e354a7042eab45bf53bbd567eed : SUCCESS in 4m 06s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab079cc7df6c415e8263386dc9cc177e : SUCCESS in 3m 51s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/62d152afbd604b1a8d62a1a3270a05e9 : SUCCESS in 18m 11s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/eef4c89ad859488dbbc9309aab1a2407 : SUCCESS in 18m 51s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c30a2e5fcbd42f09b276b4d3263275d : SUCCESS in 14m 28s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/175e4784a88748398fea9141d6f7132e : SUCCESS in 16m 22s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6fe36aa5cf8d4a4498f1bc4d232910eb : SUCCESS in 15m 54s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5511dc92d7f14c50a81e7e421d772f8e : SUCCESS in 49m 37s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/811dbb8934234bebb226bc7e38a8d60c : FAILURE in 56m 40s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53dc973eca824d67a0432060ba8490c2 : SUCCESS in 44m 45s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8cd872503d0845159416d0b078743749 : SUCCESS in 35m 54s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/43d40201c3b84534a6c060b293f30b65 : SUCCESS in 23m 54s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d21693032ef945c29e14f04c686db167 : SUCCESS in 1h 27m 07s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/58e83c9edd3f480c8c3226ea6b1130bf : SUCCESS in 1h 17m 58s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/035a9982d00b4fdca5dd5ed30198b9b4 : SUCCESS in 1h 30m 42s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3c591ae832d540409b33fb08235b61e2 : SUCCESS in 1h 00m 45s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/437c5340957443289299f3d9a5311f9c : SUCCESS in 17m 55s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/b14db517d6b743aeacabef0e405b5592 : SUCCESS in 22m 11s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1eee331051c64185b78297f403fd878b : SUCCESS in 17m 01s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae10b231a38a415183528caf0e36a1fa : SUCCESS in 17m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0fe35fe5d2a72277f75d06551f281a0ff9477f53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 16:53:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"01ad010a18d8e090070d94668004390878577740","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 18:10:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de51099a5e1d4f13bb65817a2bc28425 : SUCCESS in 4m 08s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d8cc2441c2548e294def03b9d012aa2 : SUCCESS in 3m 55s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0630748f578843fa8932f8e37270ed3f : SUCCESS in 4m 10s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/23bb40f0b9434707bee1ac958fad1bb1 : SUCCESS in 19m 22s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/b5d22090995542ffbb706c789fb12e90 : SUCCESS in 17m 17s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8fb5af91d89e4bc7ad6bcba9a711b3e3 : SUCCESS in 15m 14s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee179445c70b4755ac3f351d491da7c6 : SUCCESS in 10m 54s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ab59e56e4c1847599a052b73b92aacdf : SUCCESS in 1h 04m 06s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e4eaf0e552dc48bc896a0031a5683b77 : SUCCESS in 1h 13m 03s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/91232c67e7cb4fffb445ec4adec4a0ee : SUCCESS in 1h 10m 52s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/deeea40bb05147e4ad7d8440d2ff690d : SUCCESS in 53m 25s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/9ee80741b4164351a0bace45a9143313 : SUCCESS in 21m 47s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75613036dc8f432db6f0ac589349d011 : SUCCESS in 18m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a882b70ea206653378c57207c2c7f01cd8959588","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 18:10:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b59fca5c8067cb4463af1c376a21a1a604efc767","revisions":{"b59fca5c8067cb4463af1c376a21a1a604efc767":{"kind":"REWORK","_number":1,"created":"2022-06-23 23:30:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/66/847266/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/66/847266/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/66/847266/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/66/847266/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/66/847266/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/66/847266/1"}}},"commit":{"parents":[{"commit":"7864f0a8dc63c74c3eec52aa24c6bc4f109a9c93","subject":"NSX Plugin: nsx_api_password should be secret","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/7864f0a8dc63c74c3eec52aa24c6bc4f109a9c93"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-20 11:46:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-23 23:30:46.000000000","tz":0},"subject":"ml2+ovs: Accept a string value for permitted_ethertypes","message":"ml2+ovs: Accept a string value for permitted_ethertypes\n\n... as is done for the other ListOpt parameters. Also, this change\nensures the parameter is properly purged instead of being unmanaged.\n\nChange-Id: I49888ab1c9f7c82d79e685b2e30e93f84257709e\n(cherry picked from commit dcb4622275ba788fe84cc11233d2d1a60a38080b)\n(cherry picked from commit 69aee692c02846a9fbb69dfca75d8451f39f2629)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b59fca5c8067cb4463af1c376a21a1a604efc767"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b59fca5c8067cb4463af1c376a21a1a604efc767"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
