)]}'
{"id":"openstack%2Fpuppet-neutron~849201","triplet_id":"openstack%2Fpuppet-neutron~master~If959e3751f881107bf5e697f044df99406677bd2","project":"openstack/puppet-neutron","branch":"master","topic":"sriov-plugin-config","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-01 23:05:45.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"If959e3751f881107bf5e697f044df99406677bd2","subject":"Remove unused resrouce type to manage ml2_conf_sriov.ini","status":"MERGED","created":"2022-07-09 14:59:49.000000000","updated":"2022-08-02 05:01:23.000000000","submitted":"2022-08-02 05:00:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849201-sriov-plugin-config","meta_rev_id":"b7b66d4d41ad1dbc3741d8801d89491ae3ae7889","_number":849201,"virtual_id_number":849201,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-02 05:00:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-08-01 09:46:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2022-08-01 09:46:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-09 17:09:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-11 21:58:49.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-08-01 07:35:51.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"d3bae615a00755a4588cb536c2d48433904cebc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 14:59:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7076cb0e03cfd31b716cef049d9a9b125e22959f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 17:09:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b469ae4dedcf429998b95445864d8cbf\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53226756267b4da8b51687a040566de2 : SUCCESS in 4m 04s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/187bfa924adb4a24bbcde5c68bc6bc6b : SUCCESS in 2m 55s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76c55041e50b417bb6148ad19e69139c : SUCCESS in 2m 44s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/c34701a3bcac43e6ad84102fa6995dea : SUCCESS in 17m 10s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d2ad4cec4e54ff188c31401c3a9bbec : SUCCESS in 13m 59s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9640769abd0241c689afe4c5578c803f : SUCCESS in 14m 32s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b47a00879808437daf869dde78a332db : SUCCESS in 12m 22s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3e0901968ac4e0c800c4e3096bef311 : FAILURE in 24m 33s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59fba3bf7aa4437dad460a2c639e4b61 : FAILURE in 37m 56s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48f9f581ac99442d87e3ab6fb456a1be : FAILURE in 38m 12s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d4cccbe3f3e2478b807108c53b754eb0 : FAILURE in 21m 20s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb832b2ffbe1494d8b22ade3fa13c034 : FAILURE in 27m 57s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9db1665344e24359beffef60a7f9edd4 : SUCCESS in 19m 49s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ba86c71767c34fadb5eadb50032d2823 : SUCCESS in 55m 56s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c674cc76327a46d7a13ade6e79442732 : SUCCESS in 1h 07m 51s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/475ea5534b99404fb022a235478bea66 : SUCCESS in 1h 00m 19s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/60c72f4e13654d369155f96d9174937c : SUCCESS in 52m 25s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d87e5aaa5d93435f9f10774fdefa0856 : SUCCESS in 36m 59s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42cbab10a1fe4ca3822f5abf891bd849 : SUCCESS in 15m 37s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/f8cf2e31774d4b278b5131a5d34d45e9 : SUCCESS in 19m 25s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13db3c59de3944908acdbf592b34c654 : SUCCESS in 15m 04s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/491f12cb9b2740b3a06529761800161c : SUCCESS in 17m 42s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/3ec3890ec4384ea5bd404a6532179116 : SUCCESS in 2h 06m 06s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/3bb69b28d2784700a1f82334b8ffb979 : SUCCESS in 43m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2f33efcd9cf40afbbd2af220090de30 : SUCCESS in 3m 04s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3fa79cb298a74be28642cf2a795cd28f : SUCCESS in 59m 46s","accounts_in_message":[],"_revision_number":1},{"id":"fa53610ef2046b2c7f25a858e3d85b65fad1fcab","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-07-11 21:58:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"df2a772c539ac37bbea4441da6e6c99bed0826d0","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-08-01 07:35:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"24f2020ac50c050473172228b77b8916112e7921","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 09:41:36.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/puppet-neutron for 849201,1","accounts_in_message":[],"_revision_number":1},{"id":"739c2031b4936751f21e9bbdb8ef902646b39259","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-01 09:43:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"039ba605d7710adb66363a76c838fa682db5848d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-01 09:46:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4a1f7cd8d3924598a2990868b0598fa63e0d6bff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 12:20:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d5f91fee5234556a1e5a76f372a9732\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57b2716f1698435dba284f35fb1c773b : SUCCESS in 3m 39s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b31d5dc4a475452482518495250c985d : SUCCESS in 3m 37s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31693fdb28594bd9bafdba92249e0da7 : SUCCESS in 3m 19s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/e2fc8ed28d734d47b1fcdc210fbf20b5 : SUCCESS in 20m 06s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ced6d6dae3364d31a24e71e8d1818410 : SUCCESS in 14m 40s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7832a0572954445fbfd518bbdf8fa661 : SUCCESS in 22m 30s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49c60267fb7644999bb7e7df146b6248 : SUCCESS in 14m 47s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9be7c4c4a4e146f1824886efc599953a : FAILURE in 42m 17s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/974241b4a5bd4dbb83251f1969e95969 : FAILURE in 44m 13s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2fd296dafb404a30adaab5726a17bd17 : FAILURE in 39m 59s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b26916285b1b4641a299ec768566070a : FAILURE in 35m 37s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2a67910927248c68d8e94b6b81214ff : FAILURE in 34m 44s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7a829ee04c254de8b78c2dcb1599cb77 : SUCCESS in 23m 10s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2a1c46e378ae43e58a909d370326e2ab : FAILURE in 1h 06m 10s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d6c3058be62b485b9052fb79d220effe : SUCCESS in 1h 12m 02s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/aae8216b862846dabf554199e24ffc66 : SUCCESS in 1h 09m 47s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6107a3bc7e6647e9a8d59aeb83f672fc : SUCCESS in 53m 35s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ee190d2193b148c6b355e21020d6f371 : SUCCESS in 48m 42s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a531ce8497f244dfa8c3e2408c9e713f : FAILURE in 22m 12s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/ca9b3b184075411386bb3f02be37c4d0 : SUCCESS in 29m 28s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5fc1d7c374df4b11848603c2dde5c442 : FAILURE in 21m 53s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/769a1566a99c4cd8bded3b4b49baeef7 : FAILURE in 21m 46s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/a717c1346a604aa8846ff5eaa4e62e51 : SUCCESS in 2h 31m 06s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/2fc2d0ea36624050b084fa824adcbef1 : SUCCESS in 49m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbbc088b4fe0469e934a5f16b8fdd288 : SUCCESS in 4m 43s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e73d5553f05144979466e01d39a1a4b7 : SUCCESS in 1h 20m 24s","accounts_in_message":[],"_revision_number":2},{"id":"76d211df93ac3401eac43fdcb50cd70f2fa723c3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-01 12:22:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1d3e27d757739600981f0aa55b016928515e1547","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 15:00:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/397e69274c0449389a6382a30fcad348\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/97c6664a3bdc41b68f228b340286c81f : SUCCESS in 3m 52s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/17e0ea87ea994e2fbc3970d71f26725d : SUCCESS in 3m 48s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f921d8fd325b405a96e1e6dfbc295573 : SUCCESS in 4m 08s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/9333ea52d3394b999bd5d7e879d39aec : SUCCESS in 20m 06s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a3d77f14e214c0ea2e9db553369bfde : SUCCESS in 16m 16s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84cb2720b19841bb8a9bac91a1128e11 : SUCCESS in 15m 12s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/032b0b2cff4e420aae8ec89be45eaa7a : SUCCESS in 15m 26s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f98528d5c484a7ebbd284ef15e7fc7b : FAILURE in 36m 40s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88ec561270134893b3cf28b0da3e7872 : FAILURE in 38m 44s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9961b4e7886e40d68e7f6030421f6cff : FAILURE in 37m 51s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b5f859f8a72409fb02eb2be890c1b6f : FAILURE in 28m 54s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f273bc05a2884807b13a71e1a510d9b2 : FAILURE in 25m 40s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d59d804ac2bd4a45bb01ae864f26a2ab : SUCCESS in 21m 32s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a317df8c42bc41fe95d7ca909d27a498 : SUCCESS in 1h 02m 52s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d989b99ac0bc4ee7a2182248589a5483 : SUCCESS in 1h 06m 31s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9ec0a52b00e944cf8b092a20f6a58fb7 : SUCCESS in 1h 11m 37s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f675eff3fb9d4eefbd97de78b021b279 : SUCCESS in 51m 28s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9a7be5ce59bf45f58f40d52ec653ee9d : SUCCESS in 46m 01s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aed524ea2a024df7a6a7e63c4c5093c6 : FAILURE in 16m 48s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/e0a0ea7b37284fb3bf8922ba17507cfe : SUCCESS in 22m 32s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d4c76c7d723487597072574f3a9e802 : FAILURE in 21m 26s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc75a88686dd4b71ac2fee6fdec9839e : FAILURE in 20m 12s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/f1dbe76f81804ce5a3c8422171672286 : SUCCESS in 2h 33m 09s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/1bb498f5584d44518a8691f66bffd121 : SUCCESS in 55m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/974cd76fd9864cd8908ead8cb19ad6e4 : SUCCESS in 3m 50s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c9ecfa4e92524a05a149be154c1d79eb : SUCCESS in 1h 23m 03s","accounts_in_message":[],"_revision_number":2},{"id":"5222f72b9d764e4f65afc86ce4a143ddb92fdd79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 15:02:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bb1c9c8336eb2f70eef799c529eb2b00ddd2bdfa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 17:19:31.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b3c2f3b4b034407aa9b25ccb1a662127\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0929570265d41e6a18e2cd9e9955334 : SUCCESS in 4m 31s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f01a40f54604b22a9015c06c8cf09a2 : SUCCESS in 4m 06s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af239f9b379140ae88ea8ce5e5bd56dc : SUCCESS in 3m 46s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/2e30df4f5fb04e8c8c059f3f7d711121 : SUCCESS in 21m 34s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae9dc8b6b29040869347fe066361ae7c : SUCCESS in 13m 19s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc8d3d4e84f44b789af4a82be894d03f : SUCCESS in 15m 24s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c5e67b4a380c43d69961c19c5760040b : SUCCESS in 1h 01m 15s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/16d28c7a74f143568f9dbb1e9ea588b8 : SUCCESS in 1h 15m 51s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2e05609c8aac43a4a04336127f79fa80 : SUCCESS in 1h 09m 00s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fa673298a2b34c2d93f770551f861d1f : SUCCESS in 58m 48s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/131b7f0d59654660a4e397571cd10da4 : SUCCESS in 49m 10s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/467103b981c04190bb8f0a2465f3b15d : SUCCESS in 27m 35s\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/fd139c2644ba48f79bd999ed4e7b27ad : SUCCESS in 2h 15m 21s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/4c4c2642a3c347418520e0f4781a2c27 : SUCCESS in 40m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f86b502af4f34c398e937c176094766d : RETRY_LIMIT in 2m 10s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/20b69d6bbf794069ae6abf2534815813 : SUCCESS in 1h 13m 53s","accounts_in_message":[],"_revision_number":2},{"id":"ecf186b2ee2bcd23c62ceeb1707f91ddd46d7f75","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-01 23:05:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"813578847f984ebde6579d599afd3bfddae9790f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 01:55:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32a629ab41dd43afa197e5de9e499ff8\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b14fbf03f47341b39731de6945d6261e : SUCCESS in 3m 50s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cafa1861463547c0a9ccaa27eb311dc5 : SUCCESS in 3m 13s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4464be209ec43499ede20630fcd9d2a : SUCCESS in 3m 51s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/385c0f102bb34160887365dbeb867445 : SUCCESS in 23m 13s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b4d0d54628b4c1c9ae3b1ff4fd3e64e : SUCCESS in 16m 52s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57e36dd790bb43f89236767688f1c99d : SUCCESS in 16m 25s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c1c63fde367422c8d8e4680e5a188fc : SUCCESS in 17m 12s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b779c9a0fb314250b7fdc1c966f37c83 : FAILURE in 38m 44s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5157731dd26345d9a47be4f156fd8e69 : FAILURE in 53m 34s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/17f423bfbbbe49c5a1ef5b05a42b6919 : FAILURE in 43m 46s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c608cd4e0b844f2aaa09170b8dee68e : FAILURE in 35m 12s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1293bc022ad845d9b651449f9059c1ee : FAILURE in 28m 46s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fc9afc1151174812987195fac9654c4a : SUCCESS in 23m 36s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/344dab0df87f43858666fd73d100b576 : SUCCESS in 58m 22s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7beff08a43814bd9a7e79227c005657d : SUCCESS in 1h 14m 58s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3656b8890d5f4f7d84eecde70c847de4 : SUCCESS in 1h 12m 30s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ea74d19dad464064a1c4985928e693f9 : SUCCESS in 56m 40s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/02e8165c67974868b2eb408928c2a8fe : SUCCESS in 47m 30s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea19a1a0a42f4c69bc2cd4ad50fd0cb2 : FAILURE in 18m 08s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/defaa39ccedf4309976a0b3822915ddd : SUCCESS in 20m 38s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3261f9335e44a8e9a4cdc81242f0ea5 : FAILURE in 18m 20s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8f1e37525124d7ba82242bc9a14b309 : FAILURE in 17m 33s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/25080e630f3b47139a80ffacfd0cf796 : SUCCESS in 2h 42m 28s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/c82a9e7765e0451592dc0f2271faf9fc : SUCCESS in 50m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a291d3d44fb47efb70c9c7368586940 : SUCCESS in 3m 23s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/82131b8baf494e3482d761025a383af0 : SUCCESS in 1h 22m 03s","accounts_in_message":[],"_revision_number":2},{"id":"2310b39bfd4be8e30d7513d9ff34a8b7f789c2ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 01:56:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"72ee494d5b0778df8543be3c07482778a3130e44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 05:00:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/994b017ae44e4a078206053d96304e8f\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5f5ea3fd959415b8918c866e0a37b50 : SUCCESS in 3m 43s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33ba8342470d468993d449e309fd4762 : SUCCESS in 3m 35s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7dc4b2ef9e94fbe8fa58a809b655f95 : SUCCESS in 3m 26s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/0dc09463221947d3a16589359d109c47 : SUCCESS in 22m 04s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b6e3cc8172847ec84b976bdc21fbef9 : SUCCESS in 16m 27s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/feb8e9e9960f439285747520f5fa3843 : SUCCESS in 15m 21s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/15c249cae2b14231b832634ab379f7b5 : SUCCESS in 1h 05m 14s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6ea4e615f36f430b9f84fcd19b817f68 : SUCCESS in 1h 18m 22s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e67a6799e528406594095d1c2d20378a : SUCCESS in 1h 11m 51s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f8ce74edab0f4b3daa85f3482c59792e : SUCCESS in 1h 00m 26s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/abb32a0779794b57affb14075d2e4019 : SUCCESS in 48m 44s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/39cf532ff79f4746a13c72fdceb9cf61 : SUCCESS in 28m 00s\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/e2eacfb5ab5543deb431796665575595 : SUCCESS in 2h 57m 32s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/938565b62fa44d848525f93109702760 : SUCCESS in 53m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c4f42e411c1466586630d97b1dc7478 : SUCCESS in 3m 33s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0ce6204f54ba480186ea75b5d13a2ea2 : SUCCESS in 1h 23m 50s","accounts_in_message":[],"_revision_number":2},{"id":"30b78843d2bc3a38a3d8a803844c683b96bb0180","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 05:00:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b7b66d4d41ad1dbc3741d8801d89491ae3ae7889","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 05:01:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93938644f24149ab8d0aac6f3a1cf27a\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b58778b1bf7e416a9374b6d5fcafd91b : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"717436ecab47ce2a21d492775229b944a202e3c0","revisions":{"659d68ea2fecad263025bca058561c5e40d85201":{"kind":"REWORK","_number":1,"created":"2022-07-09 14:59:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/01/849201/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/01/849201/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/01/849201/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/01/849201/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/01/849201/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/01/849201/1"}}},"commit":{"parents":[{"commit":"7a7c469b85edfdb2a4193e13c85a2d0a13603c53","subject":"Merge \"Fix typos in parameter descriptions and tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/7a7c469b85edfdb2a4193e13c85a2d0a13603c53"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 14:58:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 14:58:14.000000000","tz":540},"subject":"Remove unused resrouce type to manage ml2_conf_sriov.ini","message":"Remove unused resrouce type to manage ml2_conf_sriov.ini\n\nThe ml2_conf_sriov.ini file was removed during Ocata cycle by [1], and\nnow no additional file is required by neutron-server. This change\nremoves the resource type because it is not used in any manifest files.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/413439\n\nChange-Id: If959e3751f881107bf5e697f044df99406677bd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/659d68ea2fecad263025bca058561c5e40d85201"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/659d68ea2fecad263025bca058561c5e40d85201"}]},"branch":"refs/heads/master"},"717436ecab47ce2a21d492775229b944a202e3c0":{"kind":"REWORK","_number":2,"created":"2022-08-01 09:43:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/01/849201/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/01/849201/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/01/849201/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/01/849201/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/01/849201/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/01/849201/2"}}},"commit":{"parents":[{"commit":"43558584cdd4493777af6d4bbfb81e413a9765c3","subject":"Merge \"Rename acceptance spec files to explain order\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/43558584cdd4493777af6d4bbfb81e413a9765c3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 14:58:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-01 09:43:22.000000000","tz":540},"subject":"Remove unused resrouce type to manage ml2_conf_sriov.ini","message":"Remove unused resrouce type to manage ml2_conf_sriov.ini\n\nThe ml2_conf_sriov.ini file was removed during Ocata cycle by [1], and\nnow no additional file is required by neutron-server. This change\nremoves the resource type because it is not used in any manifest files.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/413439\n\nChange-Id: If959e3751f881107bf5e697f044df99406677bd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/717436ecab47ce2a21d492775229b944a202e3c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/717436ecab47ce2a21d492775229b944a202e3c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[]}
