)]}'
{"id":"openstack%2Fpuppet-neutron~849462","triplet_id":"openstack%2Fpuppet-neutron~stable%2Fwallaby~Ie0f7d0029412a16e4d9d43f408be97794b3518c0","project":"openstack/puppet-neutron","branch":"stable/wallaby","topic":"ovn-tunables","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-07-13 04:58:33.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-07-13 01:42:36.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"6681":{"account":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"last_update":"2022-07-13 09:51:43.000000000","reason":"Change was submitted"},"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2022-07-12 21:23:24.000000000","reason":"removed on reply"},"28223":{"account":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"last_update":"2022-07-12 14:31:39.000000000","reason":"\u003cGERRIT_ACCOUNT_28223\u003e replied on the change","reason_account":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}}},"hashtags":[],"change_id":"Ie0f7d0029412a16e4d9d43f408be97794b3518c0","subject":"[ovn metadata] Add more tunable parameters","status":"MERGED","created":"2022-07-12 08:01:49.000000000","updated":"2022-07-13 09:52:43.000000000","submitted":"2022-07-13 09:51:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849462-ovn-tunables","meta_rev_id":"4061324abadc001560df603b12d2189031e23b58","_number":849462,"virtual_id_number":849462,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-13 09:51:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-07-13 01:42:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2022-07-12 21:23:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-12 14:31:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2022-07-13 01:42:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"}],"reviewers":{"REVIEWER":[{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-12 10:40:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-12 14:29:19.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2022-07-12 14:29:19.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-07-12 14:29:19.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"state":"REVIEWER"},{"updated":"2022-07-12 14:29:19.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"e07b0c2778120e09f5127fdfd927faf5e76d0201","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-07-12 08:01:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"561080f55b09d13834428a0bd5e88a5768956465","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 10:40:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb6ab13c4db84a5bb21f7fa9e74273e5\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f108f6897d73436ea6191ec472e2b52e : SUCCESS in 5m 28s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41d917b329b141cab6b0e6cf15a6e2b8 : SUCCESS in 5m 21s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/81ba27fc16be44e3bfc7405550f68f64 : SUCCESS in 27m 52s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/262c79d2316d43a3aea7740f7c459255 : SUCCESS in 24m 06s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/30ad6d4effb440cb925104b6bb58c15b : SUCCESS in 23m 34s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1276b66f4d14c368304f167f2a63f62 : SUCCESS in 57m 58s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96abb875f88e4019b5a5b8bd6849508c : FAILURE in 1h 17m 20s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2130fbeeb2845eea6701d0eee8463dd : SUCCESS in 51m 55s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c91efea65dee4f16b1c0c2cc2772e649 : SUCCESS in 45m 43s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4268536f66254a998fd51323d3d455c9 : SUCCESS in 1h 29m 13s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b4cb93df25a14b49968b0ae6bf69541d : SUCCESS in 1h 37m 42s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2e4d2e64ce574b719321152e50f48f1f : SUCCESS in 1h 32m 53s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/566e20309df74fdba149b3e6cce5fc68 : SUCCESS in 1h 10m 06s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/3a0bd934a87b4f48986027c05f3b63cf : SUCCESS in 32m 29s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea02c34f054d4e96b9361b03080e935b : SUCCESS in 26m 03s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/0f33fdc02d274725bc3040fc0ef1fcbf : SUCCESS in 2h 33m 22s\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/a977f8d9493943828c242b65aa4151e7 : SUCCESS in 2h 14m 15s (non-voting)\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/ca91b6ce28544971b238c03d69d83c13 : SUCCESS in 1h 00m 18s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9044c659d4c64ab7b711b955ab9a6258 : SUCCESS in 6m 03s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d9fecaa3184747ac98ca3405ab406f05 : SUCCESS in 1h 10m 04s\n- puppet-neutron-tripleo-centos-9-standalone https://zuul.opendev.org/t/openstack/build/ec850a76a4624dd38baa8345e2a9b617 : SUCCESS in 1h 06m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fb5047f28e21fed24f046db8ea00b79952622948","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2022-07-12 14:31:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6db39e64064d1c65140972f6e70009955af981b3","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-07-12 21:23:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"afec7619d96f6cf206a94111c053092b28c8415b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-13 01:42:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ff4ba9700b08cb5c8dcb4e844619c9f6448bbdb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 01:43:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"feff77c9063722a91f07beafd7fb63e1371117d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 04:21:34.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca477e5e62e24148aef665e2ec42b7d1\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f924decb8c7643f0b20af3313742c4ca : SUCCESS in 5m 50s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/808ea888095f4cf7865f941914f65396 : SUCCESS in 5m 33s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/3af9422f00544fd5831511063912a9ba : SUCCESS in 27m 39s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b242668a85634054a5e72c0c1b370ea4 : SUCCESS in 23m 07s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/93714e9007f44689ac28d7e3e7151bd9 : SUCCESS in 1h 27m 43s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/222218d43fb242be8cc8aaab3812786b : FAILURE in 1h 38m 35s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/db179cf5e06a4adbb4b2e6d2daa7951c : SUCCESS in 1h 34m 41s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a3427a9a789d41ad811cb37d789cc226 : SUCCESS in 1h 12m 57s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/6c03bfcb33a74c1db165f964eae54dda : SUCCESS in 34m 22s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/129f6e8d694744789b77f59d64a7f3da : SUCCESS in 26m 27s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/8e2be40a47c741a6b425cc961bdf8fdf : SUCCESS in 2h 28m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec2e32e96f544633975d6f3aff197904 : SUCCESS in 6m 21s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f0758322d17f4db2931161a4b744de03 : SUCCESS in 1h 08m 50s","accounts_in_message":[],"_revision_number":1},{"id":"0abdfbc4cb4586355b6a4ea8010c11f3751cab6a","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-07-13 04:58:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0c5d0af0757b716c5b344d6456fe312a237da0fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 07:13:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d18699cd5acf41dc8e9fc3785530af7f\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3e9b1b16acb4e76a1b9df2baa5a71bf : SUCCESS in 4m 35s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7c5762d38f547c587f661804352cec1 : SUCCESS in 3m 39s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/3a84a7b01c1d4a59b84ade85545cb3a1 : SUCCESS in 21m 34s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e11f5d6a682b4205a21a8fad3262673a : SUCCESS in 13m 13s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c0bd7ac76cb41ab97cac4b39cafee17 : SUCCESS in 15m 23s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4456c151e479404a8082429804bc856e : SUCCESS in 51m 26s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e15ab3666ec54a0aa4ed6e0acfa86b6d : FAILURE in 1h 06m 08s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7535d92fe534ef6b813349428a44632 : SUCCESS in 40m 50s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ae26780bee14836aeb9d62aea414e3d : SUCCESS in 36m 13s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e8d6a639c9f3477486ccc79874076116 : SUCCESS in 1h 15m 04s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/adabcc2280384404892536c7defa9592 : SUCCESS in 1h 15m 29s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ccbb32af372f4c03a27bf296db1caad0 : SUCCESS in 1h 08m 43s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3838c76e34344d6aa728af34e8530a87 : SUCCESS in 58m 24s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/4c70e01d652243c1a18b3b54281d7b78 : SUCCESS in 27m 58s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a8c0c8231624d5ba5784b655425bd9a : SUCCESS in 19m 00s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/51930053e5af44bf9ee8f9cc21c1cb5b : SUCCESS in 2h 12m 49s\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/276fa25fdccf4059b88ec2644fe6a8d8 : SUCCESS in 2h 04m 30s (non-voting)\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/a58daa21b19e44abb7896f4719492d40 : SUCCESS in 59m 16s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b8f544b1a15460ba302d1609e65f450 : SUCCESS in 4m 04s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/19962bfa42c140fc9fc001cfe30b9287 : SUCCESS in 1h 03m 18s\n- puppet-neutron-tripleo-centos-9-standalone https://zuul.opendev.org/t/openstack/build/eff91b15e4de4576b7b7d512917a45ae : SUCCESS in 1h 05m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8f840225cb91c624e2dd705ba1c7191508b0ea39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 07:15:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"62a90915c9654bdbd413d724c989976ea7282e14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 09:51:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/568e35349fca42d08092f86e7360d7af\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c8e26223f4c46dba0ac03c8558c4fcb : SUCCESS in 3m 34s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4607125b3265495eb05937230c4e1cfc : SUCCESS in 3m 16s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/6a7e07b6cca84693a4f8526541fcf1f5 : SUCCESS in 27m 25s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/879c651e0f464c6687344fd99e3eda1f : SUCCESS in 14m 45s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e6bfa3644e3645d99f1849d2467e60c8 : SUCCESS in 1h 41m 25s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6c1f9a81e3d74bcbb9182c0513481590 : SUCCESS in 1h 35m 00s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1bc19ce3ad69435fa0364d8fb015b980 : SUCCESS in 1h 15m 51s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e7daebd903984f36a315c5a1fe169155 : SUCCESS in 1h 08m 57s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/b9f727bfd5bb4779a36b59d65390a339 : SUCCESS in 35m 19s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49b1abc283304d84a00faecec3d60181 : SUCCESS in 19m 06s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/f2aa42b0fba74e4aac5b158a63cf1858 : SUCCESS in 2h 36m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35e60f04b608439fbc10d9cb31d817df : SUCCESS in 6m 31s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5ce89d13d22647b7a9931b93148940fe : SUCCESS in 1h 18m 51s","accounts_in_message":[],"_revision_number":1},{"id":"d504cbd4c1bdebfee7212f9328e7249671dff83a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 09:51:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4061324abadc001560df603b12d2189031e23b58","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 09:52:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/779b615158054e35b8ddff0d88d1de77\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e6306503fb9436f8679b56a23baa9a7 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9a9bdac5ac863da6a43c09e693057c9a8b59f21f","revisions":{"9a9bdac5ac863da6a43c09e693057c9a8b59f21f":{"kind":"REWORK","_number":1,"created":"2022-07-12 08:01:49.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/62/849462/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/62/849462/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/62/849462/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/62/849462/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/62/849462/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/62/849462/1"}}},"commit":{"parents":[{"commit":"4456238f978d37dbdac8769d3d3db89ed56ef418","subject":"ml2+ovn: Add more tunable parameters for OVN DB connections","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/4456238f978d37dbdac8769d3d3db89ed56ef418"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-07-08 15:21:15.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-07-12 08:01:28.000000000","tz":330},"subject":"[ovn metadata] Add more tunable parameters","message":"[ovn metadata] Add more tunable parameters\n\nFor ml2 conf these are already added as part of [1], these\nparameters are also needed in ovn metadata agent config\nfile, this patch takes care of it.\n\n[1] https://review.opendev.org/q/I9a93b65aa5dbf4bca934aec82aad2f7f6ab3d73c\n\nChange-Id: Ie0f7d0029412a16e4d9d43f408be97794b3518c0\n(cherry picked from commit 68a4d9a53df2e96a7c8cba361fabe03fa908f266)\n(cherry picked from commit 53fd3e801c98a25d0b228aa138c643a3c3f62d3f)\nConflicts: spec/classes/neutron_agents_ovn_metadata_spec.rb\n(cherry picked from commit 3677b7aa1d2e2fea10828918c6b07d82e0f9dca9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/9a9bdac5ac863da6a43c09e693057c9a8b59f21f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/9a9bdac5ac863da6a43c09e693057c9a8b59f21f"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[]}
