)]}'
{"id":"openstack%2Fpuppet-neutron~851846","triplet_id":"openstack%2Fpuppet-neutron~master~I91fc9238dd5209bf22381a1171b9f580a9a4e7f9","project":"openstack/puppet-neutron","branch":"master","topic":"bug/1983300","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-11 16:30:28.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I91fc9238dd5209bf22381a1171b9f580a9a4e7f9","subject":"Adapt to new type validation in puppetlabs-apache","status":"MERGED","created":"2022-08-02 11:38:49.000000000","updated":"2022-08-11 19:15:37.000000000","submitted":"2022-08-11 19:15:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851846-bug/1983300","meta_rev_id":"4689418ddf7c395303d988a548a6eb0e44c6450d","_number":851846,"virtual_id_number":851846,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-11 19:15:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-08-11 16:30:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2022-08-08 15:19:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2022-08-11 16:30:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-02 13:57:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-08 15:19:19.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"f31f3236d0d14c91bd141fc72e6942b1f4ad63eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-02 11:38:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e2a7fb202195d0a5f69b77acc0301d09e3a6bff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 13:57:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c8283be4b6f47fca5ace699aef3cd48\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/efec890420014417967bc778ea4f47bf : SUCCESS in 5m 38s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9f27d0f48cf466983c2f9cbd5a7efef : SUCCESS in 3m 38s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f6cd94fe73449208ea17a0ea1a4733a : SUCCESS in 3m 56s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/9d8912799faa4847a40caaf0258bd3ca : SUCCESS in 22m 36s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62b4d38fcf3a447887fde76565d201b8 : SUCCESS in 16m 11s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84401f22425b4926b393cc6b64099d92 : SUCCESS in 15m 32s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c57e74d832f54e7e9634ff2fda005b33 : SUCCESS in 15m 09s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/afd929c8eb044329a88ff732e1f72baf : FAILURE in 44m 29s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8482396d50be4de5836679bb1146a1f1 : FAILURE in 53m 55s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51a8eee0bb8a42579d9b93498c0d2fbf : FAILURE in 37m 29s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94aad8038552418282b0e4a501faa519 : FAILURE in 33m 44s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db73283f492f4f37957707161e4343df : FAILURE in 32m 40s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/30b1dfe27c3f4a58910b3369383e1d36 : SUCCESS in 25m 51s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc3350acde3c459ab8cfce6e9b5cef70 : SUCCESS in 1h 05m 42s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/61dac4a72c784add8a017a0cf5217a1e : SUCCESS in 1h 17m 59s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/743a024c55cf4551bc01dbf3c48972c5 : SUCCESS in 1h 08m 50s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a1134eb798f74d9fad0f930d3abaa775 : SUCCESS in 50m 25s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ba412db61b05474cbaf8d082bc279bc0 : SUCCESS in 49m 32s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16350cf4f7dd49d083796e45b1ee2275 : FAILURE in 24m 18s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc08072f570046ea9fcf07967586f8d1 : SUCCESS in 29m 10s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b64924068b54fe8b8cc5c2bfd70fc86 : FAILURE in 16m 44s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95b1c50f8eba4dd0afb5454b9ec1caeb : FAILURE in 18m 59s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/6e36c5fb32e34c1ca5fe8ffdd2040bc2 : SUCCESS in 2h 13m 23s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/dcc79095dacc47618771f3a9173f694f : SUCCESS in 42m 14s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5e47a2a09b4043d88b936a391cb54298 : SUCCESS in 1h 06m 04s","accounts_in_message":[],"_revision_number":1},{"id":"d8e0a2e2508020cb78ce716a51d72fdd784da907","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-08-08 15:19:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5aa92e1e07e1fe8fc0ef76dff68182f1835b4f15","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-11 16:30:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ab4a4d61b8eaee60bb11564980b6e6304279a56e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 16:31:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"38a255d3f8220d3988e795e16254909c40da16fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 19:15:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1b45d886cdb4754a1fc0c84cc76a5d1\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8bb95fed36454e88a231692fc8c82411 : SUCCESS in 3m 30s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ae4928da6e447b39b03f86bb954edb6 : SUCCESS in 3m 18s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2dbc06b8807a43d4a0f17ea2fe8a79ab : SUCCESS in 5m 55s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/1d37e9548d634007b82c52362b74a8e2 : SUCCESS in 27m 24s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5abbca714f8643999cdf46526eb5c161 : SUCCESS in 23m 33s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae85d706043b485692857d09cf4a58b2 : SUCCESS in 16m 32s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/385a24d4f8334be6a52ff75b5ad10f9b : SUCCESS in 1h 00m 04s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/62504b7a046345cfba8054ac54671fdc : SUCCESS in 1h 10m 56s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/265cafaadfb34020b10831613736dc1e : SUCCESS in 1h 06m 52s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4c9f66e2538f482c84e0d5ffde76b2a6 : SUCCESS in 56m 22s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c94346e744ac4e13b5889d2946c63c7c : SUCCESS in 51m 42s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/4d5611dc1e5742929e815af5e6a9056b : SUCCESS in 25m 03s\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/61263f1db9064c1fb0efc6d03e634940 : SUCCESS in 2h 38m 52s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/e7b1271ebf2042fb9a37068b829d9cbb : SUCCESS in 54m 29s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5d2846eb7fef48a7abd1be0e1c646c65 : SUCCESS in 1h 21m 09s","accounts_in_message":[],"_revision_number":1},{"id":"4689418ddf7c395303d988a548a6eb0e44c6450d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 19:15:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"386dd3c488194dfba7862b1d57f70f152eaaa284","revisions":{"386dd3c488194dfba7862b1d57f70f152eaaa284":{"kind":"REWORK","_number":1,"created":"2022-08-02 11:38:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/46/851846/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/46/851846/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/46/851846/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/46/851846/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/46/851846/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/46/851846/1"}}},"commit":{"parents":[{"commit":"bb522e0fc66b84bb3099419e1ecb51b291956692","subject":"Merge \"Deprecate support for NSX plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/bb522e0fc66b84bb3099419e1ecb51b291956692"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-02 11:38:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-02 11:38:38.000000000","tz":540},"subject":"Adapt to new type validation in puppetlabs-apache","message":"Adapt to new type validation in puppetlabs-apache\n\nThe puppetlabs-apache module is enforcing more strict data type\nvalidation[1].\n\nThis change updates the default values to adapt to that change.\n\n[1] https://github.com/puppetlabs/puppetlabs-apache/commit/f41251e336fa3e7c31c19968ccee74121cf71e47\n\nCloses-Bug: #1983300\nDepends-on: https://review.opendev.org/851652\nChange-Id: I91fc9238dd5209bf22381a1171b9f580a9a4e7f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/386dd3c488194dfba7862b1d57f70f152eaaa284"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/386dd3c488194dfba7862b1d57f70f152eaaa284"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[]}
