)]}'
{"id":"openstack%2Fpuppet-neutron~853883","triplet_id":"openstack%2Fpuppet-neutron~master~Iaba23fe52679e92017fb1a489635c8b2d95334b3","project":"openstack/puppet-neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-22 18:34:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iaba23fe52679e92017fb1a489635c8b2d95334b3","subject":"Do not create /etc/default/neutron-server in CentOS/RHEL","status":"MERGED","created":"2022-08-21 17:47:26.000000000","updated":"2022-08-22 18:34:41.000000000","submitted":"2022-08-22 18:34:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853883","meta_rev_id":"d331dafa9f879b73aa930d7e9d814198ac4a25c5","_number":853883,"virtual_id_number":853883,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-22 18:34:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-08-22 06:48:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-08-22 15:40:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-08-22 15:40:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-21 21:18:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-22 06:48:42.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-08-22 15:40:22.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"57437fedde6a75971dcfbab7a05b3c0ad13e1540","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-21 17:47:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"967d09501b86aec6176060eef8c0b51814444b9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-21 18:42:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cb769832724f07881a7a9a2bbb85fc8e3aa7d37d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-21 21:18:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/37048125bf2044ccae25beb787cd690e\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2df0ab18442146ca923fcf6e18bc502c : SUCCESS in 3m 39s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e4dc7f40ba343439ab30391f41f38e2 : SUCCESS in 3m 38s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e78a5b2e6214243a35975541d827ccc : SUCCESS in 3m 36s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/3fe7a0d6d8664e2aa235b7a464c5dd14 : FAILURE in 19m 48s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ea64b1c90c74c47855067c3fed84438 : FAILURE in 16m 21s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d62070f7a1344515b8f86237740efe27 : FAILURE in 16m 55s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/942ac396be734209a5cf2ec141eaab38 : FAILURE in 16m 51s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a56ac61f036a495ab3194ba5b9c2e584 : FAILURE in 51m 45s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bcb4e8d87bc54c06931f6de8bd8b3693 : FAILURE in 1h 02m 00s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3d9a4a9cde2471f979d6a4d284bc870 : FAILURE in 43m 35s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b1f6bbef852d405faa405b23544df499 : FAILURE in 38m 48s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/548a1f689ccc4274a1248df0353b1810 : FAILURE in 40m 41s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e6a7190f75b743e9891fce822bd4a77c : SUCCESS in 23m 38s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/41d45f95ab6b4bcab1e2d405e8323c83 : SUCCESS in 1h 04m 31s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/aab912aba5b141d6985ae3fd469fe842 : SUCCESS in 1h 08m 30s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e6ae0aac59fc4b8bb7cc5aa10e570a22 : SUCCESS in 1h 14m 36s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7f65f01916014b5eb88bd2c9519c1e6f : SUCCESS in 1h 05m 20s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f89e0ad42ba44948ac298b419bcbd969 : FAILURE in 54m 41s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54b7540af878490f982537544ce80900 : SUCCESS in 19m 08s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/58c5fae54baa48ac964526b93e6a7707 : SUCCESS in 22m 27s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49c65b1e95a14b4691414ac8389482bb : SUCCESS in 18m 18s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee24dcc2cda0479a963cb2803107dbf0 : SUCCESS in 15m 11s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/d674cb4efda44770921be19882e02e89 : SUCCESS in 2h 31m 26s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/279bd5b1b313461bbf80d74377231c90 : SUCCESS in 50m 41s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d231507aebbd464891977159baa3e71d : SUCCESS in 1h 22m 41s","accounts_in_message":[],"_revision_number":2},{"id":"99c04d0b7052c9266165fea5831c4db71901608a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-21 23:54:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7be5392b0ab374975451d0c174e907e6ffbf1312","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-22 00:04:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"19dd04769804075e78866d95967f1e03395119b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-22 00:50:00.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b3a69ece2a8a7024cf544fd1a42a02635261cd2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-22 01:12:34.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"75c96d5d16731597ae5edf1155e0a7b09eff588e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-22 02:24:31.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"83f3b3b5b6a0bb6ce8246f51c6487662cb588213","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-22 02:50:40.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"986f89a4209ae1b4780b0102c4003b4b3c74f0fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 04:44:15.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b45f245437df44cfb24af3fc9c274c31\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5694c8700c234e06b3913debc8b364ad : SUCCESS in 3m 50s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f2e39f8814848f48da59c57a477388f : SUCCESS in 3m 51s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91324547acd54f4abe414f182c009deb : SUCCESS in 3m 01s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/2d27985a7a574d2284b6ff1854990687 : FAILURE in 20m 40s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07f427a8f8eb410ebbc191880bef2755 : FAILURE in 16m 59s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca3f437d5775458bbd5474d1925dc279 : FAILURE in 16m 19s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d79ec496275d4493951c00f5a2127dbd : FAILURE in 15m 36s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/081168c24e8247d88c3ef59789ff1d87 : FAILURE in 45m 44s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6756f9ad3554a3c84c211ef377e6db2 : FAILURE in 59m 14s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c51e5cf7a464f58b460f11d6b089aff : FAILURE in 49m 48s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1cd9d7c51fbd4975a49d25db2570e298 : FAILURE in 37m 47s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a3e090c69d248b59f51bd2dbce6f76d : FAILURE in 46m 03s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/abbaf56545574a97830c44a2463bd1f4 : SUCCESS in 23m 21s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc6a5cc4f12144f1ae4ba331dc4798af : SUCCESS in 1h 02m 13s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3d0fe2c0499142f38cec0a392f702136 : SUCCESS in 1h 26m 04s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6f2566ad3d7348deb433a0e9590820cf : SUCCESS in 1h 04m 29s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5226b6d5ff8a450593069fc7c9ab4b71 : SUCCESS in 53m 26s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7080ecbb5c2f4545afe47706e0b675a0 : SUCCESS in 57m 16s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45b7861998b045a7a3d6e2e9a73a68c2 : SUCCESS in 17m 59s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/af1499f640e34e9a9a85cdcb52ed2727 : SUCCESS in 22m 43s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd90946762d8437d855eada90587fc09 : SUCCESS in 16m 45s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a24be583d1c74eb0b95ff32f1f6eebcd : SUCCESS in 16m 59s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/ab6b72f5b1b740ed9ed50efa1f5f9788 : SUCCESS in 1h 49m 58s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/41b08f8e75514c4baa75fea65c3fe9ad : SUCCESS in 35m 11s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/04b9900d0a994177ac45b57aeb44f768 : SUCCESS in 52m 15s","accounts_in_message":[],"_revision_number":8},{"id":"7269be5265a962942cb1580560b0189eb6a1dd83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-22 04:49:56.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"b5b91ff7dbf40df03afed2f3634793918b3e79de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 06:26:13.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e11364121204fb6bfdbc8461e3078c9\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/303c2b04a2244a708bf925c67a08e71d : SUCCESS in 3m 30s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8969bb1811da41ba9a41b32317e449ad : SUCCESS in 5m 14s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c091c9448d343ac8b3d0d51ebbdc7be : SUCCESS in 3m 19s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/75922b2dbd0a495ea5e25085cfc5dafb : SUCCESS in 21m 10s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e8f63895b5d4e408bee6ce05a4c8b72 : SUCCESS in 13m 41s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8db5890d7c446f49d99e3dd3abcd1aa : SUCCESS in 22m 01s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7da1c434cf8c4cd19664c69e8f9c88ae : SUCCESS in 12m 47s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4063b1ea94424eff839bb352e0ab01fc : FAILURE in 40m 03s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24b74abf56934512b6cf7ace5ca06e7d : FAILURE in 48m 44s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e2960fbce8846459ef423a50fe663d8 : FAILURE in 42m 48s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08f09bff2d7e4bf6b6246c72dc3e193b : FAILURE in 35m 34s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/670defc0805c4854a6750115a77e1bf4 : FAILURE in 35m 26s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/96b68cf8d264445dbf8e33b1cf854612 : SUCCESS in 25m 48s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/250db4eaac094306938f96b5baaec0cc : SUCCESS in 1h 07m 43s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/63af8771ca4b4e5e89c593d8d5926c1e : SUCCESS in 1h 16m 01s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cd0bde83f4c64afdb7a59fce2ed6180f : SUCCESS in 1h 07m 45s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/db0ccee7bdec4fc8b958e9ab4bb2c3aa : SUCCESS in 57m 18s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/012e19d06f8945cdaf4311ed77327adb : SUCCESS in 52m 47s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b758864420a04a19b02c8636d00d0400 : SUCCESS in 19m 17s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/3d0a90c38b9440eea70b1935117b668f : SUCCESS in 25m 39s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/893141e74aab4ebfae693d55d8ef325f : SUCCESS in 19m 03s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3cc06c0b510449f6b823dffb0dd85fb5 : SUCCESS in 16m 38s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/4790c49c454f4c58a35b58e215a34a11 : SUCCESS in 1h 34m 15s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/b98808f1f3e54741b988ec42f91c4239 : SUCCESS in 34m 47s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/921276b9ad2a4cdc874ff1f0cd6cf2f7 : SUCCESS in 53m 18s","accounts_in_message":[],"_revision_number":9},{"id":"97ce096ac63bcf1a84af5c999b9f5e9eb8518c47","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-08-22 06:48:42.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"cf50145b05fbfa7a734986c13f0213e5f4abe3f9","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-08-22 15:40:22.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"bd08e9939206bb822b76674552e7c39762163a33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 15:41:57.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"30096abe20ba44a27ebcc0d41a3f049bd08cde84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 18:34:41.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/976c66aa2bf947468d4b41aa71e3cbc4\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0cd140bdef5a433397d2f248e2394040 : SUCCESS in 4m 34s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/acc15a5e661449f68220f1aec22e2f57 : SUCCESS in 4m 15s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0c92cbe8b2646f8893ef2fe971be0ce : SUCCESS in 4m 01s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/46b8e21d1a3042438e715dfb80d2ddbc : SUCCESS in 22m 14s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b97bcca04fe649348ba5b4ea91ccaef5 : SUCCESS in 16m 59s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/014c2947c35747d6af5e09f44a5d09f4 : SUCCESS in 12m 45s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/917f22f87aa64929a5b2c04ab212cddb : SUCCESS in 1h 01m 42s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/df6b89fd22364c6c8e0c54769f23d9a6 : SUCCESS in 1h 11m 46s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f1c483f3c98e44dab352f05b676ac175 : SUCCESS in 1h 14m 26s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/66e5e6e4786e4b8393a706a3f65cc97c : SUCCESS in 57m 52s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9c8467a5a70b48ad815d90b9920ec83a : SUCCESS in 59m 14s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/fd547be73f204867b6e662feb6d83028 : SUCCESS in 21m 57s\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/df4151bde48a4015a78ecd333f9edbcc : SUCCESS in 2h 48m 27s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/d5757dc4be914dc79f5586c006332092 : SUCCESS in 52m 14s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/df5b35ee38b24479a060d81c922b85c1 : SUCCESS in 1h 27m 17s","accounts_in_message":[],"_revision_number":9},{"id":"d331dafa9f879b73aa930d7e9d814198ac4a25c5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 18:34:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"6aa4204663664cdd1ad8197503603a29f66a5200","revisions":{"352083df7834ac959ebde0bc970ab985a6826b47":{"kind":"REWORK","_number":1,"created":"2022-08-21 17:47:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/853883/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/83/853883/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/83/853883/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/83/853883/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/83/853883/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/83/853883/1"}}},"commit":{"parents":[{"commit":"581f6a31496ffc23ae2fc9f06ce5e41957bd13dc","subject":"Merge \"Remove unused items from platform_params\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/581f6a31496ffc23ae2fc9f06ce5e41957bd13dc"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-21 17:46:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-21 17:46:57.000000000","tz":540},"subject":"Create /etc/default/neutron-server only in Ubuntu","message":"Create /etc/default/neutron-server only in Ubuntu\n\n... because the file is not used in the other operating systems.\n\nChange-Id: Iaba23fe52679e92017fb1a489635c8b2d95334b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/352083df7834ac959ebde0bc970ab985a6826b47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/352083df7834ac959ebde0bc970ab985a6826b47"}]},"branch":"refs/heads/master"},"83a1b14fb3686251bb5b8ace1a15070435f1835a":{"kind":"REWORK","_number":2,"created":"2022-08-21 18:42:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/853883/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/83/853883/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/83/853883/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/83/853883/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/83/853883/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/83/853883/2"}}},"commit":{"parents":[{"commit":"581f6a31496ffc23ae2fc9f06ce5e41957bd13dc","subject":"Merge \"Remove unused items from platform_params\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/581f6a31496ffc23ae2fc9f06ce5e41957bd13dc"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-21 17:46:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-21 18:41:38.000000000","tz":540},"subject":"Do not create /etc/default/neutron-server in CentOS/RHEL","message":"Do not create /etc/default/neutron-server in CentOS/RHEL\n\n... because the file is not used in these operating systems.\n\nChange-Id: Iaba23fe52679e92017fb1a489635c8b2d95334b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/83a1b14fb3686251bb5b8ace1a15070435f1835a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/83a1b14fb3686251bb5b8ace1a15070435f1835a"}]},"branch":"refs/heads/master"},"83d685a57bf477d2dd07ea61aad9db3d3c380638":{"kind":"REWORK","_number":3,"created":"2022-08-21 23:54:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/853883/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/83/853883/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/83/853883/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/83/853883/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/83/853883/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/83/853883/3"}}},"commit":{"parents":[{"commit":"581f6a31496ffc23ae2fc9f06ce5e41957bd13dc","subject":"Merge \"Remove unused items from platform_params\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/581f6a31496ffc23ae2fc9f06ce5e41957bd13dc"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-21 17:46:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-21 23:53:56.000000000","tz":540},"subject":"Do not create /etc/default/neutron-server in CentOS/RHEL","message":"Do not create /etc/default/neutron-server in CentOS/RHEL\n\n... because the file is not used in these operating systems.\n\nChange-Id: Iaba23fe52679e92017fb1a489635c8b2d95334b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/83d685a57bf477d2dd07ea61aad9db3d3c380638"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/83d685a57bf477d2dd07ea61aad9db3d3c380638"}]},"branch":"refs/heads/master"},"b876b7ad43d0603fe683e16b1dbe1bdd5c50a67b":{"kind":"REWORK","_number":4,"created":"2022-08-22 00:04:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/853883/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/83/853883/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/83/853883/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/83/853883/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/83/853883/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/83/853883/4"}}},"commit":{"parents":[{"commit":"581f6a31496ffc23ae2fc9f06ce5e41957bd13dc","subject":"Merge \"Remove unused items from platform_params\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/581f6a31496ffc23ae2fc9f06ce5e41957bd13dc"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-21 17:46:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-22 00:04:34.000000000","tz":540},"subject":"Do not create /etc/default/neutron-server in CentOS/RHEL","message":"Do not create /etc/default/neutron-server in CentOS/RHEL\n\n... because the file is not used in these operating systems.\n\nChange-Id: Iaba23fe52679e92017fb1a489635c8b2d95334b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b876b7ad43d0603fe683e16b1dbe1bdd5c50a67b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b876b7ad43d0603fe683e16b1dbe1bdd5c50a67b"}]},"branch":"refs/heads/master"},"da3da2b49db7338f926b74cd8a3bb9527c5d58cc":{"kind":"REWORK","_number":5,"created":"2022-08-22 00:50:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/853883/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/83/853883/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/83/853883/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/83/853883/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/83/853883/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/83/853883/5"}}},"commit":{"parents":[{"commit":"581f6a31496ffc23ae2fc9f06ce5e41957bd13dc","subject":"Merge \"Remove unused items from platform_params\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/581f6a31496ffc23ae2fc9f06ce5e41957bd13dc"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-21 17:46:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-22 00:49:51.000000000","tz":540},"subject":"Do not create /etc/default/neutron-server in CentOS/RHEL","message":"Do not create /etc/default/neutron-server in CentOS/RHEL\n\n... because the file is not used in these operating systems.\n\nChange-Id: Iaba23fe52679e92017fb1a489635c8b2d95334b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/da3da2b49db7338f926b74cd8a3bb9527c5d58cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/da3da2b49db7338f926b74cd8a3bb9527c5d58cc"}]},"branch":"refs/heads/master"},"a34393a745b9040521fa69649282ae2ed517cae4":{"kind":"REWORK","_number":6,"created":"2022-08-22 01:12:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/853883/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/83/853883/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/83/853883/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/83/853883/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/83/853883/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/83/853883/6"}}},"commit":{"parents":[{"commit":"581f6a31496ffc23ae2fc9f06ce5e41957bd13dc","subject":"Merge \"Remove unused items from platform_params\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/581f6a31496ffc23ae2fc9f06ce5e41957bd13dc"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-21 17:46:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-22 01:12:26.000000000","tz":540},"subject":"Do not create /etc/default/neutron-server in CentOS/RHEL","message":"Do not create /etc/default/neutron-server in CentOS/RHEL\n\n... because the file is not used in these operating systems.\n\nChange-Id: Iaba23fe52679e92017fb1a489635c8b2d95334b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/a34393a745b9040521fa69649282ae2ed517cae4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/a34393a745b9040521fa69649282ae2ed517cae4"}]},"branch":"refs/heads/master"},"a02ef72d29133b7013c161315661c9216a6aef69":{"kind":"REWORK","_number":7,"created":"2022-08-22 02:24:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/853883/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/83/853883/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/83/853883/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/83/853883/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/83/853883/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/83/853883/7"}}},"commit":{"parents":[{"commit":"581f6a31496ffc23ae2fc9f06ce5e41957bd13dc","subject":"Merge \"Remove unused items from platform_params\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/581f6a31496ffc23ae2fc9f06ce5e41957bd13dc"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-21 17:46:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-22 02:24:23.000000000","tz":540},"subject":"Do not create /etc/default/neutron-server in CentOS/RHEL","message":"Do not create /etc/default/neutron-server in CentOS/RHEL\n\n... because the file is not used in these operating systems.\n\nChange-Id: Iaba23fe52679e92017fb1a489635c8b2d95334b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/a02ef72d29133b7013c161315661c9216a6aef69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/a02ef72d29133b7013c161315661c9216a6aef69"}]},"branch":"refs/heads/master"},"655daf20fe7b87a142a1caa3cb6e910e2596fda4":{"kind":"REWORK","_number":8,"created":"2022-08-22 02:50:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/853883/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/83/853883/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/83/853883/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/83/853883/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/83/853883/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/83/853883/8"}}},"commit":{"parents":[{"commit":"581f6a31496ffc23ae2fc9f06ce5e41957bd13dc","subject":"Merge \"Remove unused items from platform_params\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/581f6a31496ffc23ae2fc9f06ce5e41957bd13dc"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-21 17:46:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-22 02:50:25.000000000","tz":540},"subject":"Do not create /etc/default/neutron-server in CentOS/RHEL","message":"Do not create /etc/default/neutron-server in CentOS/RHEL\n\n... because the file is not used in these operating systems.\n\nChange-Id: Iaba23fe52679e92017fb1a489635c8b2d95334b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/655daf20fe7b87a142a1caa3cb6e910e2596fda4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/655daf20fe7b87a142a1caa3cb6e910e2596fda4"}]},"branch":"refs/heads/master"},"6aa4204663664cdd1ad8197503603a29f66a5200":{"kind":"REWORK","_number":9,"created":"2022-08-22 04:49:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/853883/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/83/853883/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/83/853883/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/83/853883/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/83/853883/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/83/853883/9"}}},"commit":{"parents":[{"commit":"581f6a31496ffc23ae2fc9f06ce5e41957bd13dc","subject":"Merge \"Remove unused items from platform_params\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/581f6a31496ffc23ae2fc9f06ce5e41957bd13dc"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-21 17:46:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-22 04:49:42.000000000","tz":540},"subject":"Do not create /etc/default/neutron-server in CentOS/RHEL","message":"Do not create /etc/default/neutron-server in CentOS/RHEL\n\n... because the file is not used in these operating systems.\n\nChange-Id: Iaba23fe52679e92017fb1a489635c8b2d95334b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6aa4204663664cdd1ad8197503603a29f66a5200"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6aa4204663664cdd1ad8197503603a29f66a5200"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[]}
