)]}'
{"id":"openstack%2Fpuppet-neutron~854749","triplet_id":"openstack%2Fpuppet-neutron~master~Ie105131302783088fac97a56607f5a6b2c9d49aa","project":"openstack/puppet-neutron","branch":"master","topic":"apache-logging","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-26 19:06:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie105131302783088fac97a56607f5a6b2c9d49aa","subject":"Add Apache WSGI logging parameters for pipe/syslog","status":"MERGED","created":"2022-08-26 07:58:20.000000000","updated":"2022-08-26 19:08:14.000000000","submitted":"2022-08-26 19:06:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854749-apache-logging","meta_rev_id":"928baec5ccead4c5106bed550e83f79c01c1885d","_number":854749,"virtual_id_number":854749,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-26 19:06:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-08-26 11:17:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-08-26 16:01:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-08-26 16:01:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-26 10:09:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-26 11:17:39.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-08-26 16:01:45.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"2bdc85bcc644f6335c5203a76d6a4150840d08c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-26 07:58:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db9cbbc872723cda79c3d9bf45aedcd41d40a514","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 10:09:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4e2f0bcedd4478e99bb6f834d3090ac\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88bd679a726e4861916d007384aaca80 : SUCCESS in 3m 50s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4bda5c5ac9f44080860fb8a3d482ea45 : SUCCESS in 3m 36s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d136715617940fbbb006f236d5524e4 : SUCCESS in 3m 41s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/a595cf72096f4240b8341fd71cb431dc : SUCCESS in 21m 18s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/faa113b0ddb244b28648f0c80dea0895 : SUCCESS in 16m 39s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff0f2c1c866d42d4aeee678bab5e2132 : SUCCESS in 15m 53s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ccadbe53cfa428ca1f303446de00b35 : SUCCESS in 17m 18s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9502202fc5fe4919bbe4d069e2919ab1 : FAILURE in 45m 22s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e3d42d275314fcaa35ad63c10c17b35 : FAILURE in 54m 33s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b2422c591a44aa8897cb18c8e51a576 : FAILURE in 42m 30s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e3c82cb731b446da522a346f88746e4 : FAILURE in 35m 07s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02767c8cdbf84ec390673660b42ec579 : FAILURE in 40m 49s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6b2a0041f201479e9c83036031d6d612 : SUCCESS in 32m 40s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ea06faaf1cba449d894c81e5ce6072bb : SUCCESS in 1h 02m 24s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8bb734c99ddb41429e3679d5f84763a7 : SUCCESS in 1h 01m 03s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a4f2bbd1282d4af9b750362ff7d1d9ca : SUCCESS in 1h 08m 05s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b06f7203b39243aab298730c9f24529c : SUCCESS in 55m 32s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3b5311114e08462290e7d23eef0c4fd0 : SUCCESS in 57m 04s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b4fd96241184475ac0559bbacf1a245 : SUCCESS in 19m 44s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/41dc4faaafa44a08b41db38cea9b0dfd : SUCCESS in 27m 24s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47fa6e8eed984ad1a04ace63aad5db4f : SUCCESS in 17m 50s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/187ce79de529488cb8cb8eb0c428b51a : SUCCESS in 19m 38s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/d2ac4a9c934247249bde7bbe069479b4 : SUCCESS in 1h 51m 32s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/c81e2013095e44e9b8008f0dba42136f : SUCCESS in 35m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e79ebac450b6413ea344725a43912e9e : SUCCESS in 4m 10s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/624144d4b4bd4309a56ae355533e439f : SUCCESS in 59m 31s","accounts_in_message":[],"_revision_number":1},{"id":"2ccca5c47ea065c1fc48ae69300c288ac28bf341","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-08-26 11:17:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4bffac061bfc12f72cd280d12ade1d84b63f7444","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-08-26 16:01:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"87e189fe08982780dc151b2daa352db9f35a622a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 16:03:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"33ad4716f41a95166f337956a1d6a14039c973b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 19:06:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d890dde3704c4a93816da220ba57e800\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e2ffae758a744a9a92f7e25ed013002 : SUCCESS in 3m 54s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39d0c7edf268468a87be6dfb3ef5aef4 : SUCCESS in 3m 10s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33527368adec40e1b0935b78b78237f9 : SUCCESS in 3m 08s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/c42ad154899042f589f1cdf8d5067f7f : SUCCESS in 17m 14s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3cc00baf9a11488b972d868442eab3e4 : SUCCESS in 13m 26s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2e7ff4a16ff4c06a29dc3ae7d98bf9f : SUCCESS in 12m 53s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/13bf27c13cab4ff697f450b1f995fa26 : SUCCESS in 58m 18s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/eb69eb3d405f457788daa9219727f321 : SUCCESS in 57m 50s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c6c7d4b87127435f88d6e8cc1366ea40 : SUCCESS in 52m 55s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/650fb0c9a8a34a35ab1c84f8a6ba444e : SUCCESS in 46m 14s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a6b27623cdfa43c98ca9b693c2df167b : SUCCESS in 54m 40s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/84d1ed5c240242d5a7c6e796a9fa706d : SUCCESS in 24m 42s\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/c955f4f88f974d7db0681c556d027a80 : SUCCESS in 2h 00m 03s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/530ab77170a94149bd642d2c0e4211de : SUCCESS in 40m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53a2cb4893e3473b8f1f78153536fa60 : SUCCESS in 3m 03s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/bbc5d61be17e47b48c46ae9c4b6971df : SUCCESS in 1h 03m 30s","accounts_in_message":[],"_revision_number":1},{"id":"9ec507104ea51d6cb8af8ee5ec72ddd86a2b4ccd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 19:06:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"928baec5ccead4c5106bed550e83f79c01c1885d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 19:08:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6990b402a2ab4f9d9f79b07f2bc0b0ac\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a2ce9214b694b49a86ffedafcd9aa7c : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e47fe3ba4db5c753ec9c27e867ccb65514a48bf6","revisions":{"e47fe3ba4db5c753ec9c27e867ccb65514a48bf6":{"kind":"REWORK","_number":1,"created":"2022-08-26 07:58:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/49/854749/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/49/854749/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/49/854749/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/49/854749/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/49/854749/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/49/854749/1"}}},"commit":{"parents":[{"commit":"581f6a31496ffc23ae2fc9f06ce5e41957bd13dc","subject":"Merge \"Remove unused items from platform_params\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/581f6a31496ffc23ae2fc9f06ce5e41957bd13dc"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-26 07:57:55.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-26 07:58:08.000000000","tz":540},"subject":"Add Apache WSGI logging parameters for pipe/syslog","message":"Add Apache WSGI logging parameters for pipe/syslog\n\nAdd parameters for advanced logging configurations in Apache to\nsupport piped logging and support for syslog (via mod_syslog\navailable in Apache \u003e\u003d 2.5.0)\n\nCo-Authored-By: Andy Botting \u003candy@andybotting.com\u003e\nChange-Id: Ie105131302783088fac97a56607f5a6b2c9d49aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/e47fe3ba4db5c753ec9c27e867ccb65514a48bf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/e47fe3ba4db5c753ec9c27e867ccb65514a48bf6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[]}
