)]}'
{"id":"openstack%2Fpuppet-neutron~854894","triplet_id":"openstack%2Fpuppet-neutron~stable%2Fxena~I993523dc7f0ec8ad90a50d049dc0f5ccf9f2b191","project":"openstack/puppet-neutron","branch":"stable/xena","topic":"bug/1987460-stable/xena","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-29 18:43:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I993523dc7f0ec8ad90a50d049dc0f5ccf9f2b191","subject":"mlnx: Allow workaround for duplicate resources","status":"MERGED","created":"2022-08-28 03:38:44.000000000","updated":"2022-08-29 18:44:49.000000000","submitted":"2022-08-29 18:43:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854894-bug/1987460-stable/xena","cherry_pick_of_change":854809,"cherry_pick_of_patch_set":1,"meta_rev_id":"cacc68f743f645d199b6072d1f650333be48feb7","_number":854894,"virtual_id_number":854894,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-29 18:43:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-08-29 06:39:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-08-29 15:11:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-08-29 15:11:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-28 05:05:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-29 06:39:59.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-08-29 15:11:18.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"2a4289821789b9198561a979a283b9f3c06a2e03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-28 03:38:44.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"5f1f5508b38bae6fb3f6dedd12e660bdd6ac22fc","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-28 03:39:15.000000000","message":"Topic bug/1987460-stable/yoga-stable/xena removed","accounts_in_message":[],"_revision_number":1},{"id":"8b047c4b915140748d328612312213a79aee7817","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-28 03:39:20.000000000","message":"Topic set to bug/1987460-stable/xena","accounts_in_message":[],"_revision_number":1},{"id":"5fbde77a71822dc4dc620205a21bdfed8742114f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-28 05:05:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cafb6f8688b48fc970c67147b7f52c2\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8337a8e0fad487aa1c8b87ebcc77d36 : SUCCESS in 3m 57s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34809b1f134c4f2c820489ef4100d7d9 : SUCCESS in 3m 54s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a596f7fb660405bb22e3a61e428da88 : SUCCESS in 4m 01s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/e1a03cf930764efeb470ed7078237c81 : SUCCESS in 19m 27s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/1007e07262de4a968cd3887fee06ce5d : SUCCESS in 19m 05s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6375ef353b85461fa04ebc3ea1a341ab : SUCCESS in 16m 26s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/236a8c85de9a4bc3b8b07cc761b09ece : SUCCESS in 16m 23s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13a76679606b45cba90cb5277c0e7198 : SUCCESS in 15m 07s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7bc53f5750f465ba3136a10466aad54 : FAILURE in 48m 37s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b02954eff9fd40539d95f05abd49704d : FAILURE in 46m 29s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0d99412b872483a9f322de8005fadb1 : FAILURE in 45m 10s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae22dae4c9dd42af89023f131f3c267b : FAILURE in 35m 32s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/61edf941bae54a7b9682c244d650e671 : SUCCESS in 27m 51s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f867858210f8403490174e9bac79f158 : SUCCESS in 1h 22m 39s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cd9723e3608143eda8ecbc773cf495fa : SUCCESS in 1h 22m 42s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4c0977b3ab6940afbd832eacfab361bf : SUCCESS in 1h 14m 12s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/53eb55e062ea4b9bbab4a4b3f95ba8fc : SUCCESS in 1h 05m 42s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9b635f7c59649cba804a05324c762f4 : SUCCESS in 19m 53s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/7ef1f8108dfd4c6697c543f9149c33ef : SUCCESS in 26m 58s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/348933be486d4222abb099856910fe0a : SUCCESS in 19m 46s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa833e623ec046e4965c65d00eac7dec : SUCCESS in 19m 26s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50835a562c494d81bd1e630071a87e49 : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":1},{"id":"cf88292db8eeb99e4e7369d42052be345f74bd17","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-08-29 06:39:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d04b76cac2c535deb462d7ddc7d5023e833ce423","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-08-29 15:11:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3c35bf890a803892f02016cdd996c83acb26927c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 15:15:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"decaf7f963ce326a875fb353956fc48edef6bbbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 18:43:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c38c251b2c6a4563ab05541c8d196778\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a192865af94c44b3902a8830dc8746ea : SUCCESS in 3m 44s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed7b64fa036f43ff949b4f81d48eb080 : SUCCESS in 3m 42s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/895e2767265a49f3b073805fa13d93a3 : SUCCESS in 3m 36s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/0ac0b06abcfe4bb0a210d708a1a6a9e2 : SUCCESS in 21m 02s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/8427f9a82a8c4348a8db99eda6d54d2c : SUCCESS in 20m 48s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92928e5d0ae549fd86dda29723e29432 : SUCCESS in 15m 38s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92c9d842962b48cfb774f6ecd51489ed : SUCCESS in 16m 11s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4059da3fedd042879bcee7cfde8b3ce0 : SUCCESS in 1h 31m 51s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/19914b6b9fa74a9090889a0c0d7acddc : SUCCESS in 1h 47m 44s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/15ccf6c172ea471f8748cf85cf677a4e : SUCCESS in 1h 37m 37s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e1100cef696e40efa0ef49fcdb3bff2e : SUCCESS in 1h 12m 44s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/e76fd757d3f74421b7e8e2017abf3c45 : SUCCESS in 27m 08s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a3086aac2fd4bb79c2c7d2854c42139 : SUCCESS in 17m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8da18c05d234bed99b89ab9a6f5565f : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":1},{"id":"2590f150ac84f9f59e65088f9011aaf74784f86e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 18:43:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cacc68f743f645d199b6072d1f650333be48feb7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 18:44:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48bf6669a30d48c6aaabf5558bae1eb9\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f19d6e59b4a46ea805ef93980c9a56b : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1f26bbfda91d0c5f4c8f99da2790c84dbb64ad76","revisions":{"1f26bbfda91d0c5f4c8f99da2790c84dbb64ad76":{"kind":"REWORK","_number":1,"created":"2022-08-28 03:38:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/854894/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/94/854894/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/94/854894/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/94/854894/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/94/854894/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/94/854894/1"}}},"commit":{"parents":[{"commit":"3677b7aa1d2e2fea10828918c6b07d82e0f9dca9","subject":"[ovn metadata] Add more tunable parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3677b7aa1d2e2fea10828918c6b07d82e0f9dca9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-24 00:52:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-28 03:38:44.000000000","tz":0},"subject":"mlnx: Allow workaround for duplicate resources","message":"mlnx: Allow workaround for duplicate resources\n\nThe commit b2e7de79df98a986768aaef203ddbe02bee41072 , which was merged\nduring Train cycle, introduced a few new parameters to support multi\ninterface driver, but the following two parameters are already declared\nin the corresponding agent classes and trigger duplicate resource\ndeclaration when these classes are used in a single manifest.\n - dhcp_broadcast_reply\n - interface_driver\n\nThis change allows users to workaround the problem. Details are\ndocumented in the release note.\n\nCloses-Bug: #1987460\nChange-Id: I993523dc7f0ec8ad90a50d049dc0f5ccf9f2b191\n(cherry picked from commit c4f3abe22e5f57a03c5953f0b342433d93f36731)\n(cherry picked from commit 49bfd89810d85dccd658ea9f16480c679d5955d1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/1f26bbfda91d0c5f4c8f99da2790c84dbb64ad76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/1f26bbfda91d0c5f4c8f99da2790c84dbb64ad76"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[]}
