)]}'
{"id":"openstack%2Fpuppet-neutron~856206","triplet_id":"openstack%2Fpuppet-neutron~stable%2Ftrain~I993523dc7f0ec8ad90a50d049dc0f5ccf9f2b191","project":"openstack/puppet-neutron","branch":"stable/train","topic":"bug/1987460-stable/train","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-09-09 11:07:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I993523dc7f0ec8ad90a50d049dc0f5ccf9f2b191","subject":"mlnx: Allow workaround for duplicate resources","status":"MERGED","created":"2022-09-08 02:05:39.000000000","updated":"2022-09-09 11:08:26.000000000","submitted":"2022-09-09 11:07:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"856206-bug/1987460-stable/train","cherry_pick_of_change":855558,"cherry_pick_of_patch_set":1,"meta_rev_id":"a2cc95cf7598d8b7bef619c00cd6f19e5eb438ca","_number":856206,"virtual_id_number":856206,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-09 11:07:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-09-08 19:27:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-09-09 09:13:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-09-09 09:13:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-08 04:07:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-08 19:27:27.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-09-09 09:13:31.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"18d3e88c364ff6c7eeb06231e53843ac294b928a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-09-08 02:05:39.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"78b1e7c4308601e98a50689e4bd3e4c34c7ed067","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-09-08 02:05:46.000000000","message":"Topic bug/1987460-stable/victoria-stable/ussuri-stable/train removed","accounts_in_message":[],"_revision_number":1},{"id":"ab0c2a1de5935ab43036df57f8509a414623b4f1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-09-08 02:05:51.000000000","message":"Topic set to bug/1987460-stable/train","accounts_in_message":[],"_revision_number":1},{"id":"4d59c6dc90b111d551b0238e7175a76f99782c4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 04:07:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b0b2f65e9e047bc9e30e06dba4bdb51\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/065aa869a71540da80a7a69dfc59ac3a : SUCCESS in 5m 46s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/294958be1de6425fb12d99cdc0a798de : SUCCESS in 4m 27s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/10a7f0eb16aa44a28e5851ae0ffd2308 : SUCCESS in 4m 35s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/af19ed77c6d34cdc88946d41111d75d7 : SUCCESS in 30m 49s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/c676e9fd534b4c989af413eb931cf69b : SUCCESS in 21m 48s\n- puppet-openstack-unit-6.4-centos-8-stream https://zuul.opendev.org/t/openstack/build/82a530906b1d40fc87b9d94f2c69b4c2 : SUCCESS in 21m 26s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e6aac42afbc045549dde153c561cd551 : SUCCESS in 16m 57s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c6cf9e86bbf644fda6337cda5e277e3b : FAILURE in 19m 53s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/3dfd6059bd5b4259afa28b1b84784534 : SUCCESS in 1h 03m 16s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/16a36203ff5c44fbb47c88c82c4ac9cb : SUCCESS in 1h 10m 38s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/07d4e51a3e8942cb8da2b011f6437dae : SUCCESS in 1h 11m 11s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/ea710c660ff743c9ba1ba279a89aee27 : SUCCESS in 55m 38s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/307ad722500f4fb9874bbfcd35449192 : SUCCESS in 53m 23s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/04f68a197bb2422d848ae64146c01423 : SUCCESS in 51m 57s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cd134e42fb4e44bab63d23f2be9f31df : SUCCESS in 46m 33s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/684631c071444de3ab7134a18840bce9 : SUCCESS in 44m 03s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/44fbdce31fbd4768889fa06cf99d8dd9 : SUCCESS in 59m 20s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/47eace732b704b4e92b85b1de7f2d814 : SUCCESS in 1h 01m 50s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2f72f1e1043942c291fa3c15f801dd0e : SUCCESS in 1h 05m 40s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e741f9ada9084586a144559c8c4b3270 : SUCCESS in 48m 38s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/ad4e564584ef4998b084e1e092ca526c : SUCCESS in 25m 21s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e844c8c077e049dd97770743f78c79c1 : FAILURE in 10m 25s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3d6147c8b9f46b4b4a747ac20da3291 : SUCCESS in 4m 44s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ac220fd0043e4857bb945eb0b5dd70bc : SUCCESS in 1h 43m 25s","accounts_in_message":[],"_revision_number":1},{"id":"c3bec9c404dbf8125a4e6af06496d214e3d961af","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-09-08 19:27:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3a9ec36fa867f60213b8becd5bdfe50bebdd1b50","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-09-09 09:13:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"45d36235489be051dd005f34d4a87ae639354897","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 09:15:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2ebdc636a6425d71f6a238f846cd4aaf382edd1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 11:07:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/577115e9cd3846f09fbc2c9869758cdf\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a1ed632ce30d4f06b41ffcace77f2c2b : SUCCESS in 5m 18s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b306f35790db4ad4845e61202f0ae6fb : SUCCESS in 4m 20s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/29744a9edf044fb48f781b2183bdefd3 : SUCCESS in 4m 58s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/204646df46d5421caf864963051e7e9c : SUCCESS in 36m 58s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/65b2b7a18f5148b2a5ce353c83866376 : SUCCESS in 16m 18s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/29650faf061e40bf808e8248080f1616 : SUCCESS in 1h 02m 34s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/942a670c79c34d48a6fe66a4151aedfd : SUCCESS in 1h 09m 51s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/b69faf12606749c0b9442dbad5b9ce38 : SUCCESS in 1h 11m 53s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/e3c96b441c3544b7aa99c44f4e200fa8 : SUCCESS in 54m 18s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/c5cdaf7be142410bae578b2fa5e94e44 : SUCCESS in 24m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ec7036820cc4bcaac83f953cf61abc2 : SUCCESS in 6m 16s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0d36b8d2d8ca44229c865c85b8358211 : SUCCESS in 1h 42m 49s","accounts_in_message":[],"_revision_number":1},{"id":"5dc63b21e126b69acd3a30106355a9440f620bc5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 11:07:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a2cc95cf7598d8b7bef619c00cd6f19e5eb438ca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 11:08:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b70dcaa870c4df8980be2e1eef1a117\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a35316d96778448d82a3797a66743428 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b2d406629ad1ce4b784e7c9b1be87f570089c468","revisions":{"b2d406629ad1ce4b784e7c9b1be87f570089c468":{"kind":"REWORK","_number":1,"created":"2022-09-08 02:05:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/06/856206/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/06/856206/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/06/856206/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/06/856206/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/06/856206/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/06/856206/1"}}},"commit":{"parents":[{"commit":"b793bbfc699ac35b5c789410fe82bdfc1cd3d3d9","subject":"Add report_inteval parameter in ml2 baremetal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b793bbfc699ac35b5c789410fe82bdfc1cd3d3d9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-24 00:52:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-08 02:05:39.000000000","tz":0},"subject":"mlnx: Allow workaround for duplicate resources","message":"mlnx: Allow workaround for duplicate resources\n\nThe commit b2e7de79df98a986768aaef203ddbe02bee41072 , which was merged\nduring Train cycle, introduced a few new parameters to support multi\ninterface driver, but the following two parameters are already declared\nin the corresponding agent classes and trigger duplicate resource\ndeclaration when these classes are used in a single manifest.\n - dhcp_broadcast_reply\n - interface_driver\n\nThis change allows users to workaround the problem. Details are\ndocumented in the release note.\n\nCloses-Bug: #1987460\nChange-Id: I993523dc7f0ec8ad90a50d049dc0f5ccf9f2b191\n(cherry picked from commit c4f3abe22e5f57a03c5953f0b342433d93f36731)\n(cherry picked from commit 49bfd89810d85dccd658ea9f16480c679d5955d1)\n(cherry picked from commit 1f26bbfda91d0c5f4c8f99da2790c84dbb64ad76)\n(cherry picked from commit 181975c44d803c1833cdaadaac06247d0c0fb3c6)\n(cherry picked from commit 29e3ff1ee3f787219328b4c7961a3ee544c91f05)\n(cherry picked from commit 0b225b0bd1f786212aaf21b370e7e54525442c66)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b2d406629ad1ce4b784e7c9b1be87f570089c468"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b2d406629ad1ce4b784e7c9b1be87f570089c468"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
