)]}'
{"id":"openstack%2Fpuppet-neutron~856979","triplet_id":"openstack%2Fpuppet-neutron~stable%2Fwallaby~I8e32e0ca02b841058dc71f00a72b825e6441c1d3","project":"openstack/puppet-neutron","branch":"stable/wallaby","topic":"bug/1945973","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-09-12 18:37:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8e32e0ca02b841058dc71f00a72b825e6441c1d3","subject":"ovn: Do not expect the separate plugin package","status":"MERGED","created":"2022-09-12 01:45:46.000000000","updated":"2022-09-12 18:38:02.000000000","submitted":"2022-09-12 18:37:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"856979-bug/1945973","meta_rev_id":"7cbb4e6d48cf8758cd938f42f2af39f22e686649","_number":856979,"virtual_id_number":856979,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-12 18:37:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-09-12 06:18:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-09-12 15:30:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-09-12 15:30:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-12 01:45:46.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"state":"CC"},{"updated":"2022-09-12 04:46:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-12 06:18:18.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-09-12 15:30:18.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"28686ed5f588f7b27b45a2977e37afb0d9bec160","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-09-12 01:45:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"09bc87878eab679c950dd3f31ae45c440530d005","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 04:46:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b226ca69392f4501a9c6d59244383c41\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32482663cb5244b78aebbc46670707bf : SUCCESS in 3m 23s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dad54efb491f42069ca1962a86febc57 : SUCCESS in 3m 36s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/10500878adb44db68b4d2487ed06155a : SUCCESS in 19m 49s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f7168cbaac645919c904fb496fbac67 : SUCCESS in 16m 13s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d3e2a5973904bfc9ae41fb7743c502f : SUCCESS in 14m 53s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eef07a5af7504ef0a604bf1ee33788ce : SUCCESS in 50m 01s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e041e47268b4427b837cb1750e336a7 : SUCCESS in 50m 57s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b3308dc311540c9960ce7f0d681db26 : SUCCESS in 52m 03s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3bf047cac34b4e5fa353f07527d4e115 : SUCCESS in 37m 30s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e6cc4897a25b4c7788ca896de9dc37d5 : SUCCESS in 1h 07m 36s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e486d6ab2a0a4c30a720b831e41fabed : SUCCESS in 1h 06m 05s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/46ea5abd02394c5099fd93eb7df30797 : SUCCESS in 1h 13m 29s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/48926f26679444269a22800c761e416e : SUCCESS in 44m 46s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/2d43cd16c74d4002a8168bcc368957cd : SUCCESS in 27m 39s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07e30fafaa224e7d82addf6ec31b80c3 : SUCCESS in 20m 50s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/da0e06e5155d40888cb213008c1d99bf : SUCCESS in 2h 57m 47s\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/3a08bf7654d04b9db480685a311956d9 : SUCCESS in 2h 04m 35s (non-voting)\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/8c1b12e0febe463ebf145fe4a57562d1 : SUCCESS in 57m 07s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ba4b37cbf754f1cb588bdaba9cb38b4 : SUCCESS in 3m 02s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5bb9a8a257e64d8c8123f869180b6f2b : SUCCESS in 1h 05m 44s\n- puppet-neutron-tripleo-centos-9-standalone https://zuul.opendev.org/t/openstack/build/4ec3a2077ace481c8caaeb4cc1b1c00c : SUCCESS in 1h 02m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f1e64ac5f88b18e0561f371892bc9752d6e2e7a0","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-09-12 06:18:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b1e84ac074edbb2a1cb53915eab4e217b13a668e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-09-12 15:30:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6b166c34418f30a555140e189a39fa3860b02ba6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 15:31:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a79591095c1094aa3191dbc23df624da8e1521ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 18:37:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/919b2ce5fe4e4b2c86945480b3dfb9b9\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62adcd4ab5c042c0b23309bc2cc1a5db : SUCCESS in 6m 11s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/220528e44f154a75a240c8eb74c2b5ee : SUCCESS in 5m 20s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/4a5bdc0adf1b49eb90d08fb0b8c697e7 : SUCCESS in 28m 11s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f70ce08dacd4bc7bc67e07ee0e39c53 : SUCCESS in 24m 44s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/07bacca4f7ec4efe8d5f9a1b66f7b9ab : SUCCESS in 1h 16m 25s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5a424a67c79f4ad1a576a6e3755ff99b : SUCCESS in 1h 22m 20s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c7eda9a046664cf1a868a90ddf91cf10 : SUCCESS in 1h 12m 47s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5cde4a0dab554355a688e986a0415e31 : SUCCESS in 52m 49s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/5df5a6136c874eb8b8b98ceaf370e613 : SUCCESS in 31m 13s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca99462184fb4f09bbc1a4ca5defdbf8 : SUCCESS in 18m 54s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/0ac7638bddef40c39196f056215841d5 : SUCCESS in 3h 00m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6974ce35acf34f39a018361b3fab44d6 : SUCCESS in 4m 09s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b413246c47b5467ab437ea640c96ed1d : SUCCESS in 1h 20m 45s","accounts_in_message":[],"_revision_number":1},{"id":"48cbc0c555436c08e9dddf4df4402021b25af8a6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 18:37:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7cbb4e6d48cf8758cd938f42f2af39f22e686649","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 18:38:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf72f9f5c9d64724967a062f21571bcc\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/197793e8a8c6455e9e07c235f67887c5 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e70d3e68ad7ee40465f3514e79b8ae814840da6f","revisions":{"e70d3e68ad7ee40465f3514e79b8ae814840da6f":{"kind":"REWORK","_number":1,"created":"2022-09-12 01:45:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/856979/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/79/856979/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/79/856979/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/79/856979/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/79/856979/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/79/856979/1"}}},"commit":{"parents":[{"commit":"181975c44d803c1833cdaadaac06247d0c0fb3c6","subject":"mlnx: Allow workaround for duplicate resources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/181975c44d803c1833cdaadaac06247d0c0fb3c6"}]}],"author":{"name":"Jake Yip","email":"jake.yip@ardc.edu.au","date":"2021-09-28 08:01:33.000000000","tz":600},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-12 01:45:05.000000000","tz":540},"subject":"ovn: Do not expect the separate plugin package","message":"ovn: Do not expect the separate plugin package\n\nOVN is part of Neutron since Ussuri [1]. As such, there is no longer a\nseparate python-networking-ovn package that needs to be installed.\nThis change removes the logic to install the separate plugin package,\nand also updates the config file path to use the file provided by\nthe current neutron package.\n\nIn addition, this change ensures the ovn-metadata-agent package is\ninstalled in Ubuntu/Debian as well.\n\n[1] https://docs.openstack.org/releasenotes/neutron/ussuri.html#relnotes-16-0-0-stable-ussuri\n\nConflicts:\n\tmanifests/params.pp\n\tspec/classes/neutron_agents_ovn_metadata_spec.rb\n\nCloses-Bug: #1945973\nChange-Id: I8e32e0ca02b841058dc71f00a72b825e6441c1d3\n(cherry picked from commit 854ece099a20a248a5422e4c0ba1f73ecc138d22)\n(cherry picked from commit aecfaa35e2cefceb3f0af9c6a9604a0d2bf6e719)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/e70d3e68ad7ee40465f3514e79b8ae814840da6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/e70d3e68ad7ee40465f3514e79b8ae814840da6f"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[]}
