)]}'
{"id":"openstack%2Fpuppet-neutron~864630","triplet_id":"openstack%2Fpuppet-neutron~master~I64da797bc921702f710e7328ef3e230354e2ceaf","project":"openstack/puppet-neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-11-23 17:01:42.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I64da797bc921702f710e7328ef3e230354e2ceaf","subject":"Remove support for networking-vpp plugin","status":"MERGED","created":"2022-11-16 01:42:01.000000000","updated":"2022-11-23 22:00:50.000000000","submitted":"2022-11-23 21:59:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"864630","meta_rev_id":"a4f7d8899ab6647cba148f0fe28210dac1a18d34","_number":864630,"virtual_id_number":864630,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-23 21:59:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2022-11-18 09:21:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-11-21 15:20:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-11-21 15:20:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-16 04:58:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-16 11:09:54.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-11-21 15:20:01.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"0f6bdbcc4fecc3f63da813985dfcf66e930b67c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-16 01:42:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"09a618ddb3da4c3e6a590c2dc07d543923148045","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-16 01:56:35.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"f27abed39ed42841757f3f431d7018d2ecae5f8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-16 02:23:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a5310347dd50be3b5b5ea5ef23f6c2c2695221f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 04:58:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd8ef4df5a664281b7d571777001a8e2\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/687572c5920b47dc9df30f84bed47524 : SUCCESS in 3m 04s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd8471e51def4857a908aa42543cac47 : SUCCESS in 3m 10s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c39867293cd451a9b4263a6a641d591 : SUCCESS in 3m 12s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/82c0387ec46745e0a042becb4c746e62 : SUCCESS in 16m 27s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0129370fa0174604ad1609d2defca793 : SUCCESS in 12m 42s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/724e424e37624b079ce78dee676fc7ca : SUCCESS in 12m 50s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9426e84662845c5826a256586a00ccb : SUCCESS in 13m 21s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25710adc1574426e9be1fa6f2bc9f15b : SUCCESS in 44m 28s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1369ba11fd164728a7540eb62cce01a8 : SUCCESS in 1h 00m 50s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/83470e66916041c596eba469871d39c7 : SUCCESS in 42m 28s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a80da91942214a23aed9104e75f088dc : SUCCESS in 42m 46s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee76a03ee64849d0813e07a873bb5af4 : SUCCESS in 43m 52s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1bb4a6aacd784239876f7add4711bb5a : SUCCESS in 20m 45s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/19d326e0e6604c3c9fdcf850fd108d3e : SUCCESS in 1h 07m 05s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4b513973ab6a4ea2883a528f7f80dbee : SUCCESS in 1h 20m 15s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/52ee906d2fa848fd8feaac7054a1e16e : SUCCESS in 1h 00m 48s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/972984a6a97a4dacb452072c65e16a87 : SUCCESS in 54m 31s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f19c5ed569c84d468586dc8e66985981 : SUCCESS in 50m 59s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c585eaf051ef424da9f5d92ff5012d46 : SUCCESS in 15m 10s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/da01b098424e44ce9fda30a4289f819e : SUCCESS in 19m 51s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52f741fcec764704bcf7741bf5d09acf : SUCCESS in 15m 38s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44d8ad78e6a844e0acff56e39fc01bb2 : SUCCESS in 15m 06s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/a4a2941bc28440a489694abd0a824639 : SUCCESS in 2h 13m 14s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/8c9edab1c4bf40fa8cfe1a92a7bf36f7 : SUCCESS in 46m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/115a23ae40cc4b258cbe38644c92a01c : SUCCESS in 3m 13s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4183110732164a76a98411e921a71b46 : SUCCESS in 1h 06m 09s","accounts_in_message":[],"_revision_number":2},{"id":"343ba60a9a10c024c3b2f1bee55ac613e85da59d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-11-16 11:09:54.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"28a313d3e25ec1d24ea857a6be4aea121291c492","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-17 01:51:34.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"63d0804af730aac45e7003b3ba94ece5c09dd3e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-18 01:50:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8d76450b721ba58c5272ae4ad691a800c6ef9a33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 04:51:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/14fd2ccd9b444bedaf3f47f89552d1d2\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2bcdcbf78c264c91baefd766df3d4a39 : SUCCESS in 3m 44s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef847eb6572f4244a8c44f6f4246e865 : SUCCESS in 2m 49s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c721dc83ec94c5aaf1d82162d637a4a : SUCCESS in 3m 14s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc362f1e8de34d2ca2bb4658ac047fbe : SUCCESS in 18m 09s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a87adc7d3f543acabe1be35385a3890 : SUCCESS in 12m 07s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/79b7e8ffda014fb9bff46d5d861847b2 : SUCCESS in 12m 06s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/502c417879914ab392334218e5b1ec63 : SUCCESS in 11m 44s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9c457acf4944c2e97c39ac8bd9803b9 : SUCCESS in 49m 00s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b226d4f5e7a47aca1c1cf76443ed8d5 : SUCCESS in 1h 00m 55s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6cccbcc50084aa1984f8b501f08fef5 : SUCCESS in 50m 14s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/adf300b054de479ebed37fd9b5fed5d4 : SUCCESS in 41m 47s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3a5759de76945988d0ebc451d76dd1d : SUCCESS in 44m 31s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9dba2a82412644b4aa3e02b7450b1233 : SUCCESS in 23m 07s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/35530e6d03d848b89c11ec40fb8ebe8f : SUCCESS in 1h 01m 30s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5ed76b326f2047c8bd78bce4cc0b84f2 : FAILURE in 1h 10m 26s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b4ed6bae6fb1407aaccb9d1e8b17a707 : SUCCESS in 1h 29m 54s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/be958b7b6ceb43db985c9544d6acbf3b : SUCCESS in 1h 01m 52s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/72e8ab53997e483f898b0ac1025ddf4f : SUCCESS in 59m 26s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a09eefe96e84169bf2cdfe376b40f62 : FAILURE in 14m 10s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a6b8a2ae5b045cfabd89b9d8e33756e : FAILURE in 20m 15s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab8ae95bea4c4a8b92fd5f8d49243a85 : FAILURE in 13m 55s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4dfdc31c2a064d5dacc6bf30cc0afd57 : FAILURE in 13m 26s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/c364fa837bb84028a528ad3ab24db277 : SUCCESS in 2h 54m 36s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/d8342b5a87204030abc3513e3538ff05 : SUCCESS in 53m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14ed042260eb48cbbf95f48ada215858 : SUCCESS in 2m 54s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8279236a16b8494babff218211a0727a : SUCCESS in 1h 22m 47s","accounts_in_message":[],"_revision_number":3},{"id":"255c59d84ebe21e9bcfa29d46275095849701808","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-18 05:38:28.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"19a2a4552f06fd54ec51576410ca29059c063042","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 07:51:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ada806595e14a02b0bc5481e613be93\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/79051838810d4690b11ef15286fc1fda : SUCCESS in 3m 17s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/369208c5be8c4abea2338fe46178916f : SUCCESS in 4m 00s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2d1c1998eae4598bbbaab03c9afce1f : SUCCESS in 3m 29s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/a4f664d631ff4c959433ac702ccd7946 : SUCCESS in 16m 40s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/143d1bec4e57426fbea2154ef9836a64 : SUCCESS in 14m 34s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9662fa04b8264324bfad89172db994b8 : SUCCESS in 12m 33s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27ecd612c91242ad8d4bc607748175ed : SUCCESS in 13m 41s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e165cb495504401ea6bcd5754cb620d9 : SUCCESS in 45m 26s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4353490070b143cf947efea050b14db7 : SUCCESS in 56m 02s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e000a272e6d49a5978e2adca061c6f3 : SUCCESS in 44m 34s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/884e3663b0574d699b3e8db68ac0c9a9 : SUCCESS in 42m 10s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f33952b775434b17b993d4a8bb2f78fa : SUCCESS in 36m 20s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/39eccf67faf14988977ac7c97ef20adc : SUCCESS in 21m 10s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1226e202b8fa4870b165fa7ef63258f5 : SUCCESS in 1h 02m 39s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/405273562c58434183ad2fca292ee885 : SUCCESS in 1h 13m 01s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e27ee984f3114abd86e0af186f1ece32 : SUCCESS in 1h 09m 41s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4bb58b8e6c88425fbab350dfa84946dc : SUCCESS in 55m 52s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0b558b60044c4b80b3aa01d456c82f29 : SUCCESS in 49m 57s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32245ba28d0941ddb167aa9d890e8c9c : SUCCESS in 17m 10s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/341e2d9801844afc9291becca9be484b : SUCCESS in 20m 25s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ddff5d28c517471984e008231efa307e : SUCCESS in 16m 34s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/992349c03fda4ed89a66d56f48845fb6 : SUCCESS in 16m 05s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/09c02bd62d1f471ea01f13b1738ea27f : SUCCESS in 2h 09m 30s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/e3acf94c411143a4b85e29553ac497d3 : SUCCESS in 42m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c09c4434db5c427c8c53d1d530bd9b37 : SUCCESS in 3m 44s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/33e03f777e57456f8d4b959d98338fbf : SUCCESS in 1h 08m 12s","accounts_in_message":[],"_revision_number":4},{"id":"476cb7c27fe2e20841b2ce672f09e91a2d3a6e25","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-11-18 09:21:20.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"837abf76ae3888a6ef8d71db4bdb11dec180cb8c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-11-21 15:20:01.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"eeb11fc7c7690e8122c65cf0d98eece584dfa651","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 15:21:29.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"50d70e19cff7f55a62e35f43022ea3088e9bbcba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 17:53:46.000000000","message":"Patch Set 4: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/864804, https://review.opendev.org/864805, https://review.opendev.org/864452 are needed.","accounts_in_message":[],"_revision_number":4},{"id":"bafc5bffa48a2179d23a721281efe9219b5ec7a2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-21 23:30:27.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"19580d37757a0f25811a232f54136466ca691271","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 00:48:18.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ffeb439a7254ca9b3e9d9aa23fca463\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b80554719f3c45aeb61f06287d0ba571 : SUCCESS in 3m 23s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09cf8e95706c4f8a9b73f514758675b1 : SUCCESS in 3m 28s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/127d6f8827be4e669145ea48dab73013 : SUCCESS in 2m 52s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/a24b0ccd6c3c4253b7531edb8b3fc3c7 : SUCCESS in 15m 02s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42fe17ec9ef5487990042eabe4fa34ff : SUCCESS in 11m 46s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d5557d1efca44ba9a4d5eb2fbb6d3dd : SUCCESS in 10m 51s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47aa0c57d123459aa4a7be6e3a9c72b1 : SUCCESS in 11m 44s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5abc8b82c243447d994f9e765e8e3ae5 : SUCCESS in 50m 00s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3db2072746ae49c187c46cc116926bd2 : SUCCESS in 59m 15s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc831c08a46d4061b7ed1197f1a2405b : SUCCESS in 45m 42s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b70cfb2f54d149eb9a002acbd661ebb2 : SUCCESS in 41m 20s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bddcdbaa39ff449fb74918c63a61208d : SUCCESS in 35m 44s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a8c71c3306b048258ad6730aea5e769f : SUCCESS in 16m 35s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e3b3d080df2f4cc2ba46d0bad29d2637 : SUCCESS in 59m 54s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c6c54c2deabd47ad93e2864fd75bf97a : SUCCESS in 1h 12m 33s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e4896f571bf94a6c91b87f053ee7afef : SUCCESS in 59m 15s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/203de45d97ff46c69c245ebd75452b6b : SUCCESS in 53m 16s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fa579aeab75743d09c3f2d2597907333 : SUCCESS in 51m 22s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb7585f071744c898c682710188821a2 : SUCCESS in 14m 09s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/bce32466417a4dbdba36c2c623309487 : SUCCESS in 21m 20s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ab11e6c43c349c59f284d49a9dec3b8 : SUCCESS in 16m 47s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/edcfdc86bedd451186f3dbf833a75faf : SUCCESS in 15m 33s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/f8efe7621e544e08955791b7c94b26f6 : FAILURE in 42m 13s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/444746a459d846cdb6e9dea70f249c2f : SKIPPED Skipped due to failed job tripleo-puppet-ci-centos-9-content-provider\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2c3f3bb936347fc973e4cf9166950e4 : SUCCESS in 3m 08s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7efad71807214ad4b03e2a43252f565f : SKIPPED Skipped due to failed job tripleo-puppet-ci-centos-9-content-provider","accounts_in_message":[],"_revision_number":4},{"id":"b90c10fa55c032cce3429dfc418d77e3884f31b7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-23 17:01:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"dd5beba677f19169f9955ad9bcc47979e44c2521","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 19:17:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b562b4ad5ec5467b9d0bdce58b2fddd8\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d3a13b392ee47eabea9dc6cc2f3fc8a : SUCCESS in 3m 53s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61b74f5b68124a5d96ab63860de9f01a : SUCCESS in 2m 58s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/295949c0858541eb91f2e86b35277258 : SUCCESS in 3m 03s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/de2cde3968bb46ed8b271f55b049730b : SUCCESS in 15m 58s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c210984e86434002a8eb995cb2796b1e : SUCCESS in 12m 52s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35f7a4ca9ff345eaac0561f22643b93b : SUCCESS in 12m 20s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f31d9f54d3984d198c33afadc235b783 : SUCCESS in 12m 34s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d796c67582494ca188c39566185b4200 : SUCCESS in 51m 47s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a963f76d3ee545298a49a3acbe2800d0 : SUCCESS in 58m 58s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5247c64b718c41039a77aa207ae0ebc2 : SUCCESS in 49m 43s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0a70650d2704a3e81f58cbd479e4918 : SUCCESS in 32m 43s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8e4e2b7beb04941b263bfeab8305bab : SUCCESS in 43m 00s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f668565a9da544c2a77c7d6602bca48d : SUCCESS in 23m 50s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0f7538f572a24eb8801bc90dd6576a12 : SUCCESS in 1h 04m 20s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/550ada002c164eac862ebd05865efa18 : SUCCESS in 1h 00m 43s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9a6338531e6a4f4093d79903c8bf67c0 : SUCCESS in 1h 03m 26s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ee159463ef81467d9f82d626f2e0f023 : SUCCESS in 45m 06s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/57d0d65892314d62ba76c6c45a189fd0 : SUCCESS in 54m 45s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b2873cc3e77430b8d0c9f3a138b8046 : SUCCESS in 15m 55s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/b9ce599aab3c478392e0e3f459da15bb : SUCCESS in 17m 21s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d815d761c0004aa1bb8eb623cf5c5a8d : SUCCESS in 19m 21s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a22b6a55270d4d8596053f4090f067ab : SUCCESS in 13m 38s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/160f3fd299114d7fb1db968b02966d04 : SUCCESS in 2h 13m 25s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/7a071adfc8b64e94a69ac4d1239fea45 : SUCCESS in 42m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9431035ed22b40daa30b95df94cf72ad : SUCCESS in 3m 03s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5c2ab18fc4fb41128a94209e520b97fe : SUCCESS in 1h 07m 07s","accounts_in_message":[],"_revision_number":4},{"id":"4257959a741956fa0784d6ae45935041b61690aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 19:25:58.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"87cc9005468443f2891ab8415ac08937c6ef0252","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 21:59:52.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b41d345f76c4265bd30534a3f357653\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/761b34af9c8c46a1b45bf7ff31d24b39 : SUCCESS in 3m 24s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a98bf256382451a8069247a25ba8b7f : SUCCESS in 2m 44s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d43f2696aa544d9b2d7af50fd707624 : SUCCESS in 3m 02s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/c013a6eea8c34ae49d6e9ac2976e3b41 : SUCCESS in 16m 20s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c81af92987449f59a05df055810fb18 : SUCCESS in 11m 32s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a5fdf21ee66d426bae04ab7c85e34329 : SUCCESS in 12m 29s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a9751ddac93643479055c3dd9c0ba077 : SUCCESS in 1h 00m 12s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6d64d8c8c6784e7bb4a4be90af1ec96f : SUCCESS in 1h 06m 25s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7d88066cadf64a8ea69ee763f302c5b5 : SUCCESS in 1h 07m 00s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fd85aa632e9845f6b8b3cb0f1f4682f6 : SUCCESS in 49m 26s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5c01bc83d3484f8da41f4e37bc3f4f6e : SUCCESS in 52m 43s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/9be3005c66864de39592be851e1cf25b : SUCCESS in 18m 13s\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/36757191480b475585eba4bb99c623ce : SUCCESS in 2h 31m 28s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/1ce1b394ca8d42a697ad6c61b247bbec : SUCCESS in 42m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/044eee5983fe43adb86a2a7ced993fde : SUCCESS in 3m 26s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9df5e32732d5461386e219a980b9dc41 : SUCCESS in 1h 17m 09s","accounts_in_message":[],"_revision_number":4},{"id":"1d1a1e7b74f0b2b9b877d3ac3467614b32f281bb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 21:59:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"a4f7d8899ab6647cba148f0fe28210dac1a18d34","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 22:00:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c029f04d1184665b5e826c4d79cef07\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/878bd662859a4067990520c042743a15 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"746280be8640e61e7263199c3eac28e0a3ee603c","revisions":{"fc3b2248b0fc7e991f5ab7ec70b3423722b18508":{"kind":"REWORK","_number":1,"created":"2022-11-16 01:42:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/30/864630/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/30/864630/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/30/864630/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/30/864630/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/30/864630/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/30/864630/1"}}},"commit":{"parents":[{"commit":"3206f2f9c25bda2cc54f09f303056f86df28daa7","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3206f2f9c25bda2cc54f09f303056f86df28daa7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-16 01:41:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-16 01:41:26.000000000","tz":540},"subject":"Remove support for networking-vpp plugin","message":"Remove support for networking-vpp plugin\n\n... because it was deprecated during Zed cycle by [1].\n\n[1] 7a47c5a29d530982ed85e5976de7b95992b9ed93\n\nChange-Id: I64da797bc921702f710e7328ef3e230354e2ceaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/fc3b2248b0fc7e991f5ab7ec70b3423722b18508"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/fc3b2248b0fc7e991f5ab7ec70b3423722b18508"}]},"branch":"refs/heads/master"},"ce8439d96f07b0637627b6da4684ec853ddd1c97":{"kind":"REWORK","_number":2,"created":"2022-11-16 02:23:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/30/864630/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/30/864630/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/30/864630/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/30/864630/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/30/864630/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/30/864630/2"}}},"commit":{"parents":[{"commit":"3206f2f9c25bda2cc54f09f303056f86df28daa7","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3206f2f9c25bda2cc54f09f303056f86df28daa7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-16 01:41:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-16 02:23:34.000000000","tz":540},"subject":"Remove support for networking-vpp plugin","message":"Remove support for networking-vpp plugin\n\n... because it was deprecated during Zed cycle by [1].\n\n[1] 7a47c5a29d530982ed85e5976de7b95992b9ed93\n\nChange-Id: I64da797bc921702f710e7328ef3e230354e2ceaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/ce8439d96f07b0637627b6da4684ec853ddd1c97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/ce8439d96f07b0637627b6da4684ec853ddd1c97"}]},"branch":"refs/heads/master"},"9be4f0be02b7cadcff9a6dce1d8887b93aaf3241":{"kind":"REWORK","_number":3,"created":"2022-11-18 01:50:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/30/864630/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/30/864630/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/30/864630/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/30/864630/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/30/864630/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/30/864630/3"}}},"commit":{"parents":[{"commit":"f47b5fc3968267b58ad4834c786be0371b04abe3","subject":"Remove support for NSX plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/f47b5fc3968267b58ad4834c786be0371b04abe3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-16 01:41:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-18 01:50:25.000000000","tz":540},"subject":"Remove support for networking-vpp plugin","message":"Remove support for networking-vpp plugin\n\n... because it was deprecated during Zed cycle by [1].\n\n[1] 7a47c5a29d530982ed85e5976de7b95992b9ed93\n\nChange-Id: I64da797bc921702f710e7328ef3e230354e2ceaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/9be4f0be02b7cadcff9a6dce1d8887b93aaf3241"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/9be4f0be02b7cadcff9a6dce1d8887b93aaf3241"}]},"branch":"refs/heads/master"},"746280be8640e61e7263199c3eac28e0a3ee603c":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-11-18 05:38:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/30/864630/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/30/864630/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/30/864630/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/30/864630/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/30/864630/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/30/864630/4"}}},"commit":{"parents":[{"commit":"ed0709cb48d60462e503f0c28415dfa3c96ac1b9","subject":"Remove support for NSX plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/ed0709cb48d60462e503f0c28415dfa3c96ac1b9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-16 01:41:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-18 05:38:28.000000000","tz":0},"subject":"Remove support for networking-vpp plugin","message":"Remove support for networking-vpp plugin\n\n... because it was deprecated during Zed cycle by [1].\n\n[1] 7a47c5a29d530982ed85e5976de7b95992b9ed93\n\nChange-Id: I64da797bc921702f710e7328ef3e230354e2ceaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/746280be8640e61e7263199c3eac28e0a3ee603c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/746280be8640e61e7263199c3eac28e0a3ee603c"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[]}
