)]}'
{"id":"openstack%2Fpuppet-neutron~865614","triplet_id":"openstack%2Fpuppet-neutron~master~Ied984b3f5e1515a4daa0c0a58bdf2b85e674f202","project":"openstack/puppet-neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-11-29 19:22:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ied984b3f5e1515a4daa0c0a58bdf2b85e674f202","subject":"Remove deprecated mech_driver defined resource type","status":"MERGED","created":"2022-11-25 01:42:59.000000000","updated":"2022-11-29 19:23:45.000000000","submitted":"2022-11-29 19:22:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"865614","meta_rev_id":"6c77e63bb5ae63d611d86da7a1a46a273c67d082","_number":865614,"virtual_id_number":865614,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-29 19:22:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-11-25 13:47:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-11-29 17:01:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-11-29 17:01:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-25 04:16:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-25 13:47:05.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-11-29 17:01:20.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"3fa062ae8b59da8c8d419e408341593ee06066f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-25 01:42:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"40bedd0f098f28a079a56063cb783160fb030f21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 04:16:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d5092b7e45d47a1a8f6b5ea267d3fdd\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04a5b07ba93d4b5d912fb148d3143127 : SUCCESS in 3m 31s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1cda8c02f4954ea3b36dd701fd02c2a0 : SUCCESS in 2m 48s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43ffca949d57466493da900fa588808f : SUCCESS in 2m 28s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/14b4186c194e47eca06e3cdc60b9f101 : SUCCESS in 12m 21s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3167d71f8ce54354a10be23f7ff6d02c : SUCCESS in 11m 28s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d501fc249b1749988a2f855687ce38be : SUCCESS in 10m 08s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2aea356348343a18a35b73b1d7685ed : SUCCESS in 9m 57s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df040a621b6b421b85c563ee5c4c44d7 : FAILURE in 15m 52s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/806da13c35e54171a90851bb2083dd41 : FAILURE in 14m 57s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c44e0dba0924190917f2c515e979a55 : FAILURE in 12m 16s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc4c0cb258db4ebab25eb4723aac224c : FAILURE in 19m 06s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e08b6d541ac44d3cbbf05376f0622788 : FAILURE in 13m 33s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a3e260ac18284359b6eefcd87a1e6e8b : SUCCESS in 20m 08s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/08ce37e8ceac4175bee7533a03c7da7b : SUCCESS in 1h 04m 41s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d36edbd7ea546fead3f4297f8b322c8 : SUCCESS in 1h 10m 37s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/abd2894210664d4e939602ad20760fff : SUCCESS in 1h 08m 57s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ceb3910cbd64b7ab6fb7f6a4a1d9453 : SUCCESS in 53m 23s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/edd81fd5a8744c179e504dc0c70de454 : SUCCESS in 49m 17s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c20c646c5d84153bd70749676c16765 : FAILURE in 9m 22s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/96ffc4ad5ed0400b896d3bb0ada5171b : SUCCESS in 21m 06s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f446630c18c47d1900a6539f189d505 : FAILURE in 26m 25s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d35d7688e574c898c035efb9198ef09 : FAILURE in 25m 53s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/4621100f65024fc790f01a18b5e679c5 : SUCCESS in 2h 28m 02s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/98d2f43cfe924507b3c438c0ab0a2d51 : SUCCESS in 53m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/695fc738fdde443e8b655a56c33bc5ba : SUCCESS in 4m 09s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/02de1350145540ada00a0b2be1f3cf57 : SUCCESS in 1h 13m 33s","accounts_in_message":[],"_revision_number":1},{"id":"1377a97cb57f1f3902d747c9aec0e8de51f22d27","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-11-25 13:47:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"30f1a1c7afe600134c28f8deba253240ae2b6f7f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-11-29 17:01:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d4cdc85b9c468836ceae24f27c645332c4cc57f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 17:02:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c82a8a35aecdd667e149bb483ab1604b25c3a3e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 19:22:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b3d0eb6ceb64de3bc1ce5cb4dce19ea\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/179a6d28acce41698ae6137ddf92698e : SUCCESS in 3m 08s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca87c62a9f11493cb9d63c439e38ae5a : SUCCESS in 3m 04s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d168402660d347339f48b14f3bf94792 : SUCCESS in 4m 26s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/33b5bbdccef34fb09bacb44dfe34a98d : SUCCESS in 18m 36s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3d2f28aff6349ff861dadd9ace9e12a : SUCCESS in 13m 24s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44f17ba3a9544ce8af455b0ca62e8d14 : SUCCESS in 10m 57s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/79fedbf742784f169c4fda7df24d8498 : SUCCESS in 57m 05s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e0fbc94807f5457bb293ffe5d71251b1 : SUCCESS in 1h 05m 38s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f854e66aafe54594ae5cd80a616343a0 : SUCCESS in 1h 02m 45s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a9dee432874a4e5e8fd9e4b9a0a5a01e : SUCCESS in 50m 50s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ba80db373db34182841ce70ea7c3c0db : SUCCESS in 52m 03s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/d56c59a260cf49c0a2a186735c4c15da : SUCCESS in 21m 39s\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/66f3c52517ff46b1b9ecc8e4a2fc9c06 : SUCCESS in 2h 17m 13s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/77b8e80cbcbd4261ae3f75d420f167c9 : SUCCESS in 42m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a7f6180d3b04beebaed5493593dc2c5 : SUCCESS in 3m 38s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a47c919c56be445eadabf7924841bdc7 : SUCCESS in 1h 09m 10s","accounts_in_message":[],"_revision_number":1},{"id":"1e9488f9b70a72401e68a77d945d142951eb6fe1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 19:22:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6c77e63bb5ae63d611d86da7a1a46a273c67d082","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 19:23:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e60301cae74a498c985fd000340724d4\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2bddc37d96094e79a27b7bca6c98535a : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"55c77ffd34c7d9aac618432a58fc17ed94daaa6e","revisions":{"55c77ffd34c7d9aac618432a58fc17ed94daaa6e":{"kind":"REWORK","_number":1,"created":"2022-11-25 01:42:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/865614/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/14/865614/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/14/865614/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/14/865614/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/14/865614/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/14/865614/1"}}},"commit":{"parents":[{"commit":"8011bfcab10728d688f6259f9c6c28eeea529e51","subject":"Merge \"Remove support for networking-vpp plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/8011bfcab10728d688f6259f9c6c28eeea529e51"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-25 01:42:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-25 01:42:18.000000000","tz":540},"subject":"Remove deprecated mech_driver defined resource type","message":"Remove deprecated mech_driver defined resource type\n\n... because it was deprecated during the previous cycle[1] and its\nimplementation is empty.\n\n[1] 069b2226a9d521865725d84fe950558cb7c06036\n\nChange-Id: Ied984b3f5e1515a4daa0c0a58bdf2b85e674f202\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/55c77ffd34c7d9aac618432a58fc17ed94daaa6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/55c77ffd34c7d9aac618432a58fc17ed94daaa6e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[]}
