)]}'
{"id":"openstack%2Fpuppet-neutron~866004","triplet_id":"openstack%2Fpuppet-neutron~stable%2Fzed~I67fc4d9bd3172ac35bae16c75ee8757d114a6d41","project":"openstack/puppet-neutron","branch":"stable/zed","hashtags":[],"change_id":"I67fc4d9bd3172ac35bae16c75ee8757d114a6d41","subject":"Remove wrong deprecation mark of project_* properties","status":"MERGED","created":"2022-11-30 09:28:11.000000000","updated":"2022-12-01 04:21:09.000000000","submitted":"2022-12-01 04:21:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"866004","cherry_pick_of_change":865617,"cherry_pick_of_patch_set":1,"meta_rev_id":"a56cdc0c5cc846c982e2878afba0cbd860d6baf7","_number":866004,"virtual_id_number":866004,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-01 04:21:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-12-01 01:40:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2022-12-01 01:40:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-30 11:28:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aa752b8a8c440e18fa871652339dea0776d03e03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-30 09:28:11.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9b503a58e09c47ef9791261943dfdd6eff6d089a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 11:28:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a342ed38c01b4c279dab0fbc60c1e8e9\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25075390b91347598be070a580450e52 : SUCCESS in 2m 49s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a60e1b6779f54ee391300f3f6758965e : SUCCESS in 3m 01s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad6054125170427aa6ff0acac0994f93 : SUCCESS in 2m 33s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/c30f1f92763c483888d7f0b5bade5d52 : SUCCESS in 15m 44s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a6cc114d67f4fb4a9e5e4fcb31e10c5 : SUCCESS in 12m 52s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/259b31fb747b4b139a9e4886ed6af5c6 : SUCCESS in 12m 56s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/825d0d9b8c084da0a40d48ffcccd8880 : SUCCESS in 12m 37s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d03034950bc644a99dd45488b0887c02 : FAILURE in 17m 59s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6bdd6d44f9394b08ba99aced12bfacb3 : FAILURE in 18m 12s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce7de1fe1e854647b2a7233ae8fcce00 : FAILURE in 17m 19s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee0f01e00eb146a991a5f7d795624495 : FAILURE in 19m 04s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce2dc76c3e234e8591c1b880602f0142 : FAILURE in 15m 30s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/219a913cb5814028a074583166de2eed : SUCCESS in 20m 08s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6037b1d3055a46d3bc63b3dd3f9a1532 : SUCCESS in 1h 03m 29s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/99299847726a4534a860cd778e1faeba : SUCCESS in 1h 07m 02s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/67f22e64ac58403595cc9211332b2d54 : SUCCESS in 1h 05m 05s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1d9aa4ac0748494582aebbc11a9b0504 : SUCCESS in 59m 12s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/95d87d1d3f32428d9067e9fab47be5f1 : SUCCESS in 52m 32s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a67569b8348f4c9ba0f7b88d6382a0e1 : FAILURE in 11m 49s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/9a5df6d1c47042ed9cba2cf63a57d8d2 : SUCCESS in 21m 52s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e98ee0d324db466f995a83eb3e7076b7 : FAILURE in 26m 40s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b994e721f294992a33d5b1f9a031593 : FAILURE in 25m 55s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/b2edf042ce7a4c06b35563dfa2c432a3 : SUCCESS in 1h 56m 07s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/5f427a1868bd41598668067088f1ea1d : SUCCESS in 42m 46s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/713f40d4e7c54d0ab88fd1f0782d7ff0 : SUCCESS in 58m 50s","accounts_in_message":[],"_revision_number":1},{"id":"03b8157d8fd3abf0e47febf308c02d60252f9577","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-12-01 01:40:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"01644b5d172f120caf58e758c07b1e40accda5b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 01:41:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"97030de713a5ef80b8cf51a8253759aec7477cfd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 04:21:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/695c377f7d814a9389f9dcfdd78455ca\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b97c6742a4b437fb15304ab0a41ac2a : SUCCESS in 2m 34s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a5087844ee5a4ddcb95be73bb4fcc041 : SUCCESS in 2m 41s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1110f4f5091f4abea434a35853bbc1df : SUCCESS in 2m 39s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/0cc0ab2916e04c808a7fdc150dc56c11 : SUCCESS in 15m 04s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9974308fbfd4972aa052f0889ed52f4 : SUCCESS in 11m 13s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/135fdbe013384dddb29e952bd06b1f6c : SUCCESS in 11m 39s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4e88304decdb4486ac3cd0133740186a : SUCCESS in 59m 20s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e5c8df8445c74b71b03406c2cdf166ad : SUCCESS in 1h 14m 42s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/85e0bdebe455436eb86cec84e51e1711 : SUCCESS in 1h 02m 04s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ecd0d33d851a4791b6b316823488f549 : SUCCESS in 57m 20s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7fadadd217a04741b66598dd2b19055b : SUCCESS in 52m 33s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/cb1277faa25e4af18d020c45279e826f : SUCCESS in 21m 08s\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/a437a03921de472a9007dc2bbc5003e1 : SUCCESS in 2h 35m 51s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/62161831c92743479beb5c7670cb8a00 : SUCCESS in 50m 17s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/80aca07ac53144d0a1ddeac73f07f009 : SUCCESS in 1h 20m 16s","accounts_in_message":[],"_revision_number":1},{"id":"a56cdc0c5cc846c982e2878afba0cbd860d6baf7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 04:21:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4fdb0798a6ea07f71ce9e33bf1328110e73c9106","revisions":{"4fdb0798a6ea07f71ce9e33bf1328110e73c9106":{"kind":"REWORK","_number":1,"created":"2022-11-30 09:28:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/04/866004/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/04/866004/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/866004/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/866004/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/866004/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/04/866004/1"}}},"commit":{"parents":[{"commit":"9481426efbea76b5d93dbd3d4661c8f0151a8a0c","subject":"Update TOX_CONSTRAINTS_FILE for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/9481426efbea76b5d93dbd3d4661c8f0151a8a0c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-25 03:09:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-30 09:28:11.000000000","tz":0},"subject":"Remove wrong deprecation mark of project_* properties","message":"Remove wrong deprecation mark of project_* properties\n\nThe project_name property and the project_id property are not meant to\nbe deprecated because these replaces the old tenant_* properties. This\nchange removes the (DEPRECATED) marks erroneously added.\n\nChange-Id: I67fc4d9bd3172ac35bae16c75ee8757d114a6d41\n(cherry picked from commit 7c752068f1d747f1218a54ed3f1b07ebee970d1b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/4fdb0798a6ea07f71ce9e33bf1328110e73c9106"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/4fdb0798a6ea07f71ce9e33bf1328110e73c9106"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[]}
