)]}'
{"id":"openstack%2Fpuppet-neutron~867016","triplet_id":"openstack%2Fpuppet-neutron~stable%2Fussuri~I9a93b65aa5dbf4bca934aec82aad2f7f6ab3d73c","project":"openstack/puppet-neutron","branch":"stable/ussuri","topic":"ovn-tunables","attention_set":{},"removed_from_attention_set":{"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2022-12-12 00:37:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9a93b65aa5dbf4bca934aec82aad2f7f6ab3d73c","subject":"ml2+ovn: Add more tunable parameters for OVN DB connections","status":"MERGED","created":"2022-12-08 18:22:50.000000000","updated":"2022-12-12 00:38:46.000000000","submitted":"2022-12-12 00:37:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867016-ovn-tunables","cherry_pick_of_change":867015,"cherry_pick_of_patch_set":1,"meta_rev_id":"239fbdc5ff2813c0cd4a1cbb4e2596f2454c1f46","_number":867016,"virtual_id_number":867016,"owner":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-12 00:37:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-12-09 20:20:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-12-11 22:34:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-12-11 22:34:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-08 18:22:50.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2022-12-08 19:30:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-09 20:20:26.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-12-11 22:34:32.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"98e2ea79e1ab18b8d9c13bdfdf64a932d7e96eda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2022-12-08 18:22:50.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"81faf1244739de8713b28b540a86cd573be88d54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-08 19:30:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/155b406f57364a8c80c36d45c4bff9f8\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ed990fd270aa4fcba8e6c7d3de7f375d : SUCCESS in 2m 32s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9b183960dd2f47779e99310d14e16e00 : SUCCESS in 2m 33s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d09e9be2f7664caeadde6f6ccc509b1a : SUCCESS in 2m 31s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/e36ab20b5cff4e1fb342953bc8f6d598 : SUCCESS in 13m 19s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/66d5a714f14549b1a29f06fc128a446c : SUCCESS in 14m 04s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/21f5e7a19e684d31bc47ec1688c4dd72 : SUCCESS in 9m 18s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e7f07a33b92c4d3cbfafab33348f4dcc : FAILURE in 12m 26s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/fe22b4bad2d2479cb3be0353aba7b9d8 : SUCCESS in 51m 19s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/007bf2d12eb04e75be63b904d23d7c5f : SUCCESS in 54m 06s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7a337b5d1c484859be24a8ca8bacb40b : SUCCESS in 42m 46s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/fb1407d2b6d94f8b9395da0acb0ef1c3 : SUCCESS in 38m 35s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/230683a76cea4c97843b6ea41e46fa20 : SUCCESS in 55m 06s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f163582b2d4344f4b77349ace43708c2 : SUCCESS in 1h 06m 02s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9bd38fcbcada4c83909e90a4f2b8b8bc : SUCCESS in 1h 03m 15s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0cfbc816f07a49e5b55e52575cfa10cc : SUCCESS in 45m 04s\n- puppet-openstack-beaker-centos-8-stream https://zuul.opendev.org/t/openstack/build/34a904b7a3fc42f290a603236e68bb45 : FAILURE in 26m 26s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e1b9519a60224cd2826011f4da4443cb : FAILURE in 5m 36s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c13a78431b6e4869b890857c20e67fb7 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":1},{"id":"d4fa54ef8b374f6a374053e85210fcf14b989ed7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2022-12-08 19:36:49.000000000","message":"Topic ovn-tunables-stable/victoria-stable/ussuri removed","accounts_in_message":[],"_revision_number":1},{"id":"eb2d1313411e7efc988156cdf23bfd28f67f0c38","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2022-12-08 19:36:54.000000000","message":"Topic set to ovn-tunables","accounts_in_message":[],"_revision_number":1},{"id":"261b4ecc1339c3453c87d8e39456027f4834c5d2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-12-09 20:20:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7f8b514d646a0fd4e9a41cfdca0556421fae4811","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-12-11 22:34:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"769a59868aa772f38c71083cbb253147ef84c898","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-11 22:35:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"96b077781341616e83e22884c2474096fb9011cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 00:37:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f29396c9f87e470898de5ca375f5fcfe\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b5661664cb8e4b56a442bdebd5c6998e : SUCCESS in 2m 45s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/88488d5bf39e4351932fdbef789f04dc : SUCCESS in 2m 31s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f55dbc32ba17452e8b9319845fcbb69b : SUCCESS in 2m 17s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/93cff8e17e114b4c802c4c01a9f503da : SUCCESS in 12m 57s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/6efa66799333429984d6b013c5cb79ff : SUCCESS in 13m 57s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/64373ac5fcb442aca142f62f568f467a : SUCCESS in 9m 52s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/901ddf60538446a6af2f46c3e1ef9a5c : SUCCESS in 54m 46s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/74848a6d9bfe4a6ca9b6f919d66b8be1 : SUCCESS in 1h 07m 57s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ef26e006d7a94dd29c496cdb15d9b5aa : SUCCESS in 58m 06s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0cd2a4d88f6440f4993c4e695eb5d6a3 : SUCCESS in 47m 25s\n- puppet-openstack-beaker-centos-8-stream https://zuul.opendev.org/t/openstack/build/01e8b117e3254c7d8d2a2b4339a50ac1 : FAILURE in 23m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cedb10c16edb4f8a8cfa2f8ddb906132 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":1},{"id":"b510b1ce2ba3eac9ce56ff72104fd6ec68633ee7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 00:37:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"239fbdc5ff2813c0cd4a1cbb4e2596f2454c1f46","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 00:38:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc344867384b4fb783d7feda6f0cffc0\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ef38fd188c2472d87706109daba333f : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e5c307ee0c533fbc9418aa402d2f5548542418eb","revisions":{"e5c307ee0c533fbc9418aa402d2f5548542418eb":{"kind":"REWORK","_number":1,"created":"2022-12-08 18:22:50.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/16/867016/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/16/867016/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/16/867016/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/16/867016/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/16/867016/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/16/867016/1"}}},"commit":{"parents":[{"commit":"0b225b0bd1f786212aaf21b370e7e54525442c66","subject":"mlnx: Allow workaround for duplicate resources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/0b225b0bd1f786212aaf21b370e7e54525442c66"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-09 12:16:12.000000000","tz":540},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2022-12-08 18:22:50.000000000","tz":0},"subject":"ml2+ovn: Add more tunable parameters for OVN DB connections","message":"ml2+ovn: Add more tunable parameters for OVN DB connections\n\nChange-Id: I9a93b65aa5dbf4bca934aec82aad2f7f6ab3d73c\n(cherry picked from commit 85db34491e152ff677c952310206a24270803a84)\n(cherry picked from commit 9d553d51c48c6f9c0ffbf46d409bc8c0590ef4ea)\n(cherry picked from commit 4456238f978d37dbdac8769d3d3db89ed56ef418)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/e5c307ee0c533fbc9418aa402d2f5548542418eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/e5c307ee0c533fbc9418aa402d2f5548542418eb"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
