)]}'
{"id":"openstack%2Fpuppet-neutron~867017","triplet_id":"openstack%2Fpuppet-neutron~stable%2Ftrain~I9a93b65aa5dbf4bca934aec82aad2f7f6ab3d73c","project":"openstack/puppet-neutron","branch":"stable/train","topic":"ovn-tunables","attention_set":{},"removed_from_attention_set":{"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2022-12-12 00:37:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9a93b65aa5dbf4bca934aec82aad2f7f6ab3d73c","subject":"ml2+ovn: Add more tunable parameters for OVN DB connections","status":"MERGED","created":"2022-12-08 18:23:12.000000000","updated":"2022-12-12 00:38:43.000000000","submitted":"2022-12-12 00:37:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867017-ovn-tunables","cherry_pick_of_change":867016,"cherry_pick_of_patch_set":1,"meta_rev_id":"49216adde25d62b1cec72757684fc3654cac7492","_number":867017,"virtual_id_number":867017,"owner":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-12 00:37:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-12-09 20:20:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-12-11 22:34:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-12-11 22:34:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-08 18:23:12.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2022-12-08 20:12:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-09 20:20:27.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-12-11 22:34:16.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"783bccd76233bfa6f9b82dfd8556ebfca5608cd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2022-12-08 18:23:12.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"d461d2df4644cb5731c8fcd6c840a69cf8683bbf","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2022-12-08 19:37:36.000000000","message":"Topic ovn-tunables-stable/victoria-stable/ussuri-stable/train removed","accounts_in_message":[],"_revision_number":1},{"id":"90097e3f1ea517986311d538597e1933aab56240","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2022-12-08 19:37:39.000000000","message":"Topic set to ovn-tunables","accounts_in_message":[],"_revision_number":1},{"id":"14a06058247f93c3f2b11b281f0b9a49192d5eb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-08 20:12:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f16fbdde837480a9a9d840d9136fe8d\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/09a791261eec42dabdc22324f9f7aa5a : SUCCESS in 4m 23s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fc4a819ead304670afad91a310dcb391 : SUCCESS in 3m 21s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/091c363809544582bf935e7ef9e8ae0c : SUCCESS in 4m 01s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/3163ded7ace942b9a44e2b38c0960ffc : SUCCESS in 20m 31s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/2a7f692187e542b5adb82b0981060f99 : SUCCESS in 15m 47s\n- puppet-openstack-unit-6.4-centos-8-stream https://zuul.opendev.org/t/openstack/build/c9272c70833a4cb4b7cb34cd86886ffb : SUCCESS in 14m 22s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8d1426393a524d2faf3d57d65fd0a025 : SUCCESS in 12m 19s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4d1dbe7c15f24fc1b3958b1981af5e55 : FAILURE in 12m 37s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/13d6e042682b4a588467950fe3644707 : SUCCESS in 58m 54s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/63dc9368c9c74e5bb5e75faebca2d498 : SUCCESS in 1h 11m 33s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/44bfd152de594843a1c4beed9d59670e : SUCCESS in 1h 05m 25s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/c0fa5ce9e2df40deb9140e41dd19edcd : SUCCESS in 47m 35s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/690ff3e9f93242cf9ff1c87f4c62e2d8 : SUCCESS in 45m 54s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f12f71ef002c42aaa4d493d6bc3eb1b1 : SUCCESS in 44m 37s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8bf1af7700754118b971017882b1e3a0 : SUCCESS in 44m 53s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/12c8366370b042b385e733e0e7487557 : SUCCESS in 33m 44s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/627b160bb3d34fd39fa153da4728d1d8 : SUCCESS in 1h 03m 02s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a02267b39a6d40028d9d8dd23664b905 : SUCCESS in 1h 01m 01s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/59bd3ef3087e45359e0b7a8c488b46bf : SUCCESS in 56m 01s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/13603fc004da40d68f67e035cd60fc33 : SUCCESS in 42m 38s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/bde1a2bc9c704762b6d9254f19b2ef8a : SUCCESS in 18m 50s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bdc3e1a44a6744398a96d7d8fa19e67e : FAILURE in 5m 11s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0deef8dda80944609ea5e1540a8592b6 : SUCCESS in 3m 21s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/95a820ea6b514a789c81039400020977 : SUCCESS in 1h 47m 20s","accounts_in_message":[],"_revision_number":1},{"id":"ca594fbb93a468cb3c808ade9d8ec9e65762d529","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-12-09 20:20:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5736f171110b7c8760d15ea0e2343404451f1334","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-12-11 22:34:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"362cb7972ec860e88b505324967777c0083c6667","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-11 22:35:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4d0d7279cb348349debca65120234633a59fc379","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 00:37:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df8d97479b8f47e287019176f99024d5\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e76695b5f474d3dad77e55465ad24f8 : SUCCESS in 4m 25s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ba58528922a046cc9cdb183edad6fdf8 : SUCCESS in 3m 55s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/971d1dbad70640dab4e2096b681c218e : SUCCESS in 4m 55s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/4c8c8e33dffe4bf19cd061c88a6f6a4f : SUCCESS in 22m 13s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0233440b65ec4bc5b3d2df68077a33ce : SUCCESS in 15m 00s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/2bb2384022cb4f4db240a051fd9c370b : SUCCESS in 51m 20s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/a44e8c9a9e6749cb967ca22f262d6977 : SUCCESS in 53m 07s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/9bd4fa8b339a4c8cbf23a42dd7e30e62 : SUCCESS in 51m 31s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/248a108f2c8f41f1b084cf423fe785da : SUCCESS in 46m 51s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/5858884186234995a762828eab5bc0e5 : SUCCESS in 19m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2ce3aa314ec43678f3394b9af3bf9b8 : SUCCESS in 3m 08s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/54f00cfc986c4d0abc22829faf81b40b : SUCCESS in 1h 57m 16s","accounts_in_message":[],"_revision_number":1},{"id":"557418c7a0a40050e85a913e6d1f418a5188b70c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 00:37:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"49216adde25d62b1cec72757684fc3654cac7492","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 00:38:43.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/585b0bc544824228b173effa166fc1d5\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0cfe6ed470b74499a9de5bba89a2ef98 : FAILURE in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4c4aa4335dabce0c0e1683d42049cfba498c219e","revisions":{"4c4aa4335dabce0c0e1683d42049cfba498c219e":{"kind":"REWORK","_number":1,"created":"2022-12-08 18:23:12.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/17/867017/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/17/867017/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/17/867017/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/17/867017/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/17/867017/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/17/867017/1"}}},"commit":{"parents":[{"commit":"b2d406629ad1ce4b784e7c9b1be87f570089c468","subject":"mlnx: Allow workaround for duplicate resources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b2d406629ad1ce4b784e7c9b1be87f570089c468"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-09 12:16:12.000000000","tz":540},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2022-12-08 18:23:12.000000000","tz":0},"subject":"ml2+ovn: Add more tunable parameters for OVN DB connections","message":"ml2+ovn: Add more tunable parameters for OVN DB connections\n\nChange-Id: I9a93b65aa5dbf4bca934aec82aad2f7f6ab3d73c\n(cherry picked from commit 85db34491e152ff677c952310206a24270803a84)\n(cherry picked from commit 9d553d51c48c6f9c0ffbf46d409bc8c0590ef4ea)\n(cherry picked from commit 4456238f978d37dbdac8769d3d3db89ed56ef418)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/4c4aa4335dabce0c0e1683d42049cfba498c219e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/4c4aa4335dabce0c0e1683d42049cfba498c219e"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
