)]}'
{"id":"openstack%2Fpuppet-neutron~867018","triplet_id":"openstack%2Fpuppet-neutron~stable%2Fvictoria~Ie0f7d0029412a16e4d9d43f408be97794b3518c0","project":"openstack/puppet-neutron","branch":"stable/victoria","topic":"ovn-tunables","attention_set":{},"removed_from_attention_set":{"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2022-12-12 08:52:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie0f7d0029412a16e4d9d43f408be97794b3518c0","subject":"[ovn metadata] Add more tunable parameters","status":"MERGED","created":"2022-12-08 18:25:02.000000000","updated":"2022-12-12 08:53:13.000000000","submitted":"2022-12-12 08:52:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867018-ovn-tunables","cherry_pick_of_change":849462,"cherry_pick_of_patch_set":1,"meta_rev_id":"3af1777eaa30451059ec0fdfe47e04bc86566520","_number":867018,"virtual_id_number":867018,"owner":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-12 08:52:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-12-12 00:41:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2022-12-12 06:47:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2022-12-12 06:47:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-08 18:25:02.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2022-12-08 20:09:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-09 01:04:18.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2022-12-09 01:09:24.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2022-12-12 06:47:55.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"7dc64b5586ecf1c2bd543aa5b06ae98ab28662bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2022-12-08 18:25:02.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"d6190aae5a476f5f8d87e04a10d0b4303455fc36","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2022-12-08 19:37:29.000000000","message":"Topic ovn-tunables-stable/victoria removed","accounts_in_message":[],"_revision_number":1},{"id":"651b2634c6270c6aa3338984f626f5b5e21d6076","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2022-12-08 19:37:32.000000000","message":"Topic set to ovn-tunables","accounts_in_message":[],"_revision_number":1},{"id":"1fe10e8689ea20cdf1935499356ed486b6411c8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-08 20:09:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d21718bbb3e469e88480a4be8ec0dea\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8eea311689ee4e57a508aca0b963b9ac : SUCCESS in 3m 15s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dded62244fbb40c1bd7dada81c79a681 : SUCCESS in 3m 04s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/24fe066e56c34e088cdd13b31b09b4ad : SUCCESS in 14m 51s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/be50d38d647f42f0a8425bbbcddbd8a7 : SUCCESS in 11m 39s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8cfc798cca3845f2ab83a37416bec929 : FAILURE in 12m 27s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/84a7c3faf1084c06b0e7962f232173ec : FAILURE in 5m 20s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c2eda35a4c154d2c8b7c5d73fc2e5da2 : FAILURE in 5m 08s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/79a4ba52a69e487ea1784c1c5c2e4e9a : FAILURE in 5m 42s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f3d1999653c2452babc89544c3d04514 : FAILURE in 4m 51s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/838798785211432c846bfefeb93e54da : SUCCESS in 1h 08m 38s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b657ec18b9f2408ba7b9a9b011f317f4 : FAILURE in 1h 39m 57s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b1722fe091c1426b874b809c99e18fe5 : SUCCESS in 1h 11m 25s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/185382cd842a42c3b543a92f2c081037 : SUCCESS in 1h 00m 51s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/aa14f7fb507f457f8160a3dd16d4b149 : SUCCESS in 20m 31s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/87c6bd9770a3493698203aa883e34f51 : SUCCESS in 18m 42s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/67cfd89528834a1e9b72811f4d4f94c9 : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":1},{"id":"95c060f727eea6c9bf9171760023799e9b448826","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-12-09 01:04:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b39b5752884c0598660b22f956a3a18e7d083109","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-12-09 01:09:24.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ae2401c18d3f32dcf137c392cbb5bf6c9f744621","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 02:51:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/027dee3c79a64ed0b89ad975e2b44b5f\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/699322595f024d36925209a8ad45b600 : SUCCESS in 3m 31s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eb99459466cf41f2b62c76e166fd4618 : SUCCESS in 2m 49s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/8f0824080d88463999fdf1902c116085 : SUCCESS in 15m 05s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/766397769c3e4a2faed6254068d31b1a : SUCCESS in 12m 25s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/51dc201e13b04fbdb4f1e903bef76e21 : FAILURE in 11m 23s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/19f9b8f9b4a64167b04ac78ef90ec20a : FAILURE in 5m 39s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c2f1267ec2464f46824461717abf8516 : FAILURE in 4m 32s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6e7d98d5c4e54f8cb1c3cb5293cf4eee : FAILURE in 4m 36s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3b0840c7f2804407a14edbea8d78904d : FAILURE in 5m 11s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7f1c81c6b59c491cb09b99a575263b62 : SUCCESS in 1h 03m 22s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/db5bb5451cdd4b1bb1ec9d63ef0bf106 : SUCCESS in 1h 42m 11s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e427bfa666404889a75ced7b35a06bbf : FAILURE in 1h 13m 12s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9ab779e107444b808da280cc008ceda1 : SUCCESS in 1h 05m 48s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/575b573b69274cf2ba369728a967cf0b : SUCCESS in 20m 56s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5fdb504849c04707b85b28c65eaf7457 : SUCCESS in 14m 42s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05ae9f520019499cbb1afd425ce4b219 : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":1},{"id":"82284b5739a2ecbd4411d63cc83a717dfaca9432","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-12-12 00:41:23.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5c3e03b9da091237bdac741f5695a2e8520bb9fd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-12-12 00:41:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c9d68986d229dbba5ed366153e9005dc906af499","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 02:28:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a23a24e1d2c477a830276bbd7de3f55\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b7bbb1e835dc4c44bb0b84863458e03c : SUCCESS in 3m 04s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f9ffeecb036742ec989be13f87bd8894 : SUCCESS in 2m 48s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/ae55c2d0651a409bbf729625f4a8ff94 : SUCCESS in 16m 08s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d58dd51e9744b3499e37ce3944c7002 : SUCCESS in 12m 44s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/432886d6a01445ec9082ada8b298020b : FAILURE in 14m 24s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/79f37d42a4e04173b2aa7d0fb355e419 : FAILURE in 6m 37s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/156e4b3feb104e6b9fa55bf08d80035a : FAILURE in 6m 38s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0ae3e9d50b4049c5a12dbf66934d57fc : FAILURE in 6m 40s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f9feef9c0349429dbd208389a57ed93f : FAILURE in 5m 54s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6c24dbdfdfe340ef8ce6d430868e0473 : SUCCESS in 1h 30m 24s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/114ded1aefbc49588618ecfb5c6172c2 : SUCCESS in 1h 18m 50s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0f66ce7c23754bddb22f436939edafec : SUCCESS in 1h 41m 50s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1989a83f8fff4393934fd6f410c34814 : SUCCESS in 1h 00m 09s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/c14d218363bb4ad9ba00c20bd2dab341 : SUCCESS in 21m 43s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4b694a279d984ce2a5358765d2d573ae : SUCCESS in 16m 11s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5e65f719538437187c455ffb64dedd0 : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":1},{"id":"abd2dc66596fa164f9ffec60b9cd368ab2057dc1","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-12-12 06:47:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"841f6f19b0bdc0dda5a2a4c2a2b3351ce5a2e317","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 06:48:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"466d60511b48fb374589ce689453bdce4980ca89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 08:52:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64287842ef1e4f718d8edf47187337ef\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eb00a8fac8a548b1b62a6c5223107d9e : SUCCESS in 3m 25s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3b76afdfce474238b1662a43e1272bd7 : SUCCESS in 3m 09s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/29e8b3eabc3b4cec811e41da5b94e285 : SUCCESS in 16m 31s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9efc4d5ef4e9442daf4a6a7d8d403dc4 : SUCCESS in 11m 38s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e49605894d324a649086fc2c3ed3809a : SUCCESS in 55m 57s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f420575ebc7044a2b10517d8cc69148a : SUCCESS in 1h 22m 18s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/46be1db8763347c1b22609711e431eec : SUCCESS in 1h 33m 01s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0d42a464ac1e472aaa852d314a4053e7 : SUCCESS in 1h 04m 07s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/d9caa0e968094e27834480f6d55ba7fb : SUCCESS in 23m 54s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/168495acecf2453bae02f6668451208e : SUCCESS in 12m 17s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d18fc5767a6b4b89a21fad8bb7f94f90 : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":1},{"id":"8399d6f7a431c8a4b633c630811569e9a710199d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 08:52:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3af1777eaa30451059ec0fdfe47e04bc86566520","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 08:53:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d6924f9b5764fe8a8ed49eb2f76c6b8\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99e9978bf3674a31a3f7f77587a6f243 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"608d5b362f226efb415b3edf71579249ace8061e","revisions":{"608d5b362f226efb415b3edf71579249ace8061e":{"kind":"REWORK","_number":1,"created":"2022-12-08 18:25:02.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/18/867018/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/18/867018/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/18/867018/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/18/867018/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/18/867018/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/18/867018/1"}}},"commit":{"parents":[{"commit":"29e3ff1ee3f787219328b4c7961a3ee544c91f05","subject":"mlnx: Allow workaround for duplicate resources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/29e3ff1ee3f787219328b4c7961a3ee544c91f05"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-07-08 15:21:15.000000000","tz":330},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2022-12-08 18:25:02.000000000","tz":0},"subject":"[ovn metadata] Add more tunable parameters","message":"[ovn metadata] Add more tunable parameters\n\nFor ml2 conf these are already added as part of [1], these\nparameters are also needed in ovn metadata agent config\nfile, this patch takes care of it.\n\n[1] https://review.opendev.org/q/I9a93b65aa5dbf4bca934aec82aad2f7f6ab3d73c\n\nChange-Id: Ie0f7d0029412a16e4d9d43f408be97794b3518c0\n(cherry picked from commit 68a4d9a53df2e96a7c8cba361fabe03fa908f266)\n(cherry picked from commit 53fd3e801c98a25d0b228aa138c643a3c3f62d3f)\nConflicts: spec/classes/neutron_agents_ovn_metadata_spec.rb\n(cherry picked from commit 3677b7aa1d2e2fea10828918c6b07d82e0f9dca9)\n(cherry picked from commit 9a9bdac5ac863da6a43c09e693057c9a8b59f21f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/608d5b362f226efb415b3edf71579249ace8061e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/608d5b362f226efb415b3edf71579249ace8061e"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
