)]}'
{"id":"openstack%2Fpuppet-neutron~871952","triplet_id":"openstack%2Fpuppet-neutron~master~I163fbf6adb8439bb0077a851337a494c6eaa8919","project":"openstack/puppet-neutron","branch":"master","topic":"ovn-agent","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-03-24 06:26:13.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I163fbf6adb8439bb0077a851337a494c6eaa8919","subject":"WIP: Add support for neutron-ovn-agent","status":"ABANDONED","created":"2023-01-27 05:48:11.000000000","updated":"2023-03-24 06:26:13.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b693ba445bedc4f7b340f43e5256895becb4fd83","_number":871952,"virtual_id_number":871952,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-01-27 16:22:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-27 08:31:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f66af93a464fc713c6766badd54ff09fae7cc2a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-27 05:48:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee9dae5327db6b87c355bea825f153690e63f619","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-27 06:02:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"452721f98114ff1bb5bf5cae1eaec0b6a5288f5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 08:31:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/05afdbb9a2604af2910bcfb5ea4e1732\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f731728e4b6e45d2bca82218535eb2b8 : FAILURE in 2m 40s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69f3bf2686474a27b316349b91ae5578 : SUCCESS in 2m 33s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/bec4c5473b174a5fbb7e0dcc41efb9c4 : SUCCESS in 15m 02s\n- puppet-openstack-unit-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1eb9cf8f6fdb460db178c0fad559cb3b : SUCCESS in 11m 47s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38914b5038ab4a8c8c276350b56c546e : SUCCESS in 11m 29s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9c9aa3869244e62945089d3c946d42a : SUCCESS in 53m 00s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33af4e40a86241f69f492d507d997f0a : FAILURE in 51m 18s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f7eed413437444e90bdcb7d01117750 : FAILURE in 26m 32s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ea47885b39f4b3b9b17283031f4abfb : SUCCESS in 38m 40s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9049c52aa9e4492d922c1e1d29481bcb : FAILURE in 34m 18s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/df41ea2dde4a430b8e96f7c0eafd2db2 : SUCCESS in 19m 47s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a7fa4fe82734131965de95d52029846 : SUCCESS in 56m 20s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/624a5a6c363a4144a558e41f3783711f : SUCCESS in 1h 05m 59s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e05ad1d4d04e4544b4e478ad59fda545 : SUCCESS in 58m 34s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/af99b2ff2205408e8fb90543bb356f62 : SUCCESS in 54m 51s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/36e1f0a696914fafabb339bf0f817b7a : SUCCESS in 50m 39s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/013ab72315f74761ab57a5cbfd8c14be : SUCCESS in 22m 26s\n- puppet-openstack-litmus-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a2f909dce454237b12835388490078f : SUCCESS in 15m 14s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b1151df97584371bb5a233b543e15ad : SUCCESS in 14m 30s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/46a8253b66dc4c08b2e778cc0bc21b0b : SUCCESS in 2h 22m 56s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/ee8541b36f2f44c69f3bd83a2c6c70c1 : SUCCESS in 47m 43s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0ac9948d5f724b1a9d1e143420049837 : SUCCESS in 1h 12m 50s","accounts_in_message":[],"_revision_number":2},{"id":"2515bf6e76809442960af87ae3f2e876790ac9cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-27 13:19:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4c425d8db8ec3fb363d0546dfaf7f9f2d4248a02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 16:22:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5be4de8e22694dd3a63824a7b1cfd6ad\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e423ed10ebcb4a0f8b390a0a581ed949 : SUCCESS in 2m 49s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/edf7637ae0b441c78d915f7ded3f2778 : SUCCESS in 2m 20s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/f9802b374a2d4d338ac8aa958952b293 : SUCCESS in 16m 28s\n- puppet-openstack-unit-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3787113c8e9841a98eec9ac2d6680329 : SUCCESS in 12m 08s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/40b47f2bf7144aa79adcb1dd2f094f7f : SUCCESS in 13m 06s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b72687189fb48de86d277705319f853 : SUCCESS in 43m 58s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55db2ab333694aac94a8b30cfa4a428d : FAILURE in 56m 11s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b3773490c4f4558837c8bbb1c91710a : FAILURE in 31m 23s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f8b0392fcc914586a81b0960d5e25c47 : SUCCESS in 38m 28s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3543d11c9bbf473fbae2ee102e1b9de8 : FAILURE in 39m 32s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a6c4049b98804a3f85bab54987b1caa6 : SUCCESS in 26m 00s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/57ea4fb5f9484899a795c2fb3af5c1da : SUCCESS in 1h 09m 49s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/28749a61cac849d5a40df1ad4ce5994c : SUCCESS in 1h 34m 05s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ad781871b3dd4bdd8f0b72acb5e0d8a6 : SUCCESS in 1h 05m 16s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6acda127b1744981a292c54d35ef675e : SUCCESS in 1h 03m 29s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ed46d6c03eb74510a204a3184055c021 : SUCCESS in 1h 00m 04s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/430ef98bb8db49128da52c919831ce2b : SUCCESS in 19m 28s\n- puppet-openstack-litmus-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/efe44192b3744ee79c1fc63b85b48cee : SUCCESS in 15m 08s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/542a6ebc310547a282e01f5a07480e6e : SUCCESS in 14m 24s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/6d923b99130249eaaa318a5bb9c5ae4c : SUCCESS in 2h 46m 51s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/1f7a054efbd64ea4903ba0dc2503ae8c : SUCCESS in 51m 56s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/993733a697a4402ea2c1e5af805f0ef7 : SUCCESS in 1h 27m 43s","accounts_in_message":[],"_revision_number":3},{"id":"b693ba445bedc4f7b340f43e5256895becb4fd83","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-24 06:26:13.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d648a54a70032c51e1088a99ccb4d8daa1baaaff","revisions":{"fde082425c396c8d98f73ff349d96b5ee9b1ce11":{"kind":"REWORK","_number":1,"created":"2023-01-27 05:48:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/871952/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/52/871952/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/52/871952/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/52/871952/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/52/871952/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/52/871952/1"}}},"commit":{"parents":[{"commit":"204261f9f9e8a55ce0d3a163a8f9c6d0de19f006","subject":"Switch to Ubuntu Jammy (22.04)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/204261f9f9e8a55ce0d3a163a8f9c6d0de19f006"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-27 05:47:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-27 05:47:38.000000000","tz":540},"subject":"WIP: Add support for neutron-ovn-agent","message":"WIP: Add support for neutron-ovn-agent\n\nChange-Id: I163fbf6adb8439bb0077a851337a494c6eaa8919\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/fde082425c396c8d98f73ff349d96b5ee9b1ce11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/fde082425c396c8d98f73ff349d96b5ee9b1ce11"}]},"branch":"refs/heads/master"},"e2393729ccbf186a2f9af30c1314d5c912fed6fc":{"kind":"REWORK","_number":2,"created":"2023-01-27 06:02:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/871952/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/52/871952/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/52/871952/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/52/871952/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/52/871952/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/52/871952/2"}}},"commit":{"parents":[{"commit":"204261f9f9e8a55ce0d3a163a8f9c6d0de19f006","subject":"Switch to Ubuntu Jammy (22.04)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/204261f9f9e8a55ce0d3a163a8f9c6d0de19f006"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-27 05:47:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-27 06:02:21.000000000","tz":540},"subject":"WIP: Add support for neutron-ovn-agent","message":"WIP: Add support for neutron-ovn-agent\n\nChange-Id: I163fbf6adb8439bb0077a851337a494c6eaa8919\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/e2393729ccbf186a2f9af30c1314d5c912fed6fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/e2393729ccbf186a2f9af30c1314d5c912fed6fc"}]},"branch":"refs/heads/master"},"d648a54a70032c51e1088a99ccb4d8daa1baaaff":{"kind":"REWORK","_number":3,"created":"2023-01-27 13:19:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/871952/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/52/871952/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/52/871952/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/52/871952/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/52/871952/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/52/871952/3"}}},"commit":{"parents":[{"commit":"204261f9f9e8a55ce0d3a163a8f9c6d0de19f006","subject":"Switch to Ubuntu Jammy (22.04)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/204261f9f9e8a55ce0d3a163a8f9c6d0de19f006"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-27 05:47:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-27 13:19:22.000000000","tz":540},"subject":"WIP: Add support for neutron-ovn-agent","message":"WIP: Add support for neutron-ovn-agent\n\nChange-Id: I163fbf6adb8439bb0077a851337a494c6eaa8919\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/d648a54a70032c51e1088a99ccb4d8daa1baaaff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/d648a54a70032c51e1088a99ccb4d8daa1baaaff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
