)]}'
{"id":"openstack%2Fpuppet-neutron~872290","triplet_id":"openstack%2Fpuppet-neutron~stable%2Fyoga~I3a784588435619418cfd51a1ceffd9fd0ecd5e63","project":"openstack/puppet-neutron","branch":"stable/yoga","topic":"bug/2003324-stable/yoga","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-02-06 17:49:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3a784588435619418cfd51a1ceffd9fd0ecd5e63","subject":"Fix wrong bridge name if bridge_mappings is commma-separated string","status":"MERGED","created":"2023-01-31 11:29:10.000000000","updated":"2023-02-06 17:49:45.000000000","submitted":"2023-02-06 17:49:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872290-bug/2003324-stable/yoga","cherry_pick_of_change":871558,"cherry_pick_of_patch_set":1,"meta_rev_id":"22254687c25cffc4fb8b8fb70bcc4fd5a9340000","_number":872290,"virtual_id_number":872290,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-06 17:49:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2023-01-31 15:09:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2023-02-06 16:21:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2023-02-06 16:21:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-31 13:12:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-31 15:09:43.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2023-02-06 16:21:32.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"7191e3cfcae1101c7392ae60d7344244bbc21a42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-31 11:29:10.000000000","message":"Patch Set 1: Cherry Picked from branch stable/zed.","accounts_in_message":[],"_revision_number":1},{"id":"c03c2f774d3f9204b57a708801c6982a6b793c95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 13:12:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97e7c1a928f54f6ea7be9d88c619a05c\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/17120d31f7e14316b81cecdb4b66c925 : SUCCESS in 2m 58s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72d1047a59d24848a5917c76cd91b207 : SUCCESS in 4m 07s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1cf0716efc5e49d8b43e894b39630fad : SUCCESS in 3m 06s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/5cda81a6f2db4432a92f56c86cf796e0 : SUCCESS in 19m 01s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/567b2e847aae4f6d93e8345c701b2485 : SUCCESS in 15m 08s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3f28adb3cb24643818965e510ed673b : SUCCESS in 15m 37s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d533652a74694f76901faf5893f71e50 : SUCCESS in 22m 29s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a6fe0fba52b4f7781e7d9a0805a2afa : FAILURE in 34m 04s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fcacd0a3d7c94849845edcb4bef68b93 : FAILURE in 57m 55s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fbb6b6bda31e4ddbb70e7d02b1c4a161 : FAILURE in 34m 33s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d4b9dd3b705d4691bc4d757455029195 : FAILURE in 41m 10s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/dd94555e7a0242fb89896379dd85fa3f : SUCCESS in 26m 23s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/22007e30b7d04447ba4763c912354a2a : SUCCESS in 1h 09m 12s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c03d996f1d664f329d5a4955d7ad2955 : SUCCESS in 1h 40m 44s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7742480481554f0a9a27a2ea218f8f4d : SUCCESS in 1h 09m 30s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a6008299f49f4bbdb693585603496abb : SUCCESS in 1h 11m 12s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ad76b82a24a4cb09d8d6a868af82d26 : SUCCESS in 18m 07s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b0ca6311fe6a47bba07bcddbc51b1b55 : SUCCESS in 1h 04m 08s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1f7304f96c884191ac788c6b35719a2c : SUCCESS in 1h 20m 47s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f990d79b27db4059a8d0da0f920f3947 : SUCCESS in 1h 05m 59s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/eae05927159f4ab9bd201c3d20fe4098 : SUCCESS in 53m 24s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0574611c96b460a951693192089aab2 : SUCCESS in 15m 50s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/29326866b8c94cfdaa2778b218128ffd : SUCCESS in 24m 43s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/4bb837a777bf44d8938a406eadb171a1 : SUCCESS in 16m 33s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e313e7901e9f4cd79ac0e0c147ddd193 : SUCCESS in 19m 54s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bca59d7377ee41f8addff1ab78af4ab2 : SUCCESS in 12m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0b806fee30d2481fc289b34472eeabd538c885d3","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-01-31 15:09:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"843f88d5005aaaa19b7ea5ff0990743e3e1d79d5","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-02-06 16:21:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a91dabf7c5fd4b30feddb2310c1a146efa903971","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 16:22:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c581d70a359882c4c3f5f96551bf2cd4cdd592e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 17:49:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51d2d6b65f594f538691047530d304ac\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f06c231af4a4896b5717d565c2738b5 : SUCCESS in 2m 59s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f5dd43204184ba988dc10876811966b : SUCCESS in 2m 54s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a53b69730be54821a3447549f027a100 : SUCCESS in 2m 49s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/adb16acb88cc49209f1fe46f1ec3745c : SUCCESS in 19m 45s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c20a326ef934e8bb312ec060cd7c970 : SUCCESS in 15m 28s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5763d72764c436ba2465ef6f7d0664a : SUCCESS in 16m 26s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d1bbbf243bee440f84cdd45bc4581c21 : SUCCESS in 1h 12m 23s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/11c17ec076d94a99841bf5770135fb4c : SUCCESS in 1h 22m 00s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/19e25088be2d4aea8933fa056b7301d9 : SUCCESS in 1h 16m 24s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0df35a91b4914d2ba002bb017ee3d594 : SUCCESS in 56m 27s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/0ced9d7b8c324b9488616cfbe509f7c6 : SUCCESS in 22m 59s","accounts_in_message":[],"_revision_number":1},{"id":"22254687c25cffc4fb8b8fb70bcc4fd5a9340000","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 17:49:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9540933d918d498b4e1c1d55dc2ae63ac88a948b","revisions":{"9540933d918d498b4e1c1d55dc2ae63ac88a948b":{"kind":"REWORK","_number":1,"created":"2023-01-31 11:29:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/872290/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/90/872290/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/90/872290/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/90/872290/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/90/872290/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/90/872290/1"}}},"commit":{"parents":[{"commit":"7090aa57c7253a4331220b2f152c1c35f07e7ee3","subject":"Remove wrong deprecation mark of project_* properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/7090aa57c7253a4331220b2f152c1c35f07e7ee3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-19 10:22:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-31 11:29:10.000000000","tz":0},"subject":"Fix wrong bridge name if bridge_mappings is commma-separated string","message":"Fix wrong bridge name if bridge_mappings is commma-separated string\n\nThis change ensures the value is properly translated to a list even\nif it\u0027s a flat string so that the bridge name is properly picked up.\n\nThis also fixes a few wrong assertions done in unit tests.\n\nCloses-Bug: #2003324\nChange-Id: I3a784588435619418cfd51a1ceffd9fd0ecd5e63\n(cherry picked from commit 05b043698f1fc7511e3087bea6a4dada515e54a9)\n(cherry picked from commit a2d49ef58e9a31fe6c5a99b8a2b7e63646ea2101)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/9540933d918d498b4e1c1d55dc2ae63ac88a948b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/9540933d918d498b4e1c1d55dc2ae63ac88a948b"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
