)]}'
{"id":"openstack%2Fpuppet-neutron~872812","triplet_id":"openstack%2Fpuppet-neutron~stable%2Fxena~I3a784588435619418cfd51a1ceffd9fd0ecd5e63","project":"openstack/puppet-neutron","branch":"stable/xena","topic":"bug/2003324-stable/xena","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-02-10 17:59:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3a784588435619418cfd51a1ceffd9fd0ecd5e63","subject":"Fix wrong bridge name if bridge_mappings is commma-separated string","status":"MERGED","created":"2023-02-07 01:29:55.000000000","updated":"2023-02-10 17:59:22.000000000","submitted":"2023-02-10 17:59:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872812-bug/2003324-stable/xena","cherry_pick_of_change":872290,"cherry_pick_of_patch_set":1,"meta_rev_id":"357e8827993fd1898f443e3deba6f862f6046eb5","_number":872812,"virtual_id_number":872812,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-10 17:59:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2023-02-07 22:44:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2023-02-10 15:59:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2023-02-10 15:59:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-07 03:36:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-07 22:44:34.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2023-02-10 15:59:47.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"5d1fe5e728005379a7a46c0d771b4f1b4a51deff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-07 01:29:55.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"de775656ebbac100fb1e88a2fcd1bb87e09afcb5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-07 01:30:04.000000000","message":"Topic bug/2003324-stable/yoga-stable/xena removed","accounts_in_message":[],"_revision_number":1},{"id":"2f23b427a48de3bd99215b677fc609dd971c3567","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-07 01:30:08.000000000","message":"Topic set to bug/2003324-stable/xena","accounts_in_message":[],"_revision_number":1},{"id":"268bef35aca401cbc640be79e920d52ea5a203b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 03:36:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a36e3e1562741578e2b6fd0fb8ad531\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/edde6f629ed941818aa7cedfc547dff5 : SUCCESS in 2m 54s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b04f505273b491eaec6a49ab00bb76e : SUCCESS in 2m 59s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c2fae26878c4e2697efe7c01a0a4614 : SUCCESS in 2m 32s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/f5f508d2de3c4fbb97b3f6cc65a25137 : SUCCESS in 15m 09s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/3af2c22e9500476aa474a885a4f61580 : SUCCESS in 14m 54s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0989fced5cb74704908e40627f954f47 : SUCCESS in 11m 27s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3a917a2da8b41f19dbcc42b13b58ad8 : SUCCESS in 12m 52s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1bc19e5f40684a7086cd82b7e719c800 : SUCCESS in 10m 12s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68115432235d47e2b00659d04e92bb6b : FAILURE in 32m 07s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b270f7a7e0d9475abaa718d112a8efd9 : FAILURE in 42m 49s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea363949cf2a4c98a6fded8777d80757 : FAILURE in 34m 21s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8055e6b9681745d1917a9aabde15ccc2 : FAILURE in 33m 13s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/01b44b95053f4f439dbf34d5d8d47e71 : SUCCESS in 24m 39s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/df22595479a14cfaa9e259627cf87986 : SUCCESS in 1h 12m 43s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c6d5a00a60f94580abb12dff8ac5edbd : SUCCESS in 1h 17m 07s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c97abafc89f0424b96aa0f87b98d2734 : SUCCESS in 1h 48m 56s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/20edc8005934491d9f84a585643e8d29 : SUCCESS in 55m 38s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1a5555b71cf476cbe3aa509fd1edc1e : SUCCESS in 17m 29s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/76a6402158df4873b428c2d45340edc5 : SUCCESS in 33m 59s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/896ab78455ac4a7494afcd5121892733 : SUCCESS in 17m 38s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bbe5c6d804a1471da4b67b65f1cca6a4 : SUCCESS in 15m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7eb8b74abc330f7012b1df1bc465d5a7c4950b48","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-02-07 22:44:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6bc9baf3b89a32c00780224d7a8dac6ec31f01cb","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-02-10 15:59:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"df1d93b5f276b1e4251574baf461ec373a2606e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 16:00:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3a71fe11a7d24a72b1e192b0c0dbf6eec57a2640","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 17:59:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a00d9e192f18461f9606aa5736cce752\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e969ef23a2af43d699043d015bcddd85 : SUCCESS in 3m 02s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d61f73e1bbb94bae987611170a21d52b : SUCCESS in 2m 55s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/495f8614aa3644b28c5c074beab6afb6 : SUCCESS in 3m 22s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/6b185e781703444ba9ebef47bb68c1f0 : SUCCESS in 15m 57s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/c692db0b8cdd4aafa7a08f80f35d5547 : SUCCESS in 17m 30s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7521b9cb0eeb4db09dd748179292f75d : SUCCESS in 13m 18s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/325074f8c5af4eb69af9012aef6f3ae8 : SUCCESS in 13m 27s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/734a5772952d4ac98289cb74df6ab3dc : SUCCESS in 55m 13s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b52ec23f87154297a03c8ca2afed2ce2 : SUCCESS in 1h 05m 20s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/551a5dbf35ce4543a22ad9c0e1721923 : SUCCESS in 1h 12m 35s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/390472260bc942fb83a1377cad534168 : SUCCESS in 1h 02m 02s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/2f9ae167c584406a97ba3352c8cf4223 : SUCCESS in 28m 00s","accounts_in_message":[],"_revision_number":1},{"id":"357e8827993fd1898f443e3deba6f862f6046eb5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 17:59:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4a41c221fcec82dc25b099a94180eb65734efa0e","revisions":{"4a41c221fcec82dc25b099a94180eb65734efa0e":{"kind":"REWORK","_number":1,"created":"2023-02-07 01:29:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/12/872812/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/12/872812/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/12/872812/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/12/872812/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/12/872812/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/12/872812/1"}}},"commit":{"parents":[{"commit":"5948c7101649b49cf532ba69599575750d5aa797","subject":"setuptools: Disable auto discovery","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/5948c7101649b49cf532ba69599575750d5aa797"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-19 10:22:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-07 01:29:55.000000000","tz":0},"subject":"Fix wrong bridge name if bridge_mappings is commma-separated string","message":"Fix wrong bridge name if bridge_mappings is commma-separated string\n\nThis change ensures the value is properly translated to a list even\nif it\u0027s a flat string so that the bridge name is properly picked up.\n\nThis also fixes a few wrong assertions done in unit tests.\n\nCloses-Bug: #2003324\nChange-Id: I3a784588435619418cfd51a1ceffd9fd0ecd5e63\n(cherry picked from commit 05b043698f1fc7511e3087bea6a4dada515e54a9)\n(cherry picked from commit a2d49ef58e9a31fe6c5a99b8a2b7e63646ea2101)\n(cherry picked from commit 9540933d918d498b4e1c1d55dc2ae63ac88a948b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/4a41c221fcec82dc25b099a94180eb65734efa0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/4a41c221fcec82dc25b099a94180eb65734efa0e"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
