)]}'
{"id":"openstack%2Fpuppet-neutron~872871","triplet_id":"openstack%2Fpuppet-neutron~stable%2Fzed~I4cec5a4f380a6d839c4ff06d218fa490a3c2044f","project":"openstack/puppet-neutron","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-02-10 19:56:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4cec5a4f380a6d839c4ff06d218fa490a3c2044f","subject":"Restart sriov-agent after updating numvfs","status":"MERGED","created":"2023-02-07 08:19:37.000000000","updated":"2023-02-10 19:56:54.000000000","submitted":"2023-02-10 19:56:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872871","cherry_pick_of_change":872613,"cherry_pick_of_patch_set":1,"meta_rev_id":"f60f58b956b54ef690461b64cbb6060603a90525","_number":872871,"virtual_id_number":872871,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-10 19:56:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2023-02-07 22:44:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2023-02-10 16:02:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2023-02-10 16:02:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-07 11:17:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-07 22:44:36.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2023-02-10 16:02:04.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"28bd49407a8d3ac3f9a703970010a725e4ac32c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-07 08:19:37.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b8ebb687051fe8fa00f039b9bd8a04b4a9ce21ae","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-07 08:19:44.000000000","message":"Topic bug/2004600-stable/zed removed","accounts_in_message":[],"_revision_number":1},{"id":"7f0b08c14de80dc796dcc04e31f06dafd36cfd5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 11:17:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e761c9601d44a3c8431519380575843\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9acb9afa896d4a76ac99ba8d1ca60a2a : SUCCESS in 3m 00s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8304c641534f40e583fca25e9f2c80b0 : SUCCESS in 2m 58s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88fb248ae0f041988cbda4bd715bcf99 : SUCCESS in 2m 46s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/ae8f8cb1f43a4f0287e483e01f6ae733 : SUCCESS in 15m 48s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b4fd6c16e064d3dbd49ed1c56b721f9 : SUCCESS in 14m 05s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9073ccd8926645f4926c1566db3c97c2 : SUCCESS in 12m 26s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a8f98a02aac4b748273a03a77a0d2fe : SUCCESS in 13m 18s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a706e1315a394847badefc152950d340 : FAILURE in 24m 12s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee7b9b0d96a84633a37a7e06210bf183 : FAILURE in 22m 08s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/165fde29ec2d496482ebe2ce41bec507 : FAILURE in 14m 51s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/251247f21de34a42b6bc9886c5534039 : FAILURE in 18m 44s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4da9b7a06aa04161b5501541e5cfa321 : FAILURE in 16m 53s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9e6bc813fc464878a4b3af500af5d9f4 : SUCCESS in 22m 17s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1c28a40eec9947ea8c4c787c72753525 : SUCCESS in 1h 01m 15s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/969594af79124c9495bf3331267ee16b : SUCCESS in 1h 08m 41s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/df4bdc97ceb746798fdaaa4859bf111d : SUCCESS in 1h 00m 05s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b2d6f28645a84a7a83690e00c45615a3 : SUCCESS in 55m 48s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/80fd155249bc4680a7f3b56cf5090ab6 : SUCCESS in 55m 31s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa4e561ac2eb4833bec4fd44d26fd7d0 : FAILURE in 13m 36s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/831e88b462a74615ba4bc8debe73b803 : SUCCESS in 19m 37s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b93460bc701f454da14a4fb55e644473 : FAILURE in 45m 08s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/50467aed645b431b8ca574117ac66c0e : FAILURE in 30m 29s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/0038e921fdf04c5aa8de484aa9e77498 : SUCCESS in 2h 33m 27s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/5731ce9b2b474d06a11bde902200d785 : SUCCESS in 48m 40s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b863c7081d78406a8181ec88d6114329 : SUCCESS in 1h 11m 29s","accounts_in_message":[],"_revision_number":1},{"id":"fc46dc461e57c9c0ddcea05abb0483a670be5019","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-02-07 22:44:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"747f58a8d0959d289d6e1e9fa3b8b0bd34ee9ab9","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-02-10 16:02:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2bd4ef256ac97fd0f0c2086b41bf336cb5b0337e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 16:03:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8128a8b2aff669dd98df676bcb0d1d357dc6cb68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 19:56:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e719e35505b4ad18cd5492055f2cc9d\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1425650984c041108304d68140f8847c : SUCCESS in 3m 00s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69afa9ad9e194c698f11f620586ad411 : SUCCESS in 2m 41s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c8985c1cc4441a292eed7a3cb6ba436 : SUCCESS in 2m 46s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/1e83e4e9f08042cb96195548a6d7a1c1 : SUCCESS in 17m 27s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9736c380c1634a56b04bae91e71a0c46 : SUCCESS in 12m 46s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a81eeee4f7f4674925488f05d531028 : SUCCESS in 12m 59s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/945bcfbbe3674e2e96e126b1f0d4fd79 : SUCCESS in 1h 00m 15s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3331224176cd406baf659add1cd6b921 : SUCCESS in 1h 13m 01s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ccaa263efcf847a98dc801932b340fcc : SUCCESS in 1h 06m 16s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a6e520a7825b4f538b27eabdaaff3d79 : SUCCESS in 51m 35s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/70bcf045a613453ebd55d8a8e230833b : SUCCESS in 57m 34s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/2d481c9a137b4997a2734a5616a6dd25 : SUCCESS in 21m 30s\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/21dea287e0514b4b9fd31aedd23c43da : SUCCESS in 3h 27m 19s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/b3901c43f95f4c83a05c7f69be446f66 : SUCCESS in 52m 13s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0c8bf16b20fd4dff91fee888c9d208a4 : SUCCESS in 1h 28m 52s","accounts_in_message":[],"_revision_number":1},{"id":"f60f58b956b54ef690461b64cbb6060603a90525","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 19:56:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2ee997136ba23617aa73bc2cf3e269942a81a259","revisions":{"2ee997136ba23617aa73bc2cf3e269942a81a259":{"kind":"REWORK","_number":1,"created":"2023-02-07 08:19:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/71/872871/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/71/872871/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/71/872871/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/71/872871/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/71/872871/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/71/872871/1"}}},"commit":{"parents":[{"commit":"a2d49ef58e9a31fe6c5a99b8a2b7e63646ea2101","subject":"Fix wrong bridge name if bridge_mappings is commma-separated string","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/a2d49ef58e9a31fe6c5a99b8a2b7e63646ea2101"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-03 01:18:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-07 08:19:37.000000000","tz":0},"subject":"Restart sriov-agent after updating numvfs","message":"Restart sriov-agent after updating numvfs\n\n... so that the agent detects the updated number of VFs.\n\nCloses-Bug: #2004600\nChange-Id: I4cec5a4f380a6d839c4ff06d218fa490a3c2044f\n(cherry picked from commit c588af82301ff872b70232a54e2e633f910d080c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/2ee997136ba23617aa73bc2cf3e269942a81a259"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/2ee997136ba23617aa73bc2cf3e269942a81a259"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
