)]}'
{"id":"openstack%2Fpuppet-neutron~876736","triplet_id":"openstack%2Fpuppet-neutron~master~I5baa7e7dfac33e267790525dc1e8b963c3166cc0","project":"openstack/puppet-neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-03-10 09:23:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5baa7e7dfac33e267790525dc1e8b963c3166cc0","subject":"Add support for [ovs] of_listen_address/port","status":"MERGED","created":"2023-03-07 14:10:36.000000000","updated":"2023-03-10 09:24:22.000000000","submitted":"2023-03-10 09:23:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876736","meta_rev_id":"f1286663beaf3acb80a91d12d56afdbafc38a39e","_number":876736,"virtual_id_number":876736,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-10 09:23:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2023-03-09 14:41:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2023-03-10 07:26:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2023-03-10 07:26:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-07 16:27:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-09 14:41:24.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2023-03-10 07:26:34.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"3445d8b3dd9d8fe901a758c04c7ad1ccd147a6b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-07 14:10:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"344d3070086a2dee64977b04de01f946b22a741b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 16:27:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c0de4cbf8ca4555b9bcd2293154dc90\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/afb5988ab7914e63b5e0ad486b0bc954 : SUCCESS in 3m 09s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/735bec406ecb4491bab4f45689b99621 : SUCCESS in 3m 04s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/cb90945abc4f44709c21334d11cf3f9a : SUCCESS in 14m 52s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/258140132e7d4ff4b1f3ae0410e85c7d : SUCCESS in 11m 43s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2622a7531fb4af88951bdd98cd9f61a : SUCCESS in 12m 44s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ece283fd43624bd6a6422ff1738bfa31 : SUCCESS in 47m 23s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cdac319f60ce4257b17e1e1c1c693dcf : FAILURE in 51m 19s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc076ffeee9f42998963eabc8069eb6d : FAILURE in 32m 39s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/400b4ed07e234c1fb9fc4dceb42f05bd : SUCCESS in 48m 21s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8fc39495c71944f39e0112985fed9eb4 : FAILURE in 41m 08s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/64d1435a6ac5498bb599550a8a2aa1fd : SUCCESS in 20m 19s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/68baf7b485d24ac08381671ea8bb10bb : SUCCESS in 49m 18s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/770f135eb5df4cd8bbec5e1646c1ff4c : SUCCESS in 1h 17m 22s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b0dd0ff4157c416da5637fe1b382d4ce : SUCCESS in 1h 08m 03s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/94e440e86dbc49ad91e5a2f1ad7d341b : SUCCESS in 45m 39s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/982c4918df374a539a55593f70aafcc1 : SUCCESS in 52m 01s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/cf65372849cd401d9574234e22130561 : SUCCESS in 18m 04s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/20b3c368172745d98d1d57b7f3be3de2 : SUCCESS in 13m 39s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f7f1c8c9e094242aaa7adabb9adeec9 : SUCCESS in 13m 37s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/e1c86acfb39b45228bb572f5b340410f : SUCCESS in 1h 59m 56s (non-voting)\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/46458b0fd9ee40e8b163573fdc5c1eee : SUCCESS in 38m 33s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fed767dad26940a7be2efee3a15e1912 : SUCCESS in 3m 04s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8d24baa89cd04a9ba63a0d55f0d011db : SUCCESS in 59m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3306e2f3a825982477855adcc52594df5bb28eb0","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-03-09 14:41:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"83d012cb5ddf7b834307cb73a06e51c7ca8b06d6","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-03-10 07:26:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"00c3c34f52b29de7bfadb97494a95dbab6a7be61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 07:28:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e9825a2b602b4339c16e4ec523ef01fc1944f085","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 09:23:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d996dafb9e6410c9545210e8117514c\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9beb86e69b264bc2af776f80e7f7bc4b : SUCCESS in 2m 26s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/867d4805f3c94b69af6344b614304a5c : SUCCESS in 2m 38s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/95190e28c4f945e79e9f414c2b253f39 : SUCCESS in 12m 33s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d07b72ee2324d078a0707b9d1bd355a : SUCCESS in 12m 47s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7ac9ff3fc63542d0904c5ea7e186de88 : SUCCESS in 1h 03m 17s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/afaefcd4f2364134ae566aa5ab2cbd0c : SUCCESS in 1h 05m 39s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6746aca80cd64c278b79c271c231fc72 : SUCCESS in 1h 01m 41s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/246038c9427b4eff8a24d7e55fe9d3ca : SUCCESS in 54m 20s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0b1ba407d42040c2993bbc4734ff9161 : SUCCESS in 45m 11s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/32655f6dbf1d417096a98735f88c613b : SUCCESS in 18m 45s\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/e15695ef5e6d44febf06f294b9fe312a : SUCCESS in 1h 54m 29s (non-voting)\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/a8dec2bcea6a45aeacc13302bae9dc1f : SUCCESS in 58m 13s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d69211d391d441ea8bf35b9f9618e60 : SUCCESS in 3m 05s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5bf22fa22cc5463e89f3eb3b82a75e93 : SUCCESS in 55m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"63a1326888fe8fc7a284c2a8838cc17ebc6b214a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 09:23:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f1286663beaf3acb80a91d12d56afdbafc38a39e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 09:24:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77f878a740124d459b4800c957ec820e\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96a67a1ac7eb477e972ba5b8944a4dde : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ac2dbf94c9d46872cd0e6f9eb9ad84306afd28d9","revisions":{"ac2dbf94c9d46872cd0e6f9eb9ad84306afd28d9":{"kind":"REWORK","_number":1,"created":"2023-03-07 14:10:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/36/876736/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/36/876736/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/876736/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/876736/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/876736/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/36/876736/1"}}},"commit":{"parents":[{"commit":"4f9ea57cb295fce0c579b8af8a2f4ac5270e4631","subject":"Replace legacy facts and use fact hash","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/4f9ea57cb295fce0c579b8af8a2f4ac5270e4631"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-07 14:10:04.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-07 14:10:04.000000000","tz":540},"subject":"Add support for [ovs] of_listen_address/port","message":"Add support for [ovs] of_listen_address/port\n\n... so that we can use IPv6 address instead of the default IPv4 local\naddress.\n\nChange-Id: I5baa7e7dfac33e267790525dc1e8b963c3166cc0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/ac2dbf94c9d46872cd0e6f9eb9ad84306afd28d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/ac2dbf94c9d46872cd0e6f9eb9ad84306afd28d9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
