)]}'
{"id":"openstack%2Fpuppet-neutron~877189","triplet_id":"openstack%2Fpuppet-neutron~master~Ib1d2a70a8e40680ebb582cbf4e59a0c018c21447","project":"openstack/puppet-neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-03-14 01:00:32.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ib1d2a70a8e40680ebb582cbf4e59a0c018c21447","subject":"Remove tenant_* properties","status":"MERGED","created":"2023-03-13 01:54:37.000000000","updated":"2023-03-14 03:31:55.000000000","submitted":"2023-03-14 03:30:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877189","meta_rev_id":"de202384567aa3690a05fd2b3c7b30b297c9c310","_number":877189,"virtual_id_number":877189,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-14 03:30:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2023-03-13 08:52:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2023-03-13 19:43:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2023-03-13 19:43:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-13 05:45:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-13 08:52:00.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-03-13 19:43:27.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"baab09decf850b20f15fce5615044ea3bf1e1645","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-13 01:54:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b48c5a4b688750989131f8fda1f1a6acdb87bd22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-13 03:09:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"563acfa477729a98281f6aa423e7c01e81909c20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 05:45:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08a83da779e846b9bfa57ceb8a6adf22\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f2b7bf0193e49aaa2e76765014d3fca : SUCCESS in 2m 41s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09b3ed0bc70f4514b463dd4d6b264e46 : SUCCESS in 2m 44s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/0d445b6059d64bc9ad48c19eec31890e : SUCCESS in 15m 16s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c432e4204934822b3abb76306c52734 : SUCCESS in 12m 30s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c018fe3ae7b48eda53d49520e1e347e : SUCCESS in 11m 55s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd9b74eda86d4d4f9621b51683f980bc : SUCCESS in 50m 57s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f212b6915fa0471a8686742c661e52d6 : FAILURE in 48m 51s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d5e5e9a53a134a1c8f330cb76741c855 : FAILURE in 39m 25s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb879de97af74d2c85dd0d99ab43e43e : SUCCESS in 38m 44s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f4ee0f9219f64157a98a9e3756acc6b9 : FAILURE in 42m 00s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/81bac0d1d04e472bb094d6d4493dade0 : SUCCESS in 21m 43s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a61ddf586fe44459948c3e67e037592 : SUCCESS in 1h 04m 36s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/02a4d7e39c804f61868aca198c7eaef0 : SUCCESS in 1h 15m 21s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3428daa392534d7394fe11fb9f1c45eb : SUCCESS in 1h 01m 33s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cb26aff0fd9b441c93ac39046518610d : SUCCESS in 1h 05m 20s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b1b11cc758634eb98cd80e0ad2f7eeea : SUCCESS in 59m 31s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/2519a1e682524495b092dd303c9faa41 : SUCCESS in 18m 54s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c446a6627fd4f3aa0487f67f6364cf2 : SUCCESS in 13m 19s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dcb3a84a78294f74b3a3bceffb747848 : SUCCESS in 13m 35s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/2b8541e948184f66a19c1a34f06b808b : SUCCESS in 2h 28m 47s (non-voting)\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/8aa55e4675e24fedbc50bac535b4af16 : SUCCESS in 49m 13s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99f700aed1dc43b788a15d7d4db13ba5 : SUCCESS in 3m 20s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/bb0d835661924628b603243b649aee64 : SUCCESS in 1h 15m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"79e7d5eb385e7bd38efaab8538aeb90892eaacb9","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-03-13 08:52:00.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"cf589d51bf7dd2e57084a6c7b07fd30965325bf1","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-03-13 19:43:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"005fff46aadff035a68014301d3560a80d89d946","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 19:44:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8e6ab4302cc705e0c724e508cb6f358308319c3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 22:37:15.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/99fb8dcf62c54f08bd3d0582a542704e\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a10ae0f9834c41d4832178c665498490 : SUCCESS in 2m 41s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/365f4148ebdc4a6aaa8a3a8ab8fa9950 : SUCCESS in 2m 25s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/cf3c1659040e43038ac3acdeb2314bae : SUCCESS in 17m 18s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e95580b98d82468a9fdd67c0988d301f : SUCCESS in 11m 48s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6643ac21c2694f70a13ad8dd6aef8911 : SUCCESS in 1h 09m 53s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/42e879e9643e4d2f81d32aab09065d68 : SUCCESS in 1h 29m 13s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1bed7a8b31b74c95a52262a628e3bc7d : SUCCESS in 1h 06m 21s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9b189a67b8eb43bb9c370e3ae1aa3598 : SUCCESS in 58m 02s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/08e8dd80ad8047268b112fe405c2fa8e : FAILURE in 1h 05m 55s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/3e14294919d94c08bd83eaca2bb8c50a : SUCCESS in 21m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/472c626bcb4c4bf884d7f4df5f59f7eb : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":2},{"id":"030fd93d0f2a3b2341e6bfe5943870773391fbaf","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-14 01:00:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a30fa698554aba571668a5dc0e6a7759d48ce750","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 02:19:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3f9fd3fc625433cbabc45258bb5edd3\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e6ea7f253b924d1389af703b24cbfe11 : SUCCESS in 3m 24s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd0e8635a13d41559307f0929b874e98 : SUCCESS in 2m 53s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/9d2d8022b7014d1b8876ae98f56bdd59 : SUCCESS in 16m 42s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b35b145a67e4fb09c225dbd8cc5da94 : SUCCESS in 11m 55s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/73af858a1341468d8cb4dc2b7e6af521 : SUCCESS in 11m 07s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4fd486b4d8a6458e95717cc8791ad733 : SUCCESS in 1h 04m 53s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52b326d64e4740e6bb6dab1aee385168 : FAILURE in 1h 03m 00s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99441fab850c432e8bccf02068ab2fef : FAILURE in 36m 30s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/71779d3a265a4817a6591a4f2114ad10 : SUCCESS in 45m 03s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d0b301ffc4a4ce593314925c1105896 : FAILURE in 41m 01s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/04971a3dd32b4a0eab2fdc352d5b605f : SUCCESS in 21m 28s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2731e60534b84923ad163390cb11e1ed : SUCCESS in 1h 06m 56s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1939b22fa7a84f2791a286443192d7a0 : SUCCESS in 1h 11m 43s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/201539533714461bb6d463fa21b5a540 : SUCCESS in 1h 06m 06s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7bf50c4e6f5e4f9da28effb2c9050d77 : SUCCESS in 1h 04m 14s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/afe0a6e6bf4c48779e6b229a4f4f6dfc : SUCCESS in 59m 13s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/42b0b3a58ba24b10ba76d9e8c623ab42 : SUCCESS in 20m 08s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d4eac873b34842b4b88aaf6d4e7c5c2b : SUCCESS in 18m 26s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/21b2a36e858b4b5b80a2f4fa37065626 : SUCCESS in 13m 53s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35db027d5f744e9ba7cb5a6568aaf772 : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":2},{"id":"7be89bac27a701edd655de2764c2989017738062","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 02:20:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6fa5a2e0fbc696f62dfdabf26a5d3ee8aeeee773","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 03:30:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/518c6d36efc54709a5922f3de6ba9ecd\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c74efaa49d9d4f3d8d8226f7b0b72a49 : SUCCESS in 2m 56s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0933173223804f7093fdba18e8c0552c : SUCCESS in 2m 46s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/c5d5b5c7e0de4542bd3253a9d6163022 : SUCCESS in 13m 52s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e0918da55a1e46839ca4926532939d9c : SUCCESS in 10m 55s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3e0d74c393e148bf95a4db4ea3a49d61 : SUCCESS in 59m 29s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/14807b6e4f264957aefdb4633294c251 : SUCCESS in 1h 05m 40s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f84752aff94f4b90829113701a1424ad : SUCCESS in 59m 38s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/283894bf98cf4cf5842d199e9f9a640c : SUCCESS in 57m 40s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1e2aa65ed162474fb4c732a9062d656f : SUCCESS in 50m 59s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/8916e34724644ff2acb074e7ce99bd9e : SUCCESS in 22m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f729194516f14606addcd58ed8348d92 : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":2},{"id":"06a28520bb1b841301db34b84e487ed30f171931","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 03:30:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"de202384567aa3690a05fd2b3c7b30b297c9c310","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 03:31:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a7e6dc6fb9f4a9fb9c577b4692f7747\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea2c7b4c5df3416c983697beb22fe696 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0235eb47aaeccc59505931a950f8e6cec132871d","revisions":{"a781dde235bcdb2bfe90e6899ea20ae435b1b0f0":{"kind":"REWORK","_number":1,"created":"2023-03-13 01:54:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/89/877189/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/89/877189/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/89/877189/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/89/877189/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/89/877189/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/89/877189/1"}}},"commit":{"parents":[{"commit":"4f9ea57cb295fce0c579b8af8a2f4ac5270e4631","subject":"Replace legacy facts and use fact hash","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/4f9ea57cb295fce0c579b8af8a2f4ac5270e4631"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-13 01:53:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-13 01:54:27.000000000","tz":540},"subject":"Remove tenant_* properties","message":"Remove tenant_* properties\n\nThese were deprecated[1] during the past cycle so can be removed now.\n\n[1] 91c9bc3f33e0b8dadbcaaa90a5013c1c83065e92\n\nChange-Id: Ib1d2a70a8e40680ebb582cbf4e59a0c018c21447\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/a781dde235bcdb2bfe90e6899ea20ae435b1b0f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/a781dde235bcdb2bfe90e6899ea20ae435b1b0f0"}]},"branch":"refs/heads/master"},"0235eb47aaeccc59505931a950f8e6cec132871d":{"kind":"REWORK","_number":2,"created":"2023-03-13 03:09:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/89/877189/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/89/877189/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/89/877189/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/89/877189/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/89/877189/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/89/877189/2"}}},"commit":{"parents":[{"commit":"4f9ea57cb295fce0c579b8af8a2f4ac5270e4631","subject":"Replace legacy facts and use fact hash","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/4f9ea57cb295fce0c579b8af8a2f4ac5270e4631"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-13 01:53:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-13 03:08:53.000000000","tz":540},"subject":"Remove tenant_* properties","message":"Remove tenant_* properties\n\nThese were deprecated[1] during the past cycle so can be removed now.\n\n[1] 91c9bc3f33e0b8dadbcaaa90a5013c1c83065e92\n\nChange-Id: Ib1d2a70a8e40680ebb582cbf4e59a0c018c21447\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/0235eb47aaeccc59505931a950f8e6cec132871d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/0235eb47aaeccc59505931a950f8e6cec132871d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
