)]}'
{"id":"openstack%2Fpuppet-neutron~877196","triplet_id":"openstack%2Fpuppet-neutron~master~I7fae9744cda0eca8dd5278677f4c4e56132e7524","project":"openstack/puppet-neutron","branch":"master","topic":"retire-tripleo","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-03-13 13:10:02.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I7fae9744cda0eca8dd5278677f4c4e56132e7524","subject":"Remove TripleO job","status":"MERGED","created":"2023-03-13 02:18:43.000000000","updated":"2023-03-13 16:01:31.000000000","submitted":"2023-03-13 16:01:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877196-retire-tripleo","meta_rev_id":"32ce81fefec4a827e02de6b5423e2d0990015ab1","_number":877196,"virtual_id_number":877196,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-13 16:01:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2023-03-13 05:01:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2023-03-13 05:01:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-13 04:55:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e78f152a4373a914a9954ab01175395ac7181776","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-13 02:18:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3183dca14ac8930fcd9b382307fdc0c635b3cb4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 04:55:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab6b6bf1ecce429aa341161fc197ad4d\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ce277efd75e4ab4a77cd0f0c9472541 : SUCCESS in 3m 11s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fcb837d6f8d74da1878f98379f4a3b68 : SUCCESS in 2m 43s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/587c2701f596420da1a5f3e3b295ac12 : SUCCESS in 15m 57s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d84fe9e0905c4c74acf957693a78e083 : SUCCESS in 13m 00s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b8498427fa294b5f9ebdfc27b0c3f5d6 : SUCCESS in 12m 21s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/51ea49694dc54c09b1a6d8cc9a2330e9 : SUCCESS in 54m 10s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a476326384c4221b5519441ef5248e1 : FAILURE in 57m 40s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f06c7b6575c94e7f91072ea9b0a3fa40 : FAILURE in 26m 59s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07e15b8a39ff435984d08353a4f9f702 : SUCCESS in 49m 12s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a9dcd5a22c3c445b9760795b4214d3cf : FAILURE in 44m 00s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e5381f7d7a5d4762872cb5a3f83a6e3e : SUCCESS in 35m 19s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f161a7a55f7d4e309d46f4143594fe22 : SUCCESS in 57m 13s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/18a6370fe441468a9bc37a8fce28ae08 : SUCCESS in 1h 44m 53s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/27bdd937a4694fc49e7117dc3d0ed977 : SUCCESS in 1h 05m 17s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e5114886f35d478a95f58805ed80177f : SUCCESS in 52m 28s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/90bc46aa3f2641f391ce9f54c1a1032a : SUCCESS in 1h 02m 45s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/491f09fdd2d846c3a088f7e746308f98 : SUCCESS in 18m 40s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/100a64c1ea874b57b32d8002f4d3b5f3 : SUCCESS in 14m 08s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/56ec7532523f43c099fd1ae6337a42b9 : SUCCESS in 14m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0ab70b30bdd93fee8b4f459a8e4883912b075d53","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-13 05:01:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"04709c269375d5fd8d3871876dd0601d7ab60761","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 05:01:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"62f82e4bcc770ae4a001fa3cda4f54be570dff32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 06:07:23.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a4e891c8c2574c3099995cb7123610ab\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b92c675e8694306a1c00f209d0ee161 : SUCCESS in 3m 19s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bdab70180c394927a14a6eda8c3ea891 : SUCCESS in 3m 05s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/cd1c8472343d4cdbb3fe9ee32145fb4e : SUCCESS in 16m 47s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/35b88511f8be4c8ebf211586a75e1378 : SUCCESS in 12m 09s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/089a88809d1e4456acd21a3d43ad9d2c : SUCCESS in 58m 11s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ff9aeed6bd97436a9f57e5a987dda621 : SUCCESS in 1h 03m 48s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1b7b7475c6c04858a46047ed591dfbb2 : SUCCESS in 1h 00m 07s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/849c83a9b80d4637ac1a235b3de69489 : SUCCESS in 51m 44s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc7712a84f194c3fb0a7b10b92e14d5a : FAILURE in 52m 58s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/779ee958b3c947be851ae8cb69cdf560 : SUCCESS in 20m 33s","accounts_in_message":[],"_revision_number":1},{"id":"c0cda5e22d762d33b4dc26f3055369fd9cb8c416","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-13 06:15:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a0619fcd61ff7df5e4f56e1d74514d17076ee844","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 07:25:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b98103a3b51d4c8abab46483f509617b\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1f3732e052e24068864b3f60396d83a0 : SUCCESS in 2m 43s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d6415ae483b4ef79dd636b1e250febf : SUCCESS in 2m 22s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/daa8f8a876094dfabe0b7fe46bffe5ef : SUCCESS in 14m 25s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/27b5b4e481ae4251b83db1e83f2f2909 : SUCCESS in 13m 08s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f8d0ea3031a0466d95b13052c3b8561d : SUCCESS in 12m 40s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55d00ad724ee415c9f12003acfe408db : SUCCESS in 47m 43s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/56f3ea6428e04771b9f3626f6b7c32ec : FAILURE in 56m 18s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ec64a078353438ba1830cd750ef7da3 : FAILURE in 30m 29s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f5fcf7a489340a494c084c2e41c33c3 : SUCCESS in 40m 46s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d4f24271384c47778b7cdb90ee3ec4c3 : FAILURE in 43m 36s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7bac3e14603e4b6cb21ea5b96e8c7923 : SUCCESS in 20m 12s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/44022f0e884e477f91ca98c56a06fcc9 : SUCCESS in 1h 02m 26s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b416dd2f7c5e4cd0ab4e8129c83fefa8 : SUCCESS in 1h 07m 17s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7446e957a930426a9194d73ca2c3c6bd : SUCCESS in 1h 00m 16s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1ba55cad8a254cfc917a3b384f88c3d1 : SUCCESS in 52m 29s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/44071c67a8e14e06915859f21d56ce71 : SUCCESS in 55m 56s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/7ad868cdbe304d21bda22827ee9494ba : SUCCESS in 19m 02s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8bf1e1ac006e462fbf86775173410c72 : SUCCESS in 14m 23s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9762278e1294301bb5775cd7206fa42 : SUCCESS in 18m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cc56c8122b584579bd124c76d3c40d8c6151f346","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 07:26:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4f64084747e52663e134d8c16faa7e186a748440","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 09:02:23.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/20024652bc4c4b2fbcb8af7de166dd05\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2e98c2837e57497aa9790a1133c815ed : SUCCESS in 2m 50s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/32ccadbdb83b47bdaca81364fa2d8c5b : SUCCESS in 2m 52s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/30bee8e5659346f6b1fd50f5334d755c : SUCCESS in 15m 29s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ea09d5bbf91422f8f1d87023d9bfb6f : SUCCESS in 11m 56s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/429ed193732a49a0b785d101856e7ba1 : FAILURE in 1h 09m 33s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7cfc94fa6bfa46a38dfef66de443c9c9 : SUCCESS in 1h 34m 09s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b5c30fab4b8d47e3adce7de2ec6dca8a : SUCCESS in 58m 12s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f4e0d2db692a4ad4af32a51b0c7c5bda : SUCCESS in 1h 04m 17s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c15de994263c43609e113b47102510bd : SUCCESS in 53m 07s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/0526eaf65d11416aa0bfe6b72c434dbb : SUCCESS in 22m 09s","accounts_in_message":[],"_revision_number":1},{"id":"303afa6a23e2795c9a9d36a6643d61f839d01f2d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-13 09:16:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0ad3b22634338335adc135f1c99e2622b1714658","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 10:58:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/666b5bf44a074ce38918d9d7ae8095a1\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b2f5d36a60154a269a5615db0901edd6 : SUCCESS in 2m 27s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30177f25a9de4dd7abb3067db164fac9 : SUCCESS in 2m 30s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/99e50b2e93b646ebaa76610260156aba : SUCCESS in 14m 42s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/781d3f77477142099e23b1c5aedb65c2 : SUCCESS in 12m 07s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/91729a5747bd4f5c9448e1cec1d98f10 : SUCCESS in 11m 40s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f87b7827a7a5437e84efe1e79a287532 : SUCCESS in 54m 45s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c6f5c48bfe74d4aaadfe41675ab1bf5 : FAILURE in 38m 25s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/663a2c111ebe40138a2d9d5a97768ad8 : FAILURE in 30m 00s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c0e4c8aa25db4257a2446e7b3607ffcd : FAILURE in 33m 10s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0034d61051d44e3e9699d5f25967f769 : FAILURE in 42m 10s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9885764c8f9b4ab28a97a764de1cefa2 : SUCCESS in 21m 21s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8204d89fed394fcf914d5eff1e12cfb1 : SUCCESS in 46m 04s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ae94337a013742b68957890147d92a2f : SUCCESS in 1h 23m 23s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6ddb642f406c41d591cfc8cb963aab2b : SUCCESS in 1h 07m 18s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/72ce4b68018644b2b020381655607515 : SUCCESS in 1h 34m 57s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2cd08e0f3e094e6989a971911ed6f105 : SUCCESS in 56m 36s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/c00bc9d29e57424fa95a771256300086 : SUCCESS in 21m 01s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e24d6eccf60c48fe87cd0fc90d132b58 : SUCCESS in 17m 01s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fb45682c0be94e4a9af75de7662f9cc1 : SUCCESS in 15m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"18bc10cc34e10abd447a62d1534010ad9e541ef9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 10:59:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e9a32c0c8e4b269397bb826e20b6b8b76947008c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 12:13:17.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c6fabf12c4514f7da6f9f99a38faddb9\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/842d3ad23cb0478ca3b2bc51e2c25e0c : SUCCESS in 3m 19s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f67b4fdf7509400295885e672f07bc30 : SUCCESS in 2m 28s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/ddb071457eb54bb9ad440d8b096c0c5e : SUCCESS in 15m 35s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b932dbd499247a497ac812a12621047 : SUCCESS in 14m 11s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a378e939c7064afeb7265bccde67105a : FAILURE in 53m 39s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4abdd1073931493f931f72735d2dc494 : SUCCESS in 1h 02m 16s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f9a05ff47c8f443abb4050882640d72e : SUCCESS in 1h 02m 06s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a5696390ac64994b9bd3cafc1ebff7f : SUCCESS in 50m 59s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a405418388ad4e49a739dd06b112a5fc : SUCCESS in 1h 03m 11s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/33ace62650234b5d97738daa3ee15bb8 : SUCCESS in 20m 05s","accounts_in_message":[],"_revision_number":1},{"id":"d59c61ff376221df84c13d0e5a1fed9bd88d171b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-13 13:10:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5382f34e4855afc349c039641a2ff39529a611b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 14:23:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d67055147b074185bcce4826bccccaf8\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e669243f5c304f3e9a0b5fa3651dc3bb : SUCCESS in 2m 53s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/afaae02678624c7ebc23daa5a130d037 : SUCCESS in 2m 54s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/b0e95700392c49ff8aac0a8604c6ca27 : SUCCESS in 15m 12s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c575d77227104cd6a61db2e44e720965 : SUCCESS in 13m 13s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d86afede47aa46f8aff03e636bc1d9fb : SUCCESS in 12m 56s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52e4ead5f9b547069442f739a3654ca9 : SUCCESS in 46m 29s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/732b6d06c8534bb7890f17444083eb10 : FAILURE in 1h 02m 19s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8597bc48fc143ceb348abac57776ae7 : FAILURE in 31m 39s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a5786b4ab61492c8136e1b36c849884 : SUCCESS in 42m 20s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9d86a9a7eb3c4f8abd7ce5735e652e1f : FAILURE in 43m 52s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8164c18f6bea4297b007d57967fbf6c7 : SUCCESS in 26m 24s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/845d21bcf9db4363b1bdd344d833c0df : SUCCESS in 56m 16s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/98527c7b7dc04382ac86ec5a6922fb00 : SUCCESS in 1h 12m 08s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/86f869d438b84a06b1535c4dc7337a33 : SUCCESS in 59m 32s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c79aabe29b68472f840aac84c288ef66 : SUCCESS in 58m 24s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/369e2269c82b4909ae7ae430713979bf : SUCCESS in 51m 27s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/e9c036e935b14444acb5cfcf0f4ef678 : SUCCESS in 19m 55s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f67f2c61d794a57b3698564480466c1 : SUCCESS in 21m 17s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/86616c294f3c432584c7079809eb3c06 : SUCCESS in 14m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a66f5893a12cacead77caec0be86c01002459964","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 14:24:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"af51c3be4ac8eb1b2ff4418013801025ce447135","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 16:01:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ded7e50460154f91a49722f440d8820b\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bff4014d5d424cae8d1837b0375ecb1a : SUCCESS in 2m 36s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a503285f2a5642bf8acb18ab8b1f14a3 : SUCCESS in 2m 34s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/c1a7291ea634450883c79607d5fa3adc : SUCCESS in 13m 20s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e041613e541a4ccb9acb7eb420db5d9c : SUCCESS in 10m 31s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/56b93fe1dce547db8689a9e623ab83cc : SUCCESS in 47m 53s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6e81cffe5ad94bb497c69c876af55477 : SUCCESS in 1h 08m 31s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/772607a088004501b1d6203c3edf1c4a : SUCCESS in 56m 10s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e2fd16efacad42c4a7b8ec27e8b090a8 : SUCCESS in 56m 54s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5d8692bdd7dc46b6bf587bceefb8f212 : SUCCESS in 50m 57s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/7cc1ecc1bc7d473eb9cc664529c81a76 : SUCCESS in 22m 20s","accounts_in_message":[],"_revision_number":1},{"id":"32ce81fefec4a827e02de6b5423e2d0990015ab1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 16:01:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"138b5284855c6a24fe25c21e64ccaa8e1dc7a5b7","revisions":{"138b5284855c6a24fe25c21e64ccaa8e1dc7a5b7":{"kind":"REWORK","_number":1,"created":"2023-03-13 02:18:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/877196/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/96/877196/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/96/877196/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/96/877196/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/96/877196/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/96/877196/1"}}},"commit":{"parents":[{"commit":"4f9ea57cb295fce0c579b8af8a2f4ac5270e4631","subject":"Replace legacy facts and use fact hash","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/4f9ea57cb295fce0c579b8af8a2f4ac5270e4631"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-13 02:18:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-13 02:18:33.000000000","tz":540},"subject":"Remove TripleO job","message":"Remove TripleO job\n\n... because TripleO is being retired.\n\nChange-Id: I7fae9744cda0eca8dd5278677f4c4e56132e7524\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/138b5284855c6a24fe25c21e64ccaa8e1dc7a5b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/138b5284855c6a24fe25c21e64ccaa8e1dc7a5b7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
