)]}'
{"id":"openstack%2Fpuppet-neutron~879372","triplet_id":"openstack%2Fpuppet-neutron~master~Ia502ce0a5e3f77259f2b395891f97401c507008e","project":"openstack/puppet-neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-04-13 01:19:09.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ia502ce0a5e3f77259f2b395891f97401c507008e","subject":"CentOS: Install only bgpvpn plugin package in neutron-server","status":"MERGED","created":"2023-04-03 18:14:14.000000000","updated":"2023-04-13 04:53:02.000000000","submitted":"2023-04-13 04:53:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"879372","meta_rev_id":"e91b861b21bba220a7a4857ad5c34404a1bfaa2c","_number":879372,"virtual_id_number":879372,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-13 04:53:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2023-04-10 19:43:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2023-04-11 15:05:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2023-04-11 15:05:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-03 19:25:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-10 19:43:10.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2023-04-11 15:05:37.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"2edc700a35c8749431eac7de3e8cc65d51fc2db5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-03 18:14:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1e2178288b4c6ee9015ef7cc50103833faaf11e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 19:25:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/de45d013b51a40f0b7a84de0777e3538\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/74e3b396d4b1454c92322317f1f0d41b : FAILURE in 3m 16s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09ec4c0986f449f18c18609a462b2530 : SUCCESS in 3m 15s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/ac12cd8cd2874a2895518ffd39d0c895 : FAILURE in 10m 35s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0aadb6de97743c9af38257747f54d55 : FAILURE in 7m 37s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87641136a3c94432abd5242dd9ad9e8f : FAILURE in 7m 20s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0cf0c26f62a846bb99d3321502157a33 : FAILURE in 39m 49s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/491f6d5b38b6409c87f96d0016efcb84 : FAILURE in 49m 35s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f94e4bbd1d44c17b9217c6f832c50f7 : FAILURE in 39m 35s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7443d4efc10e491ebb4585a58140efd1 : FAILURE in 29m 25s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d23e8a2e71d54d6faa80e022f0777a91 : FAILURE in 33m 28s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5fd53412610748bb8a0a6952a56a9f37 : SUCCESS in 24m 12s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ebfbe5df786648b5a5c857a7dc3ccbff : SUCCESS in 54m 04s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/18390fd591e14c519d66d09cc15d3942 : SUCCESS in 1h 03m 25s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1eb70deec86347898c3635db5f10294f : SUCCESS in 59m 04s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e389dc15faeb46d984c52bb834acea38 : SUCCESS in 52m 34s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/96da3affd36848d3b3506c20bcefab9b : SUCCESS in 53m 28s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/b89e8b74388745819ed30f4e3b135453 : SUCCESS in 21m 08s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0191398c6a5841b9b4847cf99eb25598 : SUCCESS in 15m 25s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b73315e30a15421fa88e950aa4ac149e : SUCCESS in 15m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e27d434ee8533985509ed460d28014c5ee2a8a5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-04 01:49:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f25482ea6ed8ac062e826fb05ae446ba48e376d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-04 02:06:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"96758438bbca28f2798301eb7f206aad8cf31b4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 04:13:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/45f6713661dc47c9bfe3b33e369840c3\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ab1af07b0714135ab0bba01bf096f73 : SUCCESS in 2m 57s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/684e6b192ec14e618cfe41b86804a603 : SUCCESS in 3m 49s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/decb21702d184fa2a73ee2f692c722b9 : FAILURE in 18m 36s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/412eb4342a144245a9ab7d946ba7c650 : FAILURE in 13m 54s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ef16b1f4ca64ec8b0a5e7a67c835f0f : FAILURE in 13m 34s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e3d07c5fc654a989735304024485760 : FAILURE in 45m 03s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d59bef02a2c8477b9881bae7406c2fa4 : FAILURE in 48m 47s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de938c515f1d48599eead07e20b14acd : FAILURE in 48m 19s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/864750e232f04b9482cd5ef043f3b7cf : FAILURE in 37m 00s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c6c049812f464274aa1a4dab31ac21a7 : FAILURE in 37m 33s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f7deb9e8ef89411d8bc0e01df822a87d : SUCCESS in 20m 20s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc2994d82f0d4cc98747245d95c073af : SUCCESS in 1h 11m 23s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ce3b64f8e0749a8ac166a98b9626406 : SUCCESS in 1h 30m 40s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/93a6bd6d243142b095b8e7a91641e67f : SUCCESS in 1h 10m 32s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/99029b0d5eb84ec0afc69260f1d2a692 : SUCCESS in 57m 58s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ddc9b051b4c340af801b84ff7b0260a8 : SUCCESS in 1h 01m 53s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/6ed4c3602c2e4b558c09de1aaa7f2e32 : SUCCESS in 24m 43s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d90738651ba41f3ae2f453eadaa4c6d : SUCCESS in 15m 13s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/feb36fea3c434ecb8f989a4077fae9ea : SUCCESS in 14m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9cb0cedd7ab29fb6ae0e17bd885f12688a90f6af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-04 09:10:00.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4d1f16e65e617e390ec9a7af7c7fbd9f3ae0fbe8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 10:25:40.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b805ca4866b4de287f4054b86b74bcb\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/252fb0a6f38b4cd39b1962f48c58a02c : SUCCESS in 2m 39s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5355b19949e49379a822176d473f7cf : SUCCESS in 2m 26s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/6ceea308bab344c99b6c2c609143104a : FAILURE in 15m 53s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7361390bde604d439e776e9b1fc5becf : FAILURE in 10m 16s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2553bee5f8d847059cfd8c8d402afc96 : FAILURE in 11m 12s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a1872fd8bd04bdcab968a39f0e3e789 : FAILURE in 41m 25s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9be7113e8f254354a2e6cb6aa048e137 : FAILURE in 41m 15s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5bffd25786c542d6a8603444dec514fe : FAILURE in 37m 37s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6c2ab60777b4dffa39ef6d0d05aef92 : FAILURE in 30m 33s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c4b203e3e3d349a58b1093885617f07c : FAILURE in 31m 21s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dfd5867437ca4a49b84c174496a09f8b : SUCCESS in 21m 36s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3da7b36d5b7f4f348d79d5118ed9148c : SUCCESS in 1h 02m 53s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8f568274917f4a958950c51aa3a82c31 : SUCCESS in 1h 12m 08s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6e07655e34a44bdbaf2cf081f9742278 : SUCCESS in 1h 05m 46s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/48e5a4acf80c4244928cf8b8b3ff8a0b : SUCCESS in 54m 47s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/967da0fd7cc340bb8273d3605b26e1f2 : SUCCESS in 59m 54s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/aa40a92b515548fe93247d39d4418c08 : SUCCESS in 21m 03s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe8133e9808840659b779c9d87f78734 : SUCCESS in 15m 15s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d69a48f2aca7435bac8aed1533205e5c : SUCCESS in 14m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63ec4d6843244102a8616b6d966f91df : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":4},{"id":"f56a9f4325c73dbdda8ded7c14223cddd4bf888a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-04 10:56:31.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"874a5282a98feba41a44f0ae6fdb3f35a921db12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-04 11:03:53.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"cdfe21e124812b3d0612e7be339816b4b2baf193","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-04 11:08:33.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"05ea24efc8efe2e2493a6c07b47c4f3fb757af5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-04 11:11:27.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"bb0deccff362d8bcc2294e7a149a6e9655e701fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 12:40:21.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6dfbc56f1c0142e6ac340d2579a6cf5e\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c73480989fcf48dfb1e8fdb0f39a861d : SUCCESS in 2m 57s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/487ad86e047b4db1b2386d59951d7182 : SUCCESS in 2m 52s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/4be2a2bce4c94e2a9a2310345873d915 : SUCCESS in 13m 07s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca106174181f4533b2af481b554661b9 : SUCCESS in 10m 41s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c520c0186b6454998e40e00bc81fbd1 : SUCCESS in 10m 03s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eff5a6a7751b4447a3faf14223f12ac7 : FAILURE in 37m 51s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b2c8f881f224433944f9ad1650e8eb2 : FAILURE in 49m 29s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/548af7ee75fb47d19b5b5eb1c0282055 : FAILURE in 45m 26s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4090ad5dd76d49c8b1a399fcdf49bab7 : FAILURE in 31m 00s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/012fccca3ea542068b60a1e8f163340f : FAILURE in 34m 53s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/44b480234e8e4dc3a651849c210a16ac : SUCCESS in 23m 26s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/38165120d79349ddb03aeef3bf127c50 : SUCCESS in 1h 01m 21s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1c9da6414d9f4495872131ccfa952e7f : SUCCESS in 1h 21m 36s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/28a7df1f7ac643119053905dcbf7ea7c : SUCCESS in 1h 16m 47s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/eee669f64df9433ea1be6401676124f1 : SUCCESS in 1h 00m 07s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bf157bc952b04a72aa9d4d67df6479e5 : SUCCESS in 59m 41s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/e3595199b2d844fca46d0af6142ef9c1 : SUCCESS in 19m 47s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93843ade26db478496909be36eb8ad81 : SUCCESS in 18m 28s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/165f8fd258c9402c802d5a8acc04cca0 : SUCCESS in 16m 08s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"11a1afb42dd40ab7f85844af53c056f5fefcf9f1","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-04-10 19:43:10.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"1e5085612175cfc0cd74c53395a7dd664a14c73c","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-04-11 15:05:37.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"e20e437837a351b2d17007c450a06c8c1fb9c2a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 15:06:35.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"45cd86940e64b71397de6a62b945d377580a9acc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 16:36:05.000000000","message":"Patch Set 7: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/879389 is needed.","accounts_in_message":[],"_revision_number":7},{"id":"6d34220d6394746e417262c18851f9e6e0745f55","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-12 01:16:11.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"7e142f151da6623e8cfdf0dc6a178be7263257f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 03:15:16.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcaf0ccc7c94490ea39b56509b7461d7\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/88ebb5ef709548a0b2ce8ad4bd3c14d3 : SUCCESS in 3m 29s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66b9fbad2265429c8c57edd1dad32cbb : SUCCESS in 2m 41s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/34437480a14141d99ff166ff2c492557 : SUCCESS in 12m 36s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33631f1b7bf448bab5e73659bf001e4f : SUCCESS in 13m 01s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e892835e358c4ac894902ce0fb3b816f : SUCCESS in 12m 02s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/68d31e83df124c82b0f1e8254bf6108a : FAILURE in 41m 02s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/187245fb035743c3a91cb61022e5f062 : FAILURE in 51m 14s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8cb61fc11af24861bc1a6c68c56bc3bd : FAILURE in 38m 03s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3580ab07736a439ea657ed2f13997f38 : FAILURE in 33m 19s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8ba24361873421d9b45c2a31ca9afc7 : FAILURE in 34m 31s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4881ecf58cba4df2aa20f0088c2bb3e9 : SUCCESS in 15m 43s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2e0b3a1cc8d9447bacc5bc85002d0541 : SUCCESS in 1h 00m 38s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d7a5dad82caf4aeda27e6e5ac40616f2 : SUCCESS in 1h 14m 11s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2eda3588508d4250966d88c95ae48b78 : SUCCESS in 1h 15m 44s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/240332eeca184832b56419b78b535414 : SUCCESS in 57m 40s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/47ce47f3910c4132ab29b6a0123b1a29 : SUCCESS in 1h 03m 12s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/e18d996353984b1ea634d1850cb843d0 : SUCCESS in 21m 37s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c22abb2b7dd473ca0806e80ea3adbb5 : SUCCESS in 18m 12s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce0ce46b9489499a97a7c5595881b792 : SUCCESS in 17m 09s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c90ecc66022903bf36f52f024e7a38d194f65623","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 03:16:09.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"dcbecc973401857d607815291347fe541fb57360","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 04:34:54.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9474f783daef472892ade5819fd67745\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d580e5c1192e44ebbb2d8fd9f1307364 : SUCCESS in 2m 51s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67806e3013e54b8fa5c19d39d085087d : SUCCESS in 2m 40s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/d25a49c44fcf49dc9bab4911680fd253 : SUCCESS in 15m 34s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e91921085db444a9bbc4052b8552e7b : SUCCESS in 14m 16s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a4dec2bbdcd24106b78e8d375aea0bdf : FAILURE in 1h 15m 58s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/686938829848474f914c8a6f8fc4b3c8 : SUCCESS in 1h 04m 35s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a70e11c6e00d462ea2908411b725bb81 : SUCCESS in 1h 01m 58s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/abfea29de8d74a7fb27c99a41adec96a : SUCCESS in 46m 32s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/549e7f3406654a578474c414d7810228 : SUCCESS in 46m 24s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/2a9e23357819418e825858e72ff99f66 : SUCCESS in 19m 55s","accounts_in_message":[],"_revision_number":7},{"id":"2cc7dadf423028441e6dc44468fb930e447f9640","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-13 01:19:09.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"46921faa89107a76a54e61e89b2353af75ef235f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 02:40:46.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0b880718feb46978355aa64ff579338\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a25751af5c284e04aca9bbae41c3cf42 : SUCCESS in 2m 52s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3bb233eb82b04311a90b131c8ca43928 : SUCCESS in 2m 51s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/33b994d3f25241ef8f760dba4a484455 : SUCCESS in 12m 54s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c87b9f131143495db8b7f71e7e181232 : SUCCESS in 14m 02s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05e986660b13488a956b3a89cf241086 : SUCCESS in 10m 52s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f4360ca01369453f91aaed85ee5d7222 : FAILURE in 43m 58s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e3931b4693a4d94acb366c4c191dc96 : FAILURE in 58m 56s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ed1afe374ed4810ae57de7fd8ad92e1 : FAILURE in 45m 08s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b8ae495c3c9e4919a0a719106bd507e4 : FAILURE in 28m 58s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8f5acb6d86e4718969672888679c648 : FAILURE in 41m 24s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c5ab5ba05cca4b699cb80ba77c6dc839 : SUCCESS in 15m 30s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e1a379cadfc74659a07c3a0efca7d4b5 : SUCCESS in 55m 12s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/efc94e060b2a491c81f51c16a037667e : SUCCESS in 1h 00m 52s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0fe53193b2eb4669af31f280030e8e80 : SUCCESS in 1h 04m 09s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e8a281fc7deb4c2788d49c0fe49be1b2 : SUCCESS in 55m 42s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/723740ea956c40c19cdda1e4265d35d2 : SUCCESS in 43m 27s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/2b1156e039c8453899268f47315b3c84 : SUCCESS in 17m 06s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/508e3424cdd143c5beba22210ae82c87 : SUCCESS in 13m 16s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/218ce6c32d9b42c48c3c58384928f925 : SUCCESS in 14m 40s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"20ab09545f256c50f0705805865665b402a33832","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 02:42:47.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"54eb1bb37596fcdbe0fa1be0823bf7bf9b309a8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 04:53:02.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0ff000488d943e8a7de8c81083b3860\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de735dc81ed24eb3aba03f293fe08bfc : SUCCESS in 2m 40s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f206da4cf3244ef0b459bcb9b00fe1c5 : SUCCESS in 2m 48s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/122330a95a8a4d45b3fc47a7f7aaeeae : SUCCESS in 14m 12s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84f2c042896f4b7eb6e7664d3b980709 : SUCCESS in 13m 38s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b6595059511648f69f49fe8c3d3684cf : SUCCESS in 55m 26s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1c05b965afe94d66b5d4456c344470db : SUCCESS in 1h 51m 32s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1e4b634cd51149648ed2e6c9194f3c5b : SUCCESS in 1h 10m 01s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1813d940ec41464b984ebc054e33aabf : SUCCESS in 44m 33s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e4bc80132ea34f2da84b192ef993b302 : SUCCESS in 1h 02m 28s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/248a138229e44ca1af07e86e68f64b66 : SUCCESS in 20m 00s","accounts_in_message":[],"_revision_number":7},{"id":"e91b861b21bba220a7a4857ad5c34404a1bfaa2c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 04:53:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"0214e964bef8f8707f3933d0a3ff234e56afa7dc","revisions":{"eff0dff0f590bc572d6ad152f326eb06f395696c":{"kind":"REWORK","_number":1,"created":"2023-04-03 18:14:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/879372/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/72/879372/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/72/879372/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/72/879372/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/72/879372/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/72/879372/1"}}},"commit":{"parents":[{"commit":"7ee14d9f9b3d9cf8311ec401aa56dcd93466add6","subject":"Merge \"dhcp-agent: Fix parameter description format\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/7ee14d9f9b3d9cf8311ec401aa56dcd93466add6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-04-03 18:13:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-04-03 18:13:43.000000000","tz":540},"subject":"wip: Add separate class to install dynamic routing service plugin","message":"wip: Add separate class to install dynamic routing service plugin\n\nChange-Id: Ia502ce0a5e3f77259f2b395891f97401c507008e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/eff0dff0f590bc572d6ad152f326eb06f395696c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/eff0dff0f590bc572d6ad152f326eb06f395696c"}]},"branch":"refs/heads/master"},"6f2c0c66d88e4bb2f10968690243800966ed633f":{"kind":"REWORK","_number":2,"created":"2023-04-04 01:49:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/879372/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/72/879372/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/72/879372/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/72/879372/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/72/879372/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/72/879372/2"}}},"commit":{"parents":[{"commit":"4cffd94ad0f0681a54b630d422787d9b7850d61f","subject":"Run db sync command with the service user","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/4cffd94ad0f0681a54b630d422787d9b7850d61f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-04-03 18:13:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-04-04 01:49:30.000000000","tz":540},"subject":"wip: Add separate class to install dynamic routing service plugin","message":"wip: Add separate class to install dynamic routing service plugin\n\nChange-Id: Ia502ce0a5e3f77259f2b395891f97401c507008e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6f2c0c66d88e4bb2f10968690243800966ed633f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6f2c0c66d88e4bb2f10968690243800966ed633f"}]},"branch":"refs/heads/master"},"6cda80f939b6790274b152734b0d8a4c2e053892":{"kind":"REWORK","_number":3,"created":"2023-04-04 02:06:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/879372/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/72/879372/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/72/879372/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/72/879372/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/72/879372/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/72/879372/3"}}},"commit":{"parents":[{"commit":"aa39b06d5ad3db5478564851a63b1b84b2dba78e","subject":"Run db sync command with the service user","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/aa39b06d5ad3db5478564851a63b1b84b2dba78e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-04-03 18:13:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-04-04 02:06:36.000000000","tz":540},"subject":"wip: Add separate class to install dynamic routing service plugin","message":"wip: Add separate class to install dynamic routing service plugin\n\nChange-Id: Ia502ce0a5e3f77259f2b395891f97401c507008e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6cda80f939b6790274b152734b0d8a4c2e053892"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6cda80f939b6790274b152734b0d8a4c2e053892"}]},"branch":"refs/heads/master"},"16d94fd75a99bbc2e2d685a642116fc1f06411ab":{"kind":"REWORK","_number":4,"created":"2023-04-04 09:10:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/879372/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/72/879372/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/72/879372/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/72/879372/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/72/879372/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/72/879372/4"}}},"commit":{"parents":[{"commit":"aa39b06d5ad3db5478564851a63b1b84b2dba78e","subject":"Run db sync command with the service user","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/aa39b06d5ad3db5478564851a63b1b84b2dba78e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-04-03 18:13:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-04-04 09:07:50.000000000","tz":540},"subject":"Add separate class to install dynamic routing service plugin","message":"Add separate class to install dynamic routing service plugin\n\nThis change introduces a separate class to install service plugin code\nfor neutron-dynamic-routing. This is required in deployments which have\ndragent running in different nodes from neutron-server.\n\nChange-Id: Ia502ce0a5e3f77259f2b395891f97401c507008e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/16d94fd75a99bbc2e2d685a642116fc1f06411ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/16d94fd75a99bbc2e2d685a642116fc1f06411ab"}]},"branch":"refs/heads/master"},"a6394ba68702de5f2601d9acb1e0ac7b80503b00":{"kind":"REWORK","_number":5,"created":"2023-04-04 11:03:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/879372/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/72/879372/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/72/879372/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/72/879372/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/72/879372/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/72/879372/5"}}},"commit":{"parents":[{"commit":"aa39b06d5ad3db5478564851a63b1b84b2dba78e","subject":"Run db sync command with the service user","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/aa39b06d5ad3db5478564851a63b1b84b2dba78e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-04-03 18:13:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-04-04 11:03:42.000000000","tz":540},"subject":"CentOS: Install only plugin package in neutron-server","message":"CentOS: Install only plugin package in neutron-server\n\nRDO does provide the separate python3-neutron-dynamic-routing package\nwhich contains only service plugins so we don\u0027t have to install\nthe dragent package in neutron-server.\n\nAlso the dragent package depends on python3-neutron-dynamic-routing so\nwe don\u0027t have to install the library package explicitly.\n\nChange-Id: Ia502ce0a5e3f77259f2b395891f97401c507008e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/a6394ba68702de5f2601d9acb1e0ac7b80503b00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/a6394ba68702de5f2601d9acb1e0ac7b80503b00"}]},"branch":"refs/heads/master"},"97ba31977380d5eb7e9de03f12f3f0208030954b":{"kind":"REWORK","_number":6,"created":"2023-04-04 11:08:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/879372/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/72/879372/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/72/879372/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/72/879372/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/72/879372/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/72/879372/6"}}},"commit":{"parents":[{"commit":"aa39b06d5ad3db5478564851a63b1b84b2dba78e","subject":"Run db sync command with the service user","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/aa39b06d5ad3db5478564851a63b1b84b2dba78e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-04-03 18:13:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-04-04 11:08:20.000000000","tz":540},"subject":"CentOS: Install only plugin package in neutron-server","message":"CentOS: Install only plugin package in neutron-server\n\nRDO does provide the separate python3-neutron-dynamic-routing package\nwhich contains only service plugins so we don\u0027t have to install\nthe dragent package in neutron-server.\n\nAlso the dragent package depends on python3-neutron-dynamic-routing so\nwe don\u0027t have to install the library package explicitly.\n\nChange-Id: Ia502ce0a5e3f77259f2b395891f97401c507008e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/97ba31977380d5eb7e9de03f12f3f0208030954b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/97ba31977380d5eb7e9de03f12f3f0208030954b"}]},"branch":"refs/heads/master"},"0214e964bef8f8707f3933d0a3ff234e56afa7dc":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2023-04-04 11:11:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/879372/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/72/879372/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/72/879372/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/72/879372/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/72/879372/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/72/879372/7"}}},"commit":{"parents":[{"commit":"aa39b06d5ad3db5478564851a63b1b84b2dba78e","subject":"Run db sync command with the service user","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/aa39b06d5ad3db5478564851a63b1b84b2dba78e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-04-03 18:13:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-04-04 11:11:13.000000000","tz":540},"subject":"CentOS: Install only bgpvpn plugin package in neutron-server","message":"CentOS: Install only bgpvpn plugin package in neutron-server\n\nRDO does provide the separate python3-neutron-dynamic-routing package\nwhich contains only service plugin codes, so we don\u0027t have to install\nthe dragent package in neutron-server.\n\nAlso the dragent package depends on python3-neutron-dynamic-routing so\nwe don\u0027t have to install the library package explicitly.\n\nChange-Id: Ia502ce0a5e3f77259f2b395891f97401c507008e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/0214e964bef8f8707f3933d0a3ff234e56afa7dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/0214e964bef8f8707f3933d0a3ff234e56afa7dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
