)]}'
{"id":"openstack%2Fpuppet-neutron~886301","triplet_id":"openstack%2Fpuppet-neutron~master~Ib124f05394b5f53cace06cd062ee2079afb311af","project":"openstack/puppet-neutron","branch":"master","topic":"replace-validate_legacy","attention_set":{},"removed_from_attention_set":{"7888":{"account":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"last_update":"2023-07-12 13:53:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib124f05394b5f53cace06cd062ee2079afb311af","subject":"replace validate_legacy with proper data types","status":"MERGED","created":"2023-06-17 12:44:07.000000000","updated":"2023-07-12 13:53:26.000000000","submitted":"2023-07-12 13:53:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886301-replace-validate_legacy","meta_rev_id":"a34171d5ba003053dd86b6fb65387694a2781680","_number":886301,"virtual_id_number":886301,"owner":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-12 13:53:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2023-06-19 19:43:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2023-06-21 08:36:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2023-06-21 08:36:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-17 13:18:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-19 19:43:05.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2023-06-21 08:36:14.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-06-22 03:44:32.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"}],"messages":[{"id":"0a170ee639e7d686d379a2e9e5e50e23492f5395","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"date":"2023-06-17 12:44:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"302021f272ec778099e8ffe5a60244bc8b6b16c3","author":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"date":"2023-06-17 12:45:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8a3ad6bc079e77e0978cbdd36f6cf8f36970f1fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-17 13:18:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/76d1ee1d72bb4149a006c757c805cd51\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a0313041919457592a46508662d8f74 : SUCCESS in 3m 24s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e335b507c344580b85771f59942409c : SUCCESS in 2m 32s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/e3afe10922b046518b256a70eebefd31 : SUCCESS in 4m 21s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/00fa5b94b46643be9413e95e0439a3cf : FAILURE in 13m 57s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e06d1959b9f143ce8848a59def3ffe13 : FAILURE in 8m 20s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/ab7f7c3959834ce7a8f99ae70f2a1395 : FAILURE in 11m 24s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9d64ddc067f34f4782a7c7faa060d9e0 : FAILURE in 12m 47s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e4556601241481f8cc13450e527f1d7 : FAILURE in 7m 42s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6693d4854bef4d55a343c2ab7289299f : FAILURE in 6m 42s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c41b11a7b88453fb4d484460f3d2c85 : FAILURE in 8m 51s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6779ad5f49824fb08c143edb95349f6f : FAILURE in 4m 57s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aee622d8788a4977b5336e1a3b727af1 : FAILURE in 4m 56s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/23448beaac1d408e888d63f3b665e1d0 : SUCCESS in 14m 29s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e925256618f74d5bb40b85ed41faf192 : FAILURE in 8m 27s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d3ade25c219f4e208a36d23675865543 : FAILURE in 7m 55s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a0a42e9287e74a39a319bf6a6cb20f92 : FAILURE in 8m 32s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9dd834c928914e8b84345df91af27fb7 : FAILURE in 8m 24s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0d74aa2d10ab4a43ba25c118b9e79ce2 : FAILURE in 8m 57s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/94486940baa641a19d2ec420f48f54c0 : FAILURE in 8m 13s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/767cd12d1d2a4a17b3486e85f37ff571 : FAILURE in 5m 29s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb73832b79e54ecabeaffd1bf469a54b : FAILURE in 5m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d84db28c81970249a66f00c91a9ad2cc1c95fa4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"date":"2023-06-17 13:32:12.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"315a2ae7c58f09f8ef430b4ef0f85eb1cd2ef71c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-17 14:23:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fce5b84879fc4592a3b72810116e0fce\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/af8969afa821417a92779eb67368230b : SUCCESS in 2m 23s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f335d2c93bc44649c9dafaf7f627f54 : SUCCESS in 2m 21s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/c2cec84030264a31b6c3c36c675519a6 : SUCCESS in 4m 27s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/86a468c945674a61a4f80615057710ec : FAILURE in 10m 40s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1bae8d1ddc714bc382f2ff2d6e9fdd27 : FAILURE in 9m 19s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/a7344a64e5134853975f199760893bdd : FAILURE in 11m 57s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f86766c57746428bad1f93a52c353c7f : FAILURE in 9m 09s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc05fa8538504e72a9e2bc887cb0eb1d : FAILURE in 6m 28s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ba87f49fd254712b27e94eeaa2ac38b : FAILURE in 6m 00s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc4865aeb11e48dea4f73b4af3b8b9a1 : FAILURE in 5m 55s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c665bf6061f44a8eb71d53b0a21c0dc3 : FAILURE in 5m 48s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce1031c4dd6845759237004f12456b84 : FAILURE in 6m 18s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8bc2426b9be74e428803fab35e3f5216 : SUCCESS in 16m 20s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c5f824d4b2e84c0c8ebbcc95bc88fcd1 : SUCCESS in 47m 44s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd2fd2fc9a204cb99e8164c53155f48e : FAILURE in 9m 36s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ffb9a4c483074d5ca6413d052e00d4d8 : SUCCESS in 49m 48s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b6649cd8c7dc4111858ccc0f61749e27 : SUCCESS in 43m 18s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1a5b172660ab4ab5836aec6046beae2c : SUCCESS in 42m 25s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/150945be5c164c36ad41bd323049e664 : SUCCESS in 21m 07s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3eb72644af614ab4bc7d1a16e049fb06 : FAILURE in 6m 32s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a411feaa02e44a2a28fe359a6e2cb26 : FAILURE in 6m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fface758e74e9e0e0b412c17ab2d72fe5dce29bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"date":"2023-06-17 15:01:05.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"05bb542f63c5fae551ee94959cd5c4f93886936f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-17 16:18:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0873a2e984a6460ea47f50d44e0d9132\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8fc8323f8a5d4c98ab6e00d456d55206 : SUCCESS in 3m 15s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1411d7f5f4724805825fa7d0a9d2b7d4 : SUCCESS in 3m 25s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/af2460c03df24d779413de6080ce671d : SUCCESS in 4m 40s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/5fdf10c7871c4c6ca2c2261760e29001 : SUCCESS in 11m 03s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/327ba9d2c6b64c088f8a4ca37c0e97a2 : SUCCESS in 12m 54s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/411a68e52e994da79fbcf38d3e2aece5 : SUCCESS in 16m 00s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7416479960d24083bf3b6e07cf7d2c94 : SUCCESS in 11m 56s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a61436dc483c4f49960b1de90dc6ceea : FAILURE in 6m 29s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9320cf194134443eb02294c983e0198b : FAILURE in 6m 48s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f32ac441d02d4de58e82214b8fe8db93 : FAILURE in 6m 22s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6fcb611ac32644f7abecaed9b2c34b8d : FAILURE in 9m 22s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dac57994c8504b0e9ab8d9625d4b6a83 : FAILURE in 8m 52s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b661458d98824e32b48df3221fd8ce99 : SUCCESS in 18m 15s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a4fea65d115d4b7aa47495fcb90b97ab : SUCCESS in 57m 59s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5adc1f8ae95243c4b56ee3715cdc59fd : SUCCESS in 56m 27s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8fe9e4ce1caf4e6aacd2c10d5005e123 : SUCCESS in 1h 10m 03s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fd526faa72d94144b4e9758731635cf3 : SUCCESS in 55m 44s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5969ae416d1b47dca8e092053f31212e : SUCCESS in 56m 02s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/7221d83408b24f42bd639eddb1ba9cb8 : SUCCESS in 15m 43s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b206e59880c4a84a34ab87281059def : FAILURE in 5m 51s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/078d57f7261d48ae88a1e55873f538fb : FAILURE in 5m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"66c579b49744b708bf368edec3635e144ef69fbc","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"date":"2023-06-17 16:21:11.000000000","message":"Topic t removed","accounts_in_message":[],"_revision_number":3},{"id":"e258783ece2a875aaeaf967b1c8b26d593284d3f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"date":"2023-06-17 16:21:16.000000000","message":"Topic set to replace-validate_legacy","accounts_in_message":[],"_revision_number":3},{"id":"e24a3c161d08479eb3ce7defa1d4f911399b537a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-06-19 19:43:05.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"756eb4a843e2b9e33300bdf7f82f562fd4c8484f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-06-21 08:36:14.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2877303a3ef8c7f9137a6e341b8131a4a10a9ab0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 08:37:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3bb338bc66127c0a4b6a1aa0b3fcc34d9dbdef5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 10:08:47.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/17c3244a2565423aa69c8b980fcf8ae9\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/19c443bcc1a94d0489810aba195675ab : SUCCESS in 2m 56s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8950d27af88b4263b1af8422dbb1bb86 : SUCCESS in 3m 04s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/9ba643eef62a4e3da1dfe1c646ea2f2c : SUCCESS in 16m 27s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ceaf03b0c83646a3b9bc9781422b2df7 : SUCCESS in 12m 31s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/4e7a150538d64af697894c5b4f22fd91 : SUCCESS in 13m 23s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f3bf3670ab1b42c49b45b019b1e8ff5c : FAILURE in 21m 39s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5864d4ee158743d39a64000d4e615e3f : SUCCESS in 1h 17m 45s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/60e92f79bbc0436ebd90b79a45f4e41c : SUCCESS in 1h 23m 47s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0bb5228b83fd4230b01de26cab1e888a : SUCCESS in 1h 01m 46s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4314605209604d45b7b4fd3c0846b292 : SUCCESS in 1h 02m 49s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/7f00cb0aead64ebc87e4f322cb2ecc9d : SUCCESS in 19m 38s","accounts_in_message":[],"_revision_number":3},{"id":"475b8c4f57999cc548d6dc811921c264bae78545","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-22 03:44:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"019fb8d88b40f3fe0eb80d8fb678a2cf8333f120","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 05:28:50.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/172f22d3ee1647389946634d40af6558\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2dd5dd0a90946afbe4861177e3830c4 : SUCCESS in 2m 54s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e67c2d2543f4704b4423072540e51ff : SUCCESS in 3m 31s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/d3b0dc4ee58547f6875e0e794903bae3 : SUCCESS in 6m 29s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/400ae7780b0f4841a48dd8d41d80bd42 : SUCCESS in 18m 02s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/08be8c17099b4d438c5ce333e98f61ac : SUCCESS in 12m 17s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/334341c3e3f64e7e8f96ab9dbff9674f : SUCCESS in 13m 34s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/769273942e7f4562a4a6373386976c8c : SUCCESS in 10m 02s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d0f0ee1151041b2a6a0c2a43d5c82f8 : SUCCESS in 51m 44s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9407c5365e484ca78b09788bf4982abd : FAILURE in 41m 19s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57b25bceb41141ab8c576382bd974559 : SUCCESS in 54m 08s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b640833a989442982a92bc09bfadcd0 : SUCCESS in 38m 49s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/288e7e7cf77c4f809d0a90657748d7f4 : SUCCESS in 44m 13s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/36830a3b38614590b589242b4b3e223b : SUCCESS in 29m 45s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9126d384cf0f4eb0ac70d78ec96dd985 : FAILURE in 57m 34s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9d9d65576f8c459d87297836e2b492f8 : SUCCESS in 1h 16m 18s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a0da36de638d43309ac2858564ea171c : SUCCESS in 1h 09m 36s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6702a987c503407bbecbfff31d8f02a4 : SUCCESS in 1h 10m 14s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0e3ede24e1d44fff9bffcf98d29865ec : SUCCESS in 59m 26s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/ae36c83ab68d44b6b7964bca7a495626 : SUCCESS in 21m 56s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/413506cffd8c4f84a2de93c6fe183039 : SUCCESS in 19m 09s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7174a8d1b36643d0b06b5afe5448d20f : SUCCESS in 15m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bcb4890ef7c0a97dfac29d13e482601367dcc8b5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-29 03:23:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0665cac1c67c5e57865cca5a40b3ca8c49c5bb63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 05:02:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53b5503ba2874e8a838dfaf0c7ed0269\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd84f71afccd4fd7bd88ba9f8cacfa50 : SUCCESS in 2m 47s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ece60b92ed647c7bed93e15fb8b69ec : SUCCESS in 2m 47s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/33a1a999c2324af3989a39d5bdf742cb : SUCCESS in 5m 20s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/d756370ef87d4c8fa70c9075e62efce6 : SUCCESS in 14m 37s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e8efc1cca70450b8198660b8fa7fb9d : SUCCESS in 12m 54s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/b365f1c1cc114511b6f464e5cc3e701e : SUCCESS in 16m 33s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/acf5117ba46040b7acf26190793fd989 : SUCCESS in 10m 13s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/491688ed2d9a426d83ee6387d954e3a3 : SUCCESS in 45m 23s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b00965d4c3fd43fb97fb51887b6965f3 : SUCCESS in 53m 44s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/35869cba01804236987268fbbbae68ac : SUCCESS in 1h 04m 26s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/085e9c573abe41b480728aea81bd40df : SUCCESS in 50m 15s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fcc55b7a736040dfbf5893aaf9474180 : SUCCESS in 42m 12s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5c51ff6cb08249359fc537aef79ac41a : SUCCESS in 22m 27s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/44dae1669ad84f3e97fde169b418f0c4 : SUCCESS in 56m 55s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/378959209bed40bfb77b1aa0fec901c4 : SUCCESS in 1h 27m 35s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bdcdc806ea3a42c6a73b75afb47a5c66 : SUCCESS in 1h 05m 47s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/81efc9cee32b40ce8860119bf675fd43 : SUCCESS in 1h 00m 09s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/23217793d50345208284b4e26805155c : SUCCESS in 1h 11m 37s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/8ffadd1a9b3248db88f9d0a6bc7f5b20 : SUCCESS in 19m 22s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c4edb7ecf34042aea9c1f2f22f8c29bb : SUCCESS in 18m 22s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2183bc2caa45405a98e555304647af1b : SUCCESS in 13m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0bd69d70a35d1aeb41e013cbd28287b9b6eac8d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 05:02:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b6691a0b053b6f3fcc1e37349e6a6af362fdd8f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 06:22:34.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cdc18fcb43ff4d1aa870268863256b5c\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/91478181bf784575a103c0892b28151b : SUCCESS in 3m 03s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd7a3f70c2de4e449fa2d2a4e5cce72d : SUCCESS in 3m 02s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/f06fd935f3804206944f5d745a064cfc : SUCCESS in 14m 29s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d8084c4607964a619a8d7ac34e404408 : SUCCESS in 11m 39s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/63e521234731449b91677ed740e6b400 : SUCCESS in 11m 47s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/592d74f8c85a4d0c95925042c2bce91a : SUCCESS in 1h 01m 47s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8386b08b1c32470892982dc2a5bce730 : FAILURE in 1h 19m 32s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/68f5fcd2d51c4e0ca866a0c1f6b4d97a : SUCCESS in 1h 05m 37s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/655ef89fa72b43b499529a89702a943e : SUCCESS in 50m 19s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/92bb8979f95d4e74810e5570914f3c0a : SUCCESS in 59m 06s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/c981169a01aa4fea94fbcdd9b7b0d4c6 : SUCCESS in 20m 57s","accounts_in_message":[],"_revision_number":3},{"id":"36c211b6314b91d56f2692b3382880a2f9580ace","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-12 06:48:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"347b3a4e09e9f409c54426eba6c04ec71ed7e457","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 07:57:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b280cb457584a96aedd096986a21fd7\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d59005e97a94b37aeaf93dbffcc9156 : SUCCESS in 5m 04s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a6c8398686443db9cbba2271a76361c : SUCCESS in 3m 14s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/a54643b89c2d47089b37ee45169ab9ab : SUCCESS in 6m 05s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/121e27f0e2cd48878316285002d9bc99 : SUCCESS in 13m 32s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f516fc089ea54c33a5535d69717e2056 : SUCCESS in 12m 15s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/4272c985b402400cb2baa6606d5bb674 : SUCCESS in 14m 47s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a786077655f540e1bdbc0f370a12d9b9 : SUCCESS in 11m 16s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f90044033ac74dc18a8a16fe30b4ab0f : SUCCESS in 50m 17s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/931ae3c1bb9140ff8213762f5405e16c : SUCCESS in 57m 17s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57b2025cc64742ea93ce58b49c1b25ee : FAILURE in 49m 38s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/44ca82ed8ae545b0997f1299793c52bf : SUCCESS in 43m 24s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b8f8df6535d44b14a3a4076c1b7037f6 : SUCCESS in 39m 49s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4bfc2f9ddfb34ebe892304e270469e88 : SUCCESS in 22m 08s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b8826bd34a3a40df8bfe597c0bb50667 : SUCCESS in 54m 24s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/826da49b575d48e6923c58443dea2cad : SUCCESS in 1h 04m 04s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ca6d1f7070b944e0b4003a38c7125643 : SUCCESS in 56m 18s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f55979b1945a4ea4b61280ae10e36c88 : FAILURE in 1h 00m 14s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5c85a79bb55b45f4bf84ac2b099f2749 : SUCCESS in 50m 49s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/85859b9953e84df38c9e82f2d43d3946 : SUCCESS in 20m 52s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f5fe118cf674274983b75a562cdb59e : SUCCESS in 14m 00s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/394258e52e6a424db8b915779698e06e : SUCCESS in 14m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b0c7881d01334267da3e6da38f228bd392145d14","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-12 10:55:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"29ff7df01fb00bbe2b5a9a653a272635d08875f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 12:12:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/627965727ba6490e8aec0bc4fd6cd1a6\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/818b75edbfe34df68ef6ba2d113dc488 : SUCCESS in 3m 00s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3313dbaa117b4cf98424ecc85a645135 : SUCCESS in 2m 31s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/dccfab4f6c23466da9c9add45b77f0a3 : SUCCESS in 5m 46s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/72c31f91355241c69be9dd2252e53fa8 : SUCCESS in 14m 46s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/153a91adcdde41d794510d267eeaa1bb : SUCCESS in 11m 37s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/d86e13be424e4f49bde49bd9cca27b6b : SUCCESS in 18m 51s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/056067e29b3b4351a99d69260f2738a5 : SUCCESS in 10m 42s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8ade63999f764317aa82cf8dbaee5ab7 : SUCCESS in 43m 19s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05498f8d26ac43409b71d778b56adb48 : SUCCESS in 1h 04m 25s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9b9e9e9f0fb4c1e9ea502937deeed05 : SUCCESS in 57m 20s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c397477684f4919b3e270857f0f7714 : SUCCESS in 45m 24s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/312ce670f35a48b6a5f0074e73166c13 : SUCCESS in 44m 44s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6e549efa333c49e7b0afd508dc9af4c6 : SUCCESS in 20m 26s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3aa38d91ca344cf88c3e04cbc9e510f4 : SUCCESS in 1h 02m 42s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4f109167b538497f9ee302f375466a87 : SUCCESS in 1h 11m 38s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bf4371374777450e831e89b0fe813a54 : SUCCESS in 1h 11m 30s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8563e6bc82de4fa4b8e28be14e017476 : SUCCESS in 48m 30s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b9af4805b6be41228ab5d4b04b115622 : SUCCESS in 53m 27s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/684152597dd14c3296bd0a654f4c693f : SUCCESS in 22m 27s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07f92dcbd9744958a27a2a02aef8a1a5 : SUCCESS in 14m 57s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aac4c139565e4bbb931ef1529db9f0bf : SUCCESS in 16m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8664fc44e9bff9686e0defa4ea16720a52083b40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 12:13:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"15ac323579cfdc682b81b23311b096889be1f047","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 13:53:26.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/279799c161bf4727a1ebcd3afb4a815f\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fbf1702e5f45442597c013d5821628a1 : SUCCESS in 2m 43s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d919c209fd3a4691880094e110d55599 : SUCCESS in 2m 56s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/639c9745f71242eebf472139cdb9a9f8 : SUCCESS in 17m 15s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e13d0b6a4a6f4d00bded287eba9db9e5 : SUCCESS in 9m 25s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/4e6e72b33b8c46d2841830e01c42410a : SUCCESS in 17m 13s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ad9b6746e570430a94f92ddc5232e576 : SUCCESS in 1h 00m 52s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/443254eb7ea54f5d9289f2bba176167f : SUCCESS in 1h 30m 41s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/69f10db4c4634f4487e487d34bba5039 : SUCCESS in 1h 02m 58s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ad203ec9fd174cd781ff6c45a6f0c43d : SUCCESS in 59m 54s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/af0bcf76e16a4baca46fb259e4d1daa1 : SUCCESS in 57m 38s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/b82057cff9e643f98734a97b20b78d16 : SUCCESS in 19m 50s","accounts_in_message":[],"_revision_number":3},{"id":"a34171d5ba003053dd86b6fb65387694a2781680","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 13:53:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"be045c24794d750821a962e1a8f21828073df3f6","revisions":{"fbd3af4c344e7d06746c8f463ad275251de220c9":{"kind":"REWORK","_number":1,"created":"2023-06-17 12:44:07.000000000","uploader":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"ref":"refs/changes/01/886301/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/01/886301/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/01/886301/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/01/886301/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/01/886301/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/01/886301/1"}}},"commit":{"parents":[{"commit":"b44f40170f496e01a03b2c20f17ea97b84d00b35","subject":"Merge \"Deprecate support for networking-odl\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b44f40170f496e01a03b2c20f17ea97b84d00b35"}]}],"author":{"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","date":"2023-06-17 12:43:28.000000000","tz":120},"committer":{"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","date":"2023-06-17 12:44:04.000000000","tz":120},"subject":"replace validate_legacy with proper data types","message":"replace validate_legacy with proper data types\n\nthe validate_legacy function is marked for deprecation in\nv9.0.0 from puppetlabs-stdlib.\n\nChange-Id: Ib124f05394b5f53cace06cd062ee2079afb311af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/fbd3af4c344e7d06746c8f463ad275251de220c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/fbd3af4c344e7d06746c8f463ad275251de220c9"}]},"branch":"refs/heads/master"},"54bf3c563185e4f92a5d9f579d82ac28c5901883":{"kind":"REWORK","_number":2,"created":"2023-06-17 13:32:12.000000000","uploader":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"ref":"refs/changes/01/886301/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/01/886301/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/01/886301/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/01/886301/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/01/886301/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/01/886301/2"}}},"commit":{"parents":[{"commit":"b44f40170f496e01a03b2c20f17ea97b84d00b35","subject":"Merge \"Deprecate support for networking-odl\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b44f40170f496e01a03b2c20f17ea97b84d00b35"}]}],"author":{"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","date":"2023-06-17 12:43:28.000000000","tz":120},"committer":{"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","date":"2023-06-17 13:31:59.000000000","tz":120},"subject":"replace validate_legacy with proper data types","message":"replace validate_legacy with proper data types\n\nthe validate_legacy function is marked for deprecation in\nv9.0.0 from puppetlabs-stdlib.\n\nChange-Id: Ib124f05394b5f53cace06cd062ee2079afb311af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/54bf3c563185e4f92a5d9f579d82ac28c5901883"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/54bf3c563185e4f92a5d9f579d82ac28c5901883"}]},"branch":"refs/heads/master"},"be045c24794d750821a962e1a8f21828073df3f6":{"kind":"REWORK","_number":3,"created":"2023-06-17 15:01:05.000000000","uploader":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"ref":"refs/changes/01/886301/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/01/886301/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/01/886301/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/01/886301/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/01/886301/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/01/886301/3"}}},"commit":{"parents":[{"commit":"b44f40170f496e01a03b2c20f17ea97b84d00b35","subject":"Merge \"Deprecate support for networking-odl\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b44f40170f496e01a03b2c20f17ea97b84d00b35"}]}],"author":{"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","date":"2023-06-17 12:43:28.000000000","tz":120},"committer":{"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","date":"2023-06-17 14:59:26.000000000","tz":120},"subject":"replace validate_legacy with proper data types","message":"replace validate_legacy with proper data types\n\nthe validate_legacy function is marked for deprecation in\nv9.0.0 from puppetlabs-stdlib.\n\nChange-Id: Ib124f05394b5f53cace06cd062ee2079afb311af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/be045c24794d750821a962e1a8f21828073df3f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/be045c24794d750821a962e1a8f21828073df3f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
