)]}'
{"id":"openstack%2Fpuppet-neutron~889441","triplet_id":"openstack%2Fpuppet-neutron~master~I20204bc6c63336d2e04d3e19a058e8b2f23b0778","project":"openstack/puppet-neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-08-01 17:22:37.000000000","reason":"Change was submitted"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2023-08-01 08:05:17.000000000","reason":"\u003cGERRIT_ACCOUNT_16137\u003e replied on the change","reason_account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}},"hashtags":[],"change_id":"I20204bc6c63336d2e04d3e19a058e8b2f23b0778","subject":"Accept a string and hash for contrail_extensions","status":"MERGED","created":"2023-07-25 14:26:56.000000000","updated":"2023-08-01 17:22:37.000000000","submitted":"2023-08-01 17:22:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889441","meta_rev_id":"ef0518a3e7266eb227834af86c7521ce9ba7ca21","_number":889441,"virtual_id_number":889441,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-01 17:22:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2023-08-01 08:05:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2023-08-01 16:02:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2023-08-01 16:02:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-25 16:15:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-28 12:09:58.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"CC"},{"updated":"2023-08-01 08:05:17.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-08-01 16:02:06.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"39240de773ce38cf183fa608caca85b1bd1933f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-25 14:26:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"537e8ac4cc83a2cc26c7c1100697c53620090078","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 16:15:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6b8754a30e6447c5b1925ab5b36b5627\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb2497d3dc2548f4b3a2cb89fc3df8d6 : SUCCESS in 2m 59s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d7282cc0d81c40388baf952cabe5ba3e : FAILURE in 2m 45s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/13f0cc9d71ef4fff9658d8be33860931 : FAILURE in 6m 48s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/798effaa91f747c986dab33d5a5ea454 : FAILURE in 17m 12s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/86cc2ede6fe54477ad2f83b72de8da18 : FAILURE in 12m 45s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/bc6098c92a3a41e6a18d37e4d2a2c1ce : FAILURE in 16m 39s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b85cdc12d28b4e69a2c9bbf6b8ed3bea : FAILURE in 13m 20s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1880f5d7e964a228e3ee564be4f58c7 : SUCCESS in 45m 42s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0086e47da47e4a45865172c9e1781920 : SUCCESS in 1h 09m 14s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/23d28ff9efc34886b907b4cd69763d44 : SUCCESS in 40m 32s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65d596fcad0e421d833abc05607a3722 : SUCCESS in 56m 56s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1966574a470a4c98b03d4f354ff69e46 : SUCCESS in 41m 09s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0ad5e2fe5b72401ab4460d0b11edf683 : SUCCESS in 1h 15m 09s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/82bc478a3933473f909d793aeb5f82b3 : SUCCESS in 1h 23m 24s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6b100ba34dbb4aa1a37868c497935636 : SUCCESS in 1h 11m 59s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9ac686f1a7044271a8a91442ab9d194c : SUCCESS in 59m 27s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2d5a5951f68b4e8eab5bac9271371dc7 : SUCCESS in 1h 06m 54s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/85c50949182d46aaa90f4eedce76fabc : SUCCESS in 22m 06s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a6d6d194ea2437eabffe104c6a93664 : SUCCESS in 15m 01s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99612ce1f3534336a46aed8079638147 : SUCCESS in 16m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e9b45564769d3d2f9d72dba15469f5a721799453","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-26 01:40:42.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a64d694cb914cb602a474671cedfd17c3c3fe4e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 03:00:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac33142097fd48d9805efff3d30eaaff\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24b70465b7544dfa88c8631f9a33b907 : SUCCESS in 2m 35s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b076c6c843144f4da3ce31d1c4f4b7a7 : SUCCESS in 2m 32s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/dfb6d4f4da18432d97cdc452b412daea : SUCCESS in 5m 08s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/a878bec1fb884b59bfc9122f3d5a59e7 : SUCCESS in 17m 52s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/309ed91a6fb94d92a968bcb922c7d61f : SUCCESS in 9m 54s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/fb26a12e5b6d40c4acdaddbd72027671 : SUCCESS in 12m 31s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cc50148c6d374d6f8dc6e11d46790c1e : SUCCESS in 12m 37s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e02429b569c40f3ba75d87d29c94347 : SUCCESS in 51m 12s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e17ec03cf68a4e17bef6a56988b44648 : SUCCESS in 44m 33s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/50f030bf668c45bbb4b7fe405ce620db : SUCCESS in 39m 40s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/01f1129a2a75401c8a9c7ec1e0b71a7e : SUCCESS in 48m 54s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c2774446f8e2454eb3565d593e40228e : SUCCESS in 52m 37s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/45b5ea06d3c14b3781ea88478831e3ab : SUCCESS in 48m 44s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/004b0d36b3d343a6872cc5489bd826a4 : FAILURE in 1h 18m 03s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fd4740e4286c4a8e8311322a5ff719ef : SUCCESS in 1h 11m 33s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/12c3f7d3367d46c68bcc3f53a151feff : SUCCESS in 1h 02m 40s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/880585571ff74999ba4d26b7562aa370 : SUCCESS in 57m 16s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/7a439dd1ead345bc99ca2571262285c5 : SUCCESS in 15m 21s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/58f4c48ede284fab844dd6124502d3f1 : SUCCESS in 16m 27s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab50928d8b5c4bf2abc6aeac749c5a29 : SUCCESS in 15m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"10b61492e6609c7b4321fb5586326a06dcfb31d9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-26 03:57:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"697ae1b7dd9cf1acd98148990f8cd24c5cae255c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 05:02:05.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/60808e64e3a04bbc9873f46483da5270\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/128bf8d12cb64a588e25ad9909254cca : SUCCESS in 2m 42s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e061fcf354a84bc5be1d5c718591cf3c : SUCCESS in 3m 12s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/1d05ce3007c04da9a83d8c5dbe14e157 : SUCCESS in 5m 36s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/143ddc8e9b054582917114e57ba54abe : SUCCESS in 13m 11s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0d1014268b6a4c0bb2a3295134f47ddc : SUCCESS in 11m 32s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/eb5e9b35c7a94a6ca56b6e5761890b8b : SUCCESS in 12m 48s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2045ba5f13c4098a29656b3932ac52b : SUCCESS in 11m 54s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7903133097484ca3a63669a8363de753 : SUCCESS in 48m 42s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8fe3bf9bbfd4d96bb623ec3fceb5a92 : SUCCESS in 59m 26s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ff0e2a04e5b4dc0988c6a6b76541712 : SUCCESS in 53m 29s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/390c6de6265e4b9da327f9185f09c86f : SUCCESS in 46m 21s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6cea58804e1540e3b4595fe97c6696dc : SUCCESS in 55m 50s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c26f8b77a3c1407da652ffc68b7c341d : FAILURE in 22m 17s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ae2419518e6b46ac943ae7eab3bb51f2 : FAILURE in 21m 50s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b4cd221287984019a96e0d24cb59c875 : FAILURE in 21m 55s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9486151f83124c67a7e57addc9d11e7b : FAILURE in 23m 05s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c69b29169ea14b30a9a2357790fb6cd7 : SUCCESS in 48m 22s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/956dcce31a8a459580af193e4508714d : SUCCESS in 21m 14s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3a4e075d27c4e15b6f81146aa5dc2b9 : SUCCESS in 15m 27s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e367a2c8704423fa1d7dc88e51c72ca : SUCCESS in 13m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"23707a0aa0edacc87e998a5dda1abda442fd3579","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-28 02:22:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"054eeee5106ab8557ee697ba5270bd05953b95e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-28 04:03:45.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6123baa796564ede9ce6d4a4f0808070\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c48009fe90143ae9f51760f16d00955 : SUCCESS in 2m 43s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/10b0c603a38d4c4fb97ec7bf4b1a3e87 : SUCCESS in 2m 39s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/0fcae82f7cac46dc812cced1e749410b : SUCCESS in 6m 24s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/bae1aefd98a9406f8a8a0c15c4d5f1ea : SUCCESS in 18m 33s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7d4a215f0904de7af1c58e15ff7deee : SUCCESS in 11m 51s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/0bbf86221b2f41c883dec5ab2ad6c3be : SUCCESS in 18m 18s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82c0190cc06f4adabbf1f9e2c5661c76 : SUCCESS in 12m 51s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/acc3c2b2c4dc43e9a5ef0ffa4a007dec : SUCCESS in 54m 21s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/28ecf522f0cf496f9615f171c4091993 : SUCCESS in 1h 04m 44s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9108a04dd2634f4a869c44782e5ef8ce : SUCCESS in 51m 47s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d328980f0304e6d87ddff73cdc84b6a : SUCCESS in 46m 43s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b922ba340bca4cf5a94a14c36b85396b : SUCCESS in 46m 56s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e0f1d9c7b5b042099814502a1b17de63 : SUCCESS in 1h 16m 42s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ff8031e10f8e4d0dbbdaa72af5e50c30 : SUCCESS in 1h 32m 35s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/35601c16e65440ad8504c6c1e3606fd7 : SUCCESS in 1h 08m 38s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d4d274ed144c4cc59ce9a79965eae33d : SUCCESS in 1h 07m 26s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/de262d8e40704bd7b95b2cb0a2d5c5da : FAILURE in 1h 14m 25s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/1268318a14a6470ea9766b953e8c626a : SUCCESS in 22m 03s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d3aac3061854f5591635926043f3750 : SUCCESS in 16m 48s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/32b59f7a712448a8a9981281ceffa0ec : SUCCESS in 14m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d333b86f2c0703ed2ce7e02d16f5d18804966a27","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-28 04:08:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"561064abe0e33fb423765f0bb20c2508963a87ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-28 05:49:18.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e5c8c707da34ef29585c1be6cd9511f\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f8d1f6b3bb0249408980be8a0adfbfe6 : SUCCESS in 2m 54s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f94f9427b5a64fe7be71446d22fcdb92 : SUCCESS in 2m 45s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/283286d7d3244893984cc92e6708e358 : SUCCESS in 5m 56s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/5b1c127e18054bb2bb84d533b20b7fb9 : SUCCESS in 14m 58s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dbc104cddca24d16a378b06b711eead8 : SUCCESS in 11m 43s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/7a8cf713a25342f980fba668155956ec : SUCCESS in 15m 39s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d2d626e1da0490bb262e59cd541cb07 : SUCCESS in 11m 38s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07d8869eebc4441b919f58d24be18c81 : SUCCESS in 44m 28s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/68062123030748d89134f80298698289 : SUCCESS in 1h 12m 03s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5aa13a27860242039a7d83f5920433c7 : SUCCESS in 54m 48s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/990fe3736ddc4e6e8d8b454f44004c93 : SUCCESS in 51m 23s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/32df7ae6162b41c0817948d85b824adb : SUCCESS in 49m 43s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2109c3788b004828a88335d386d800cb : SUCCESS in 1h 00m 44s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/34ca0a2ff867476594ca186bfcf180d5 : SUCCESS in 1h 31m 33s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/85418fa34a51462fab5ebadc13db5397 : SUCCESS in 1h 18m 58s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/142d33c4fc1048c7a1a8250267ca00cf : FAILURE in 55m 33s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5f3bb33ce38345f3b7d510a475f31c24 : SUCCESS in 1h 09m 26s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/e3148bb4959b4981a15c5552baa8a4dd : SUCCESS in 18m 22s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b9812dc5b5b4edfaef5563dd69e09aa : SUCCESS in 17m 36s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b358d6b3f9e64a1187c214a6b54261ae : SUCCESS in 16m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2a4eb6b6bb3f09a9c27c33908adfa1a0b6434c29","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-28 07:38:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f474ce138e0b8053e7e6d864c5cefabd38ce6ba0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-28 09:37:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19197e62f6eb440391407015ba4c1c43\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6d5ebee7a09d46c38ad247a562668f3c : SUCCESS in 2m 53s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/51e7f1efb82a4f8c8de1c35b6542c866 : SUCCESS in 3m 23s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/b6cd5117e6334c88b0f1f0b3f90fb140 : SUCCESS in 5m 11s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/15789361dc2e44fe81fa77acf166e7c7 : SUCCESS in 18m 22s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/926e6b9483d74c01a309a78d91170f1a : SUCCESS in 9m 51s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/5e4b0f855f6341ddbdd106ee9c93741c : SUCCESS in 16m 05s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e4d4a6af1fd455b99eab4d0fbca4979 : SUCCESS in 13m 03s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7100ea78ff414be181bbc36851ccb63b : SUCCESS in 56m 54s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/448089b772ba46ff8dc650481153a6af : SUCCESS in 49m 50s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/71f32eb6f1124ee4b2470a8c6a101bf5 : SUCCESS in 51m 13s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/89153a530e8f4ac9a06f04ffdb067747 : SUCCESS in 46m 13s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cc6e32a3034b4450a450e0cf7c7d1754 : SUCCESS in 47m 52s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/151dc0e9a4324e1ab2f050c13f92de67 : SUCCESS in 1h 05m 20s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c7d4964a70d44ce397f1019f170b120a : SUCCESS in 1h 22m 08s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/080e6a1aec2f4e11b6b99ce2a7736ce5 : SUCCESS in 1h 10m 00s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d932ff6e2cc9410c922afeeef509f65e : SUCCESS in 1h 03m 26s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/610bc8dcc8724410b31a7170760d6ba6 : SUCCESS in 56m 22s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/839695f6ab0d46b7b73809a0a80ecbf3 : SUCCESS in 19m 07s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8688016b9da14d2cb2e322d5d6c7a237 : SUCCESS in 19m 29s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb1c74f599374498aa0dce08ad5ae92f : SUCCESS in 16m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"09304c0e2d028414801ac844fc1eb42ad24372c7","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-07-28 12:09:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"daff3d4e96c1432621caa1c0efd0900ed66ea937","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-01 05:02:29.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"ce2a36a67459573b2689ccf6f4f2912c32e4fa5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 06:38:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62cfe9eba3a04648bc6adbe9647748e0\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c4af34fa71bb4230bcdf78a5e44b5603 : SUCCESS in 2m 45s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f0df50da5da64aa9916f3ff4182732e9 : SUCCESS in 2m 36s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/bc8eedf826ce480584219722f5ea562a : SUCCESS in 5m 42s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/baa1447f37df4d52bb58909b5b5d98a8 : SUCCESS in 15m 36s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b9694fe3858459692c4105398ccab63 : SUCCESS in 10m 55s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/2104fc7bab7b4c84bf74a88d76ef4f47 : SUCCESS in 17m 19s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d13712f108d14fc5bd222ffe9e8512e6 : SUCCESS in 10m 54s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe8cb50ed6834975bb2e35688c5dac11 : SUCCESS in 1h 10m 00s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57fefa558a6d444cbe9845b12629fa8d : SUCCESS in 52m 59s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1f169960717b40458e99a32092f2063e : SUCCESS in 50m 01s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d74d48ac321466eb81999aacf8990f2 : SUCCESS in 51m 42s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0abd766defcc45619012d05d5a702700 : SUCCESS in 34m 39s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9032f7aa45384a9f8c66a3607aa9f897 : SUCCESS in 1h 03m 13s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/aa953f44c4664bfc9d87fd7a2deafb22 : SUCCESS in 1h 27m 51s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9ad4ce86a9fc46eea633f9a65aa8be63 : SUCCESS in 49m 46s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7278d9d3519348548b114f9cb2c85663 : SUCCESS in 1h 00m 36s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cd50f95b1e5e450ea6559ab83933a78b : SUCCESS in 51m 29s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/55cf071598e14cfea05fdbb95ecb1697 : SUCCESS in 20m 27s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eed2f028b1cb426b8bb9f44bee92b471 : SUCCESS in 20m 29s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/14cf2f76df874354a220818201d7c56a : SUCCESS in 14m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0dbb44d929767e5d5fa860630cee846501875477","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-01 07:06:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"297f9b39f315c709a7280adfaa8d38cfd928310c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-01 07:13:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"eb0c92e3098ccaa09ab460b45e290c983b38f6d6","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-08-01 08:05:17.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"33cd75c17e069b4fac0537278519f036b0d25cca","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-08-01 16:02:06.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2c96d2cd1988bf0998c3ebce08bff54b580704b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 16:03:01.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"86ed5a03d6d1880031bd1bf5409fe697daa7d6a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 17:22:36.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e12a36a968a4fada8a784e9aee3ee56\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52bee578d9d645b1bcf22c724ef6e705 : SUCCESS in 2m 29s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4c255b528874bf780945c490d3de759 : SUCCESS in 2m 35s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/16f2db0974d34a0da8a15f268dbbbc72 : SUCCESS in 11m 58s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8cd87872473f49929d3e15a33d7a3dd7 : SUCCESS in 12m 33s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/126aab927c6645b9b2a4c38ba2cbcf6f : SUCCESS in 15m 47s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dfa1c0b3632241a9a48f5b130cc0467e : SUCCESS in 1h 08m 46s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/18c926dfe9d24b039d91f88a6b21501d : SUCCESS in 1h 19m 01s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8b6ff8e4a18f45e880cdf6d6def7927e : SUCCESS in 1h 06m 08s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b845df4e89ec479dae1cd7fcb09b28d9 : SUCCESS in 1h 01m 22s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b88cc5c302df4b8fbce879dd147332ba : SUCCESS in 1h 00m 39s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/ebe06463644f4130a574d739ec82898e : SUCCESS in 19m 40s","accounts_in_message":[],"_revision_number":3},{"id":"ef0518a3e7266eb227834af86c7521ce9ba7ca21","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 17:22:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d70cabeb526c17dc1e4ce1966ee275e48b521b3c","revisions":{"7ac26c3f7f542d02b97d72cbf24a38adc57a3705":{"kind":"REWORK","_number":1,"created":"2023-07-25 14:26:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/889441/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/41/889441/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/41/889441/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/41/889441/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/41/889441/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/41/889441/1"}}},"commit":{"parents":[{"commit":"dbad032e33586748ef96caf03571fbbc75a0fc8d","subject":"Ensure purge_config takes a boolean value","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/dbad032e33586748ef96caf03571fbbc75a0fc8d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-07-25 14:24:04.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-07-25 14:26:48.000000000","tz":540},"subject":"Accept a string and hash for contrail_extensions","message":"Accept a string and hash for contrail_extensions\n\n... following the logic we have for DictOpt options.\n\nChange-Id: I20204bc6c63336d2e04d3e19a058e8b2f23b0778\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/7ac26c3f7f542d02b97d72cbf24a38adc57a3705"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/7ac26c3f7f542d02b97d72cbf24a38adc57a3705"}]},"branch":"refs/heads/master"},"3f0639f18a80d20a397a0078ab89b1fb8b147056":{"kind":"REWORK","_number":2,"created":"2023-07-26 01:40:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/889441/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/41/889441/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/41/889441/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/41/889441/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/41/889441/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/41/889441/2"}}},"commit":{"parents":[{"commit":"dbad032e33586748ef96caf03571fbbc75a0fc8d","subject":"Ensure purge_config takes a boolean value","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/dbad032e33586748ef96caf03571fbbc75a0fc8d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-07-25 14:24:04.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-07-26 01:40:42.000000000","tz":0},"subject":"Accept a string and hash for contrail_extensions","message":"Accept a string and hash for contrail_extensions\n\n... following the logic we have for DictOpt options.\n\nChange-Id: I20204bc6c63336d2e04d3e19a058e8b2f23b0778\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3f0639f18a80d20a397a0078ab89b1fb8b147056"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3f0639f18a80d20a397a0078ab89b1fb8b147056"}]},"branch":"refs/heads/master"},"d70cabeb526c17dc1e4ce1966ee275e48b521b3c":{"kind":"REWORK","_number":3,"created":"2023-08-01 05:02:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/889441/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/41/889441/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/41/889441/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/41/889441/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/41/889441/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/41/889441/3"}}},"commit":{"parents":[{"commit":"dbad032e33586748ef96caf03571fbbc75a0fc8d","subject":"Ensure purge_config takes a boolean value","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/dbad032e33586748ef96caf03571fbbc75a0fc8d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-07-25 14:24:04.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-01 05:02:21.000000000","tz":540},"subject":"Accept a string and hash for contrail_extensions","message":"Accept a string and hash for contrail_extensions\n\n... following the logic we have for DictOpt options.\n\nChange-Id: I20204bc6c63336d2e04d3e19a058e8b2f23b0778\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/d70cabeb526c17dc1e4ce1966ee275e48b521b3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/d70cabeb526c17dc1e4ce1966ee275e48b521b3c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
