)]}'
{"id":"openstack%2Fpuppet-neutron~889679","triplet_id":"openstack%2Fpuppet-neutron~master~Ib3fcfa088414512fca3882e8b1bd94d1a5ca0e1b","project":"openstack/puppet-neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-07-27 08:36:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib3fcfa088414512fca3882e8b1bd94d1a5ca0e1b","subject":"Remove unused neutron-db-sync-service tag","status":"MERGED","created":"2023-07-26 08:26:07.000000000","updated":"2023-07-27 08:36:03.000000000","submitted":"2023-07-27 08:36:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889679","meta_rev_id":"203da322abee56de24fc62892cecd7d4a3bf7233","_number":889679,"virtual_id_number":889679,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-27 08:36:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2023-07-26 18:09:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2023-07-27 06:01:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2023-07-27 06:01:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-26 09:45:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-26 18:09:02.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2023-07-27 06:01:35.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"fc8c10f3e18d45033b63c743d87bf4ca588219e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-26 08:26:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92390dc6732bd0fd1dcd566353109593ebf90d87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 09:45:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e41f8176a624171a0d1b39441cbc083\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b1138e3a90dd4cdf904f19f666244f90 : SUCCESS in 3m 17s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d29274113f94a849daa5e0d6bbc8281 : SUCCESS in 2m 34s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/37e0e117394c461a8271536f3ce7865b : SUCCESS in 6m 07s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/50365ea8e5694a9c82192d164dc921e2 : SUCCESS in 13m 28s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d16c836603104cbe816ff101af69d469 : SUCCESS in 12m 16s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/3ee204f081b24e88ae86b9ab8af1492f : SUCCESS in 14m 32s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2268ba53a03e456ba0f92e7eece1a2e4 : SUCCESS in 11m 22s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7909395d533747dbaec3aa648f261428 : SUCCESS in 52m 02s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b2328c8fa7b421580b9ea9e646a6678 : SUCCESS in 1h 04m 08s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bfab9cb65e3a406a90ed0aeb3bdeacb0 : SUCCESS in 1h 01m 27s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57a6fef4e498403089471758890ccce1 : SUCCESS in 59m 27s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d615096588ab4cdb8f1e071811e6cba3 : SUCCESS in 34m 27s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2d4f82879c12453a8892f0b117fb05eb : SUCCESS in 52m 48s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f27ee0052c9b48b48eef9bf193c8bf88 : SUCCESS in 1h 13m 51s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/429cde8527744b53bb0d29bb132c59e1 : SUCCESS in 1h 00m 20s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3799d10d4063420794a1ef0635a25965 : SUCCESS in 53m 47s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/53b6c68665c74ef38bf407f14629e1fc : SUCCESS in 53m 51s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/9fbf0c1cd8784f8bafcf045c505512d2 : SUCCESS in 20m 32s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ccd5cf03029041ac919af2ee03189ad2 : SUCCESS in 14m 16s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4912092b00df472aa67d1ff4d6899ccc : SUCCESS in 14m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3f545c51af92dee2869af30c11c022437f9819a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-26 12:39:30.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"78e95a3bbbfdf110a091bf5fe5ba0b8db77caf7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-26 12:42:10.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"16c8fe1111652965500033f23111e5519af37b0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 14:58:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/018f4826d6114eeda91d0d5dd064e8c9\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3efc5be518e948d48fe5e9437a58e5c4 : SUCCESS in 3m 44s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0d274ecea8e2417ab8b39035e147789e : SUCCESS in 2m 38s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/e771389948e346f9b055db40628b505a : SUCCESS in 5m 44s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/1220797c7eb549cebc7cc676906b7ee1 : SUCCESS in 17m 00s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6d68a10bf7b74fcbb15b17c980853ccb : SUCCESS in 12m 52s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/7e596f9eb417424a89fa1226b16cd431 : SUCCESS in 16m 50s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0cda6fe474a749d49d6f0c272a514427 : SUCCESS in 12m 13s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8098e1ac8bfe4623b491766b16e63e5b : SUCCESS in 53m 01s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a1569d9f276b4888b8b6bd9596c33972 : SUCCESS in 1h 03m 06s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9cf3db8e5dfa4b72982c1e38f3f99e74 : SUCCESS in 52m 44s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d06347c6d9d4c04830774c9d62a4b10 : TIMED_OUT in 2h 12m 14s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/20f468cdca15467fa047060fd341909b : SUCCESS in 43m 50s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b21f570df0ba4633b6ea5a9fe6b98405 : FAILURE in 1h 06m 09s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/af024fa318884f6cb93e2381db8e3f7a : SUCCESS in 1h 15m 14s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d9e94b324bdd4e74b1b9eda1f48135d9 : SUCCESS in 1h 11m 41s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/68a3bdce4e864b19964e7815983f6eab : SUCCESS in 58m 24s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cfe5f2dc3a5640dfb4ae1a2da1d66841 : SUCCESS in 52m 09s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/2bccd564d43f4fd3afa8c69d7bf7ba87 : SUCCESS in 23m 32s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e8c5a5ce63384b869f243c01a2e6243b : SUCCESS in 17m 07s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/28eba4d661ff4ce189a79f1134e02390 : SUCCESS in 18m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0aee56410d3db5f6e00461b1fdbd9930d2c7a7e3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-26 16:14:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"15e8d7e853814fb99736df8de8e5b80c61f07464","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 17:28:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e9d8d82a36341fbb8e63dfe2733aa27\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/874dde67116b4fe78514e58b516a0227 : SUCCESS in 3m 03s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/030f05cc8ac44fd6b8374205416b244b : SUCCESS in 2m 52s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/58182e753bb6456688857a7366caee99 : SUCCESS in 4m 43s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/3952bfb6f2504f449f2222f777091564 : SUCCESS in 14m 19s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d89894255b564d5aad889861ab8f641b : SUCCESS in 9m 34s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/3f62d75117ad4935b0c0d08cc4525f78 : SUCCESS in 18m 18s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7115b358475406b944b19bd387beb12 : SUCCESS in 10m 10s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ecb17e4496b4078804071d257cc8c42 : SUCCESS in 56m 08s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e770998d8934e3c9a71ef73b4b5083c : SUCCESS in 1h 01m 03s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1488523213fc450db768772e294c72c1 : SUCCESS in 53m 43s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f38d424ad8b04157aa78c30ff9a99985 : SUCCESS in 59m 18s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7403190d92af471a90ec6efbd67fbed5 : SUCCESS in 50m 23s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e47d77f00def40c699afb5d8202053f6 : SUCCESS in 55m 42s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c4bee233d17415dacbaadb21c489d88 : SUCCESS in 1h 08m 55s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/805df149013640dfb5614874ff31b169 : SUCCESS in 1h 06m 44s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8036e7097d684755a25b051a0edc25b6 : SUCCESS in 57m 02s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c04f8105d0dc453dae4b01be0acb100e : SUCCESS in 54m 47s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/330b1bf34b49466e8f4405c94d9ee234 : SUCCESS in 23m 16s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/511abad8426c46b2b3c29080e3b6b387 : SUCCESS in 20m 23s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/900d6afc23934760ad7260a4e177c566 : SUCCESS in 15m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f9b288170fc2e664ebaa8573a1ceafc89c8f07c2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-07-26 18:09:02.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"171afdf65023ef40f6206cbc895e913795624fbe","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-07-27 06:01:35.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"051b11808639a63b7ace1724c8c121f58cb15522","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 06:02:35.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"632fdbfde6502a34c2e73985367d70fc4bfccc88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 08:36:03.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4a7d34ef7f04948a85cb38d1d9d996a\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8ad44e7889e34a7fb18905e7601fae03 : SUCCESS in 2m 32s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8b80544607b48e2bb9856e6cd7976c3 : SUCCESS in 2m 30s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/b8ad76a1336942a1987dd2e5bd31ac5d : SUCCESS in 12m 35s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65e268324be145e5bc309395d2d29230 : SUCCESS in 11m 47s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/50a53e8053b946a183ce5e5ed5ceb1a3 : SUCCESS in 12m 49s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d4b7f716c79f4863b61353ee89d6dd91 : SUCCESS in 56m 09s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cd5900fb7dec44f48c77926799ec4094 : SUCCESS in 1h 25m 56s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e3314bfb219d4d3294ffc9d692f38c5e : SUCCESS in 1h 12m 57s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ad6696080fa746da93c27784b0158503 : SUCCESS in 58m 14s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/808a7b6325a44d81ab1e55fba136d220 : SUCCESS in 57m 21s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/22f5579f75a24b42a28a45654a9e7a80 : SUCCESS in 19m 52s","accounts_in_message":[],"_revision_number":3},{"id":"203da322abee56de24fc62892cecd7d4a3bf7233","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 08:36:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"462573bd459059b3d018f7bd520d842106df1acd","revisions":{"13a69255dcd92c4ba4610676802cd8f334e157ee":{"kind":"REWORK","_number":1,"created":"2023-07-26 08:26:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/889679/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/79/889679/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/79/889679/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/79/889679/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/79/889679/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/79/889679/1"}}},"commit":{"parents":[{"commit":"7e1f8d71af7f132994a692236faea3e712fada7d","subject":"Remove redundant deps for some resource types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/7e1f8d71af7f132994a692236faea3e712fada7d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-07-26 08:25:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-07-26 08:25:09.000000000","tz":540},"subject":"Remove unnecessary neutron-db-sync-service tag from ovs-agent","message":"Remove unnecessary neutron-db-sync-service tag from ovs-agent\n\novs-agent is irrelevant to dbsync.\n\nChange-Id: Ib3fcfa088414512fca3882e8b1bd94d1a5ca0e1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/13a69255dcd92c4ba4610676802cd8f334e157ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/13a69255dcd92c4ba4610676802cd8f334e157ee"}]},"branch":"refs/heads/master"},"35005e50f3e904d3147c5784ee48a30f37e36ad4":{"kind":"REWORK","_number":2,"created":"2023-07-26 12:39:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/889679/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/79/889679/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/79/889679/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/79/889679/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/79/889679/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/79/889679/2"}}},"commit":{"parents":[{"commit":"7e1f8d71af7f132994a692236faea3e712fada7d","subject":"Remove redundant deps for some resource types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/7e1f8d71af7f132994a692236faea3e712fada7d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-07-26 08:25:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-07-26 12:38:14.000000000","tz":540},"subject":"Remove unused neutron-db-sync-service tag","message":"Remove unused neutron-db-sync-service tag\n\nThis tag is not used by any logic and can be removed.\n\nChange-Id: Ib3fcfa088414512fca3882e8b1bd94d1a5ca0e1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/35005e50f3e904d3147c5784ee48a30f37e36ad4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/35005e50f3e904d3147c5784ee48a30f37e36ad4"}]},"branch":"refs/heads/master"},"462573bd459059b3d018f7bd520d842106df1acd":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-07-26 12:42:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/889679/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/79/889679/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/79/889679/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/79/889679/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/79/889679/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/79/889679/3"}}},"commit":{"parents":[{"commit":"aca88eb0817b3a571c5008b5ffee8c6c0dddeb19","subject":"Remove unused unit test implementation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/aca88eb0817b3a571c5008b5ffee8c6c0dddeb19"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-07-26 08:25:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-07-26 12:42:00.000000000","tz":540},"subject":"Remove unused neutron-db-sync-service tag","message":"Remove unused neutron-db-sync-service tag\n\nThis tag is not used by any logic and can be removed.\n\nChange-Id: Ib3fcfa088414512fca3882e8b1bd94d1a5ca0e1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/462573bd459059b3d018f7bd520d842106df1acd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/462573bd459059b3d018f7bd520d842106df1acd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
