)]}'
{"id":"openstack%2Fpuppet-neutron~891456","triplet_id":"openstack%2Fpuppet-neutron~master~Ic057a86a67d819e49ff2965bc3387fd63a8e76a4","project":"openstack/puppet-neutron","branch":"master","topic":"apache-vhost-setenv","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-08-17 04:46:53.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2023-08-16 14:28:31.000000000","reason":"removed on reply"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2023-08-16 13:33:57.000000000","reason":"\u003cGERRIT_ACCOUNT_16137\u003e replied on the change","reason_account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}},"hashtags":[],"change_id":"Ic057a86a67d819e49ff2965bc3387fd63a8e76a4","subject":"Revert \"apache: Expose setenv\"","status":"MERGED","created":"2023-08-16 12:26:37.000000000","updated":"2023-08-17 07:56:24.000000000","submitted":"2023-08-17 07:55:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"revert_of":890927,"submission_id":"891456-apache-vhost-setenv","meta_rev_id":"d51ba84fc48e8875a6d2fae7c39b756579ae748c","_number":891456,"virtual_id_number":891456,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-17 07:55:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2023-08-16 13:33:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2023-08-16 14:28:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2023-08-16 14:28:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-16 12:26:37.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-08-16 12:26:37.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2023-08-16 12:26:37.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8f9e4da9ccad98a1d2f4bc5f5f0a0df910535728","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-16 12:26:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2f56b76a87151c698f900d63b9ab45db5607145","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-16 12:27:06.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f7b4cfd5b276a5dacac18e7dc1f489b6febdc3e7","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-08-16 13:33:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"46b02a699cd121547fdc0459a3c2737df9023157","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 14:03:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50d4ccf9af9940f09fe92dc5e7d09ed5\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ee49b5cb741441ca89d32c0041730d3 : SUCCESS in 2m 38s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c7604d4b45854566809074ff9b7e3d8a : SUCCESS in 2m 41s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/07df5af619b642d18ebd8d5a7374a3df : SUCCESS in 6m 18s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ed65fc300e74f09ac96a037a7673245 : SUCCESS in 12m 01s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7aae835bad0c4b4ab188152275ff24b8 : SUCCESS in 12m 45s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/255e2e031f9641d2bf36041139898fd7 : SUCCESS in 13m 05s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d02dcc5357fc4f139baea8576a6b228c : SUCCESS in 12m 55s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b997b48731f545fcb134e96b4eff2059 : SUCCESS in 42m 14s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c67e9589a9a544b6b45aa1f2b1f65030 : SUCCESS in 1h 02m 12s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a50f4fe71624d17b8158cc5863f2e6c : SUCCESS in 41m 42s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5496ff4ef2954c409428bed39e9e6d50 : SUCCESS in 45m 10s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/470fa20deebb474db11b1b7b4c01a534 : SUCCESS in 44m 27s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/513c49751c624fea8f425ec14190745b : SUCCESS in 1h 01m 27s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/01ac9aedef834beaa3710c045ed5e838 : SUCCESS in 1h 33m 47s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8e60c2fe641e423f801fee9525b112a9 : SUCCESS in 1h 04m 12s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bfbea6b8ba8d4340bc691696a58b2677 : SUCCESS in 1h 10m 58s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/058c6741c3604de1a3dbdc1623842747 : SUCCESS in 56m 35s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/b601f577e3de45858e313f97a70cdb6d : SUCCESS in 21m 10s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b3fc30fad7e4bc587366fd221b0b618 : SUCCESS in 14m 40s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e114c2e8a4a44570a0725f4e97a17d69 : SUCCESS in 15m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/723282ab14ac461ea4bfe624a615b240 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":2},{"id":"9d320635b28b2fba93910bf97529763322f82ec3","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-08-16 14:28:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a15fb6005b600f290b927c298f42f02d2a09d33c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 14:29:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"83875155ef07a1c7f2360a6d30b1d98610424a18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 16:07:50.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a9f3be45dbd41b59cbe2a83fcc32285\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/783d919e81dc4479aca5bbfe2781d16e : SUCCESS in 2m 47s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/070a5e6c623c48ffb5d929e32308952f : SUCCESS in 2m 26s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/51e13e8077af487d90bbfa9df2ef73fd : SUCCESS in 14m 26s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/37a3b99dbc5640ff8cec3eccbae75679 : SUCCESS in 11m 18s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/7e64eafa3cba4387aacaaa843344df8a : SUCCESS in 14m 47s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/99af58b4de8746d19df6ac13a4d51877 : FAILURE in 1h 05m 30s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cde8e71436dd481099273529f12a04f7 : SUCCESS in 1h 24m 04s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/32a8cc5ed89e48d990ff87ba0eb3b2ce : SUCCESS in 1h 13m 25s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/392417c50c4646ecb4fb1ea12c47b1c9 : SUCCESS in 1h 11m 31s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a58d14e568146c89d6a5da6cb999a03 : FAILURE in 1h 06m 27s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/1aa1efec4caa431181e92c8a35690c1d : SUCCESS in 20m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5ffa09db756447297cd1c3b940ee430 : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":2},{"id":"810c5cf92dfd272e1ae631dd412ef10fe1bd4586","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-17 00:57:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8e99636590e7d01ab5eb61d3542889b085c39acd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 02:19:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91fb1a7b24c1495ab56df3d97661b74b\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/113e1b47f80744a897787f1ebd70c62a : SUCCESS in 2m 47s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/af63ea57ba084d2fb96804ef9837a489 : SUCCESS in 2m 32s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/d550bd250db54a3fb41fe01d0d8e099e : SUCCESS in 5m 00s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/17c314ceccb2450b9a117cc1a45c943f : SUCCESS in 13m 37s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad92a15882ba448b837de14b602797e3 : SUCCESS in 12m 02s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/a0f905ef37694866ad16ee0cf5df8036 : SUCCESS in 13m 44s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7d6f512ecf7492695c29b1a4cd93212 : SUCCESS in 11m 32s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/faa666e0f71e4a639f66bd377cea4c39 : SUCCESS in 41m 57s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f20b26594abd4c46bef2642680e4f6ad : SUCCESS in 57m 58s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f7c7686959745f7b7b1a30092c55a71 : SUCCESS in 50m 43s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0b31951aa724aa39aab00fcd63cc78e : SUCCESS in 39m 18s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c45f8951fee24de6afe58f8c0cce355f : SUCCESS in 39m 11s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/db2b71def877444f92128fd5113a9eae : SUCCESS in 1h 07m 26s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5276fd13d96d4314aab19d933a327d1d : SUCCESS in 1h 16m 40s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a2ca8925d8664d708c2df69c8dc4ac1f : SUCCESS in 1h 00m 14s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c1987e13adfb4a2d86f293fa463e8167 : SUCCESS in 55m 10s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3a0549e2ddc44c21a8edef8a4e239393 : SUCCESS in 57m 52s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/9cb1b7ceb23f4340b9666839607cb449 : SUCCESS in 20m 06s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97ecc98d56b74562838bdb53fd6ed19b : SUCCESS in 16m 27s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/560327ff6d45453bb8b655b997acc34a : SUCCESS in 16m 08s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa3ca9335e234799b4e8abdfa37f62e5 : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":2},{"id":"35c45c7b15991f3e26c5af7803be93ed4cdbe667","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 02:20:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"99f5d4fabdc77fda03b5a857ab9520e4fed44f3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 04:27:51.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/22659dfbb0944dc59241d19b54150d9e\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55c8f5721d3a487c94f623a79b8f71ea : SUCCESS in 2m 49s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8444423694cc4a22b473247a084f83e9 : SUCCESS in 3m 14s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/7f20f773ced145dfb989fa2230390067 : SUCCESS in 15m 43s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/83463a4a4f7e480587fa7ed76bece87e : SUCCESS in 11m 06s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/f01cf14394b041a99babfc89b9e8af11 : SUCCESS in 17m 54s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d5f2969ec47747b282d37b92004cda3d : SUCCESS in 1h 00m 42s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/56b6b0a2374945db887110d744aa3b8f : SUCCESS in 1h 54m 21s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f63e4d98a0094a918a2aeebb812b70b8 : SUCCESS in 1h 40m 14s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/97dfeb2310d44d4f96f34a5d3c3642a8 : FAILURE in 1h 00m 00s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/02e58f86dade40239ae19b910d864e87 : SUCCESS in 57m 10s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/f9e5145a3b7f48eab97bdb2c31bdbc1e : SUCCESS in 22m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d96bb5eba17a41e0add344a6f7f69d29 : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":2},{"id":"ff6c3db6a31f3a77f0025c36c3457603a01da183","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-17 04:46:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"98aa3564eb9c79ab49ced8a70dfd8b448fccf974","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 06:15:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35ec362216694380819c56bf63e840d8\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/45126eeadf514726a84694c89051e53c : SUCCESS in 2m 28s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca229ca6c79d4ffea62da1cf22b3ebaf : SUCCESS in 3m 26s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/08bc2929325e4116b9f8e30f434854d4 : SUCCESS in 4m 30s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/7a54592e3f1a466cbb1be6149d5e7ac1 : SUCCESS in 14m 15s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/49184dbcb8da441cb8eb25eae0ac969c : SUCCESS in 9m 00s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/5180dc3218e34e94800e62707d20c69f : SUCCESS in 16m 27s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c637adbb8f34ce8b649c3815f90e13d : SUCCESS in 12m 37s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a34b6a4bdd64ef69caf88b6c68dadf9 : SUCCESS in 40m 11s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82931f3ae54b4708b677541937dc1195 : SUCCESS in 54m 45s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e9c7d1f99ab4af985f417ba258a7b93 : SUCCESS in 51m 39s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c11df426d448486d925f1940d6ebf876 : SUCCESS in 38m 55s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aeeff99c8569466bad34ab32ac8f61d5 : SUCCESS in 39m 25s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ce51ac06c2cf409dbd96b25b0e105016 : SUCCESS in 47m 46s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7023fde5a2a64b55a42ca96035729914 : SUCCESS in 1h 20m 44s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5e0eb836cfe74bbaa95cd2e528b8ac6c : SUCCESS in 1h 09m 50s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1698855cc57941b2b970af0927d56c01 : SUCCESS in 59m 14s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3415235044954031a24bd9b053f7a282 : SUCCESS in 45m 19s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/35e599c2bbae421fb6826ac654069548 : SUCCESS in 20m 48s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f3f0a549f7b4ae4983f50303fa388bd : SUCCESS in 12m 40s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/31d32a2519a645a8a51de83ba044bace : SUCCESS in 16m 22s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aed7730cfdcf418798e729d552aa18d2 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":2},{"id":"a4ced27df076d969e14e43b3016254eb9f0529c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 06:16:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"52afb6347154b69dab4f5346c583c1f7d849a253","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 07:55:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85ed1e82b51d4ebfabccb6e45466d6ae\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5002dd6f68b1400e987081924e392aef : SUCCESS in 3m 10s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e7e92139d084205834d65a4a7e55c60 : SUCCESS in 3m 05s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/c2125fb64b744cee9d55a7af47b5b152 : SUCCESS in 16m 13s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b4913aa28d94fefb13bf8d385e0359e : SUCCESS in 11m 21s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/f85950dd621e4578917309db5f7c4113 : SUCCESS in 12m 47s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bce76585eb3b48f8a805552477494314 : SUCCESS in 59m 21s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6a4fc7d8b84c44c687c82f68020921df : SUCCESS in 1h 32m 33s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7e9cce60cfcd4ec2b23aad1b7eb2c41b : SUCCESS in 1h 03m 31s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/54d24b62efcc42e1a2b1b3d2ff9405b3 : SUCCESS in 51m 57s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c8656a4cb14b486e81e24e3981f61b0e : SUCCESS in 49m 55s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/c24b54b6b2cb4a259804215f8b507c17 : SUCCESS in 17m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f805417914a4e4da5e7824915d7831d : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":2},{"id":"a91f7455aeabfc6858644a23691ae7c8f6ea2f0f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 07:55:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d51ba84fc48e8875a6d2fae7c39b756579ae748c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 07:56:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c340ffaf3fd412c88648581dc04bf5d\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0fc786d103764c7cacc1bf44e883cf38 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"293786f2a1ffd2d78c7fc263ffc44061ff135115","revisions":{"fd3252912c1604ff535c13eebf0c665f918069bd":{"kind":"REWORK","_number":1,"created":"2023-08-16 12:26:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/56/891456/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/56/891456/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/56/891456/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/56/891456/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/56/891456/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/56/891456/1"}}},"commit":{"parents":[{"commit":"ee4d51197f175d03a4edd2c9af2188f5df2d8aa2","subject":"apache: Expose setenv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/ee4d51197f175d03a4edd2c9af2188f5df2d8aa2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-16 12:26:37.000000000","tz":0},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-16 12:26:37.000000000","tz":0},"subject":"Revert \"apache: Expose setenv\"","message":"Revert \"apache: Expose setenv\"\n\nThis reverts commit ee4d51197f175d03a4edd2c9af2188f5df2d8aa2.\n\nReason for revert:\nIt later turned out that setenv does not override os.environ in wsgi\napplications and this option can\u0027t be used to customize the environment\nvariable as intended. Let\u0027s revert the change to keep the interface\nsimple as much as possible instead of keeping the useless option.\n\nChange-Id: Ic057a86a67d819e49ff2965bc3387fd63a8e76a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/fd3252912c1604ff535c13eebf0c665f918069bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/fd3252912c1604ff535c13eebf0c665f918069bd"}]},"branch":"refs/heads/master"},"293786f2a1ffd2d78c7fc263ffc44061ff135115":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-08-16 12:27:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/56/891456/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/56/891456/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/56/891456/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/56/891456/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/56/891456/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/56/891456/2"}}},"commit":{"parents":[{"commit":"ee4d51197f175d03a4edd2c9af2188f5df2d8aa2","subject":"apache: Expose setenv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/ee4d51197f175d03a4edd2c9af2188f5df2d8aa2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-16 12:26:37.000000000","tz":0},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-16 12:27:06.000000000","tz":0},"subject":"Revert \"apache: Expose setenv\"","message":"Revert \"apache: Expose setenv\"\n\nThis reverts commit ee4d51197f175d03a4edd2c9af2188f5df2d8aa2.\n\nReason for revert:\nIt later turned out that setenv does not override os.environ in wsgi\napplications and this option can\u0027t be used to customize the environment\nvariable as intended. Let\u0027s revert the change to keep the interface\nas much simple as possible instead of keeping the useless option.\n\nChange-Id: Ic057a86a67d819e49ff2965bc3387fd63a8e76a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/293786f2a1ffd2d78c7fc263ffc44061ff135115"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/293786f2a1ffd2d78c7fc263ffc44061ff135115"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
