)]}'
{"id":"openstack%2Fpuppet-neutron~893410","triplet_id":"openstack%2Fpuppet-neutron~master~Ie2d9dbebb11a3e6faa2a01b72216847f2db14bd6","project":"openstack/puppet-neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-09-05 16:01:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie2d9dbebb11a3e6faa2a01b72216847f2db14bd6","subject":"Avoid using false for non-boolean parameter","status":"MERGED","created":"2023-08-31 15:49:38.000000000","updated":"2023-09-05 16:01:26.000000000","submitted":"2023-09-05 16:01:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893410","meta_rev_id":"3e5c809fce5e801ea1305d166ee93ec97c7a0194","_number":893410,"virtual_id_number":893410,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-05 16:01:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2023-09-01 06:12:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2023-09-05 14:42:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2023-09-05 14:42:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-31 17:06:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-01 06:12:36.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-09-05 14:42:48.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"d0a08d40bffbf2be0635aa14d80fd4ad3890303c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-31 15:49:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8eca2693e603c094908a2835a1752d84f9f3a87d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 17:06:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4393d6a44374779bf28032d72940e6e\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ec47d6af91747c1a75ab032ffd4279c : SUCCESS in 2m 56s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e9860a7ca004e3b98449862a2b4b879 : SUCCESS in 3m 36s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/cbb7718c99894e0b90744f606f66d918 : SUCCESS in 6m 00s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/2a4b4c2936d64917896bb11cc8bffbc5 : SUCCESS in 18m 21s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc789aff6b064f59a8878aa49dffb38c : SUCCESS in 16m 00s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/c131bfdcffcd43649e170c2f747d94ea : SUCCESS in 15m 45s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6da109cb42784742910039bd09a12721 : SUCCESS in 13m 43s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87a68b076cdb4f80b8ea596068f41b08 : SUCCESS in 47m 54s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb004dff2b824ef798ef30b0bc8a5970 : SUCCESS in 53m 55s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4624c96bd7b0472a96e91f78d9503e86 : SUCCESS in 43m 11s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d8159b1984994ae996a2c4edf9f03280 : SUCCESS in 46m 28s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/483f9047a4b44c50b6411bb3a587e75f : SUCCESS in 42m 25s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a68de56fec2c40c08dbe5f19620793e6 : SUCCESS in 1h 09m 25s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/586d3aa8e22a47f797754e8ce0439ce0 : SUCCESS in 1h 05m 01s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6d795f35f67542778546e56ec79d5582 : SUCCESS in 1h 09m 49s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a7387032362441078abfe95062bb1f98 : SUCCESS in 54m 10s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f5c40aea40d9406885062a467560aa4f : SUCCESS in 1h 09m 46s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/6265bbdb3a034342b621d4f40d57ffba : SUCCESS in 22m 56s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b89a59f9fd941b39b3a221a8a1e0519 : SUCCESS in 15m 28s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99d95448f3b5413883fb9df5118510a4 : SUCCESS in 14m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"213309cd228095579eeb36329856e8a17e9a46c2","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-09-01 06:12:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e56b39374b8e0ce91e26ec5538bad2b580c90ceb","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-09-05 14:42:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7f9828052b7b775f740d2cb898d3c33286010896","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 14:44:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"02df04cd2d96fdb89f094dafe0519c6705557f3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 16:01:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c13e5d66b87640d5ac0ce8b0b9a5151f\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d031ba2d6764433aef1bfc39c0aa6c9 : SUCCESS in 2m 49s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b810f8148f245679bd265716d201bad : SUCCESS in 2m 44s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/7ae38082c97d4ee7a8793003ba5198de : SUCCESS in 16m 54s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/58601831818f458793e455bfe32d6f2d : SUCCESS in 10m 15s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/a07b944ba6204b768e9096ef02682b79 : SUCCESS in 16m 38s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ca4a25efeab64e2592e855b6988e964f : SUCCESS in 59m 38s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b7c2d21c537641e3b45fa918930aa5ee : SUCCESS in 1h 13m 12s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fe296cfdf1ff41d4a2e0a5a0a8c0060f : SUCCESS in 1h 02m 46s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/206906c4074243609b2d64a78f5026d6 : SUCCESS in 50m 24s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e2502425cf5b4e189879d98076f2188c : SUCCESS in 52m 30s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/d7c884e860d1438b8e5f63df1b0d8a36 : SUCCESS in 19m 56s","accounts_in_message":[],"_revision_number":1},{"id":"3e5c809fce5e801ea1305d166ee93ec97c7a0194","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 16:01:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"10c3000fd46565d2c01386ae04895916682b9abd","revisions":{"10c3000fd46565d2c01386ae04895916682b9abd":{"kind":"REWORK","_number":1,"created":"2023-08-31 15:49:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/10/893410/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/10/893410/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/10/893410/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/10/893410/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/10/893410/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/10/893410/1"}}},"commit":{"parents":[{"commit":"162a1bf84b594c97298b446f3122a36113274f88","subject":"Merge \"Ensure openstackclient is installed at the right timing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/162a1bf84b594c97298b446f3122a36113274f88"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-31 15:48:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-31 15:48:46.000000000","tz":540},"subject":"Avoid using false for non-boolean parameter","message":"Avoid using false for non-boolean parameter\n\nThe local_ip parameter is not a boolean parameter. We should use undef\nfor its default value instead so that we can implement a better\nvalidation using Optional in the feature.\n\nChange-Id: Ie2d9dbebb11a3e6faa2a01b72216847f2db14bd6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/10c3000fd46565d2c01386ae04895916682b9abd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/10c3000fd46565d2c01386ae04895916682b9abd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
