)]}'
{"id":"openstack%2Fpuppet-neutron~898381","triplet_id":"openstack%2Fpuppet-neutron~master~I3acb3fe9e84f38933ad92536bc4706761966197f","project":"openstack/puppet-neutron","branch":"master","topic":"reno-2023.2","hashtags":[],"change_id":"I3acb3fe9e84f38933ad92536bc4706761966197f","subject":"Update master for stable/2023.2","status":"MERGED","created":"2023-10-16 15:18:11.000000000","updated":"2023-10-16 18:26:32.000000000","submitted":"2023-10-16 18:25:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898381-reno-2023.2","meta_rev_id":"50a28fa8c6b2caa0c6d7083416cad6b89072f006","_number":898381,"virtual_id_number":898381,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-16 18:25:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2023-10-16 18:19:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2023-10-16 18:19:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-16 15:33:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-16 18:19:58.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"e4f1f8cee44f7b3c583790b778a8f0e134311f81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2023-10-16 15:18:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"51bffbed3a7cffae682b55640dfed4be7b3f4c59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 15:33:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dca70b270cff407eb9e0ce93e13626f9\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a4a370287664874b8944d05dacda560 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":1},{"id":"a2e77a73b9bf86eaf5f31a2cab23a8772afa753b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-16 18:19:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"098ff65ffa932042bc23a0480fe3c9835abe00dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 18:20:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7cab8a74ebe8a861cd3cbf99fd9135b0d42af6e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 18:25:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/685949fe128d43dca31d59b5f0fd365c\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a05885cbaf3847c0920eb8eff93bb2f0 : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":1},{"id":"e60b818c46dd8f44260b68525642eb629d9c4e37","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 18:25:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"50a28fa8c6b2caa0c6d7083416cad6b89072f006","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 18:26:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4660ed080ba4af7b33c64ccd0cdc575\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8808cb67bde4a478a6587aba5b451dd : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c1736c9c215d56a8566fb15fc4b9dc2a617d1054","revisions":{"c1736c9c215d56a8566fb15fc4b9dc2a617d1054":{"kind":"REWORK","_number":1,"created":"2023-10-16 15:18:11.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/81/898381/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/81/898381/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/81/898381/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/81/898381/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/81/898381/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/81/898381/1"}}},"commit":{"parents":[{"commit":"2ef559bd3b8f2e0dc03a0e9f26f4e1d0511a2363","subject":"Remove usage of deleted manifest_dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/2ef559bd3b8f2e0dc03a0e9f26f4e1d0511a2363"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-10-16 15:18:10.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-10-16 15:18:10.000000000","tz":0},"subject":"Update master for stable/2023.2","message":"Update master for stable/2023.2\n\nAdd file to the reno documentation build to show release notes for\nstable/2023.2.\n\nUse pbr instruction to increment the minor version number\nautomatically so that master versions are higher than the versions on\nstable/2023.2.\n\nSem-Ver: feature\nChange-Id: I3acb3fe9e84f38933ad92536bc4706761966197f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c1736c9c215d56a8566fb15fc4b9dc2a617d1054"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c1736c9c215d56a8566fb15fc4b9dc2a617d1054"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
