)]}'
{"id":"openstack%2Fpuppet-neutron~899007","triplet_id":"openstack%2Fpuppet-neutron~master~I4ffbcecfde80791bde946fe6add006d1fd75c290","project":"openstack/puppet-neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-10-26 17:41:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4ffbcecfde80791bde946fe6add006d1fd75c290","subject":"Remove support for netwokring-odl","status":"MERGED","created":"2023-10-23 02:01:55.000000000","updated":"2023-10-26 17:42:47.000000000","submitted":"2023-10-26 17:41:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"899007","meta_rev_id":"dac2c215eaa588bf929b936a6cef8c8184ba245e","_number":899007,"virtual_id_number":899007,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-26 17:41:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2023-10-23 07:45:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2023-10-26 16:10:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-23 09:35:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2023-10-26 16:10:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-23 03:32:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-23 07:45:48.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-10-23 09:35:33.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2023-10-26 16:10:04.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"aa58befd0e1bdbc9642273bbc88c17f16d6e1254","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-23 02:01:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8ce61e6d34d6043652f4bc3cd52f0f1d9d1ec56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 03:32:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27137065bc9f4471990861caf42bd5ad\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/552f0552433b42febd25505d3f6b6091 : SUCCESS in 2m 57s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/079e74452818476aa277388a21283478 : SUCCESS in 3m 18s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/d4ecd006bcc44e5f8af8b7620930ce9c : SUCCESS in 7m 01s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/f538e1fd5949474294ddeb5573f89f5d : SUCCESS in 14m 50s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/840e97b321be4fff9d262bc1e937d603 : SUCCESS in 13m 42s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/51b33041c949443690f455efd2b14782 : SUCCESS in 13m 49s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f4d0bc6c2a84384a76ddb1510f1f15f : SUCCESS in 14m 02s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0bf7e27c6c6d458d9fc0cd1826455b78 : SUCCESS in 48m 36s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/15f04d9babf34f3a9d9c009c6e0c6a13 : SUCCESS in 53m 03s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/56a0585060374808a211649f21d23844 : SUCCESS in 47m 47s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d6976861d3e9404a8e2d35bc1852255f : SUCCESS in 42m 41s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b39a86fc00e40c8b02fcabd48b8a248 : SUCCESS in 44m 03s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/be2ae4a40e06447bb5b4e9d8ae482218 : SUCCESS in 1h 02m 20s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9873484f7b1242fcba8943c49b9fa655 : SUCCESS in 1h 12m 11s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/deece72f956742118933ce5fc67a208d : SUCCESS in 1h 03m 03s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a0ff29d7fc374253b17d3991bc1186cf : SUCCESS in 55m 59s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1c29a9050a864e909c86d90caa42f833 : SUCCESS in 50m 52s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd60cb56a1f648e59afcc38eeb47a9ba : SUCCESS in 20m 33s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e427a2d8d642470d9824b54c9abe6900 : SUCCESS in 16m 07s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e68f494fce0427ba736821ea74f855b : SUCCESS in 15m 31s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2cf2ef0d3e934cf38a5802fd3b84c713 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":1},{"id":"3d86810b469565816c2149677133d3cb74388bbb","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-10-23 07:45:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"554e6b38d63c568dec87a73d075a96f79091a64c","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-10-23 09:35:33.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"163db6050bdda8036b2745a95eadc28d7deac3c5","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-10-26 16:10:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8abcead76029832fb5eef3edc5e48a12fdf3c60c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 16:11:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9d4173dd5d3c819056571d9fea7352a6a45392eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 17:41:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbe8539cb1a145bf80fc681a7880b678\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa0e29f75c6149cca35f2d3b77dca326 : SUCCESS in 3m 04s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c43d772a28444a9392ae53f1d7d8612f : SUCCESS in 3m 17s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/c33ec9a8e2da4252988deeba39187a1b : SUCCESS in 11m 56s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1109cd731cb6449982dc46633fb68673 : SUCCESS in 12m 21s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/365be6406d6f497581a5aedc23440696 : SUCCESS in 18m 08s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/315a3516afd7496d866bf9a3751eab46 : SUCCESS in 1h 04m 43s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4faa07e886234620a8a3768f9d932225 : SUCCESS in 1h 26m 03s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/eae1daa9499d44cc95e2af06b47bcd83 : SUCCESS in 1h 13m 07s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2e5e50f1af4e4cbc9d330a9a367487ef : SUCCESS in 1h 02m 24s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ae77d401df1f4600b6b6b7b627d14298 : SUCCESS in 58m 12s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/eae39b6c85f14ff7958c9adbd62a316d : SUCCESS in 20m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7b94bf6595447068bb4d6feae2b6c52 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":1},{"id":"698ea688f214028b4603108e9bc13f9a86f332d4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 17:41:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dac2c215eaa588bf929b936a6cef8c8184ba245e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 17:42:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/567759b88b90412bb38f89bffb9cf80e\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8a5613ca3ce4dabb6dfc4ac8f2fc412 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6c8337263110bf1bf97f5aff18f1d0ee91afac4a","revisions":{"6c8337263110bf1bf97f5aff18f1d0ee91afac4a":{"kind":"REWORK","_number":1,"created":"2023-10-23 02:01:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/07/899007/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/07/899007/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/07/899007/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/07/899007/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/07/899007/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/07/899007/1"}}},"commit":{"parents":[{"commit":"7288da601da475f77b49a41ab3beb77ad4be6332","subject":"ovs: Accept hash for bridge_mapping","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/7288da601da475f77b49a41ab3beb77ad4be6332"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-23 02:00:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-23 02:00:59.000000000","tz":540},"subject":"Remove support for netwokring-odl","message":"Remove support for netwokring-odl\n\nThe networking-odl project was deprecated during 2023.2 cycle.\n\nChange-Id: I4ffbcecfde80791bde946fe6add006d1fd75c290\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6c8337263110bf1bf97f5aff18f1d0ee91afac4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6c8337263110bf1bf97f5aff18f1d0ee91afac4a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
