)]}'
{"id":"openstack%2Fpuppet-neutron~899011","triplet_id":"openstack%2Fpuppet-neutron~master~Ifba4c6d2061aab72294d4cee46fadfae0bf30dc8","project":"openstack/puppet-neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-10-26 23:28:21.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ifba4c6d2061aab72294d4cee46fadfae0bf30dc8","subject":"Deprecate support for networking-mlnx again","status":"MERGED","created":"2023-10-23 02:42:38.000000000","updated":"2023-10-27 02:43:27.000000000","submitted":"2023-10-27 02:41:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"899011","meta_rev_id":"d18e815d753d168bb2635cbc49c47f91211cbcfb","_number":899011,"virtual_id_number":899011,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-27 02:41:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2023-10-23 07:47:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2023-10-26 16:10:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-23 09:34:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2023-10-26 16:10:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-23 04:17:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-23 07:47:10.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-10-23 09:34:02.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2023-10-26 16:10:18.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"85535eb72f9007292fbb1e3a0cdeeec5f7e0a566","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-23 02:42:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"af9e8747a0ba8ce7e81bd2bfda6319e0b41402fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 04:17:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c9ed05a881540218d34e67a17ca4f1a\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b53541ce0e254921acfb54556cc57c47 : SUCCESS in 3m 21s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/25eb98ef986441b5a725324d30c44492 : SUCCESS in 3m 36s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/292695703ee0475e930b4dbb7d2e172e : SUCCESS in 6m 57s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/c0daaed06e534507a8b56970ad4328a1 : SUCCESS in 16m 28s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2bd142cddde5429988101affc438c3ac : SUCCESS in 13m 05s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/da05d616927949f4b0929e557b113ed2 : SUCCESS in 15m 33s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6830482014ac49f597bef688751789a8 : SUCCESS in 14m 14s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0dac1d01522a40f8b96e91b0d8552638 : SUCCESS in 47m 42s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0bd4eedcd7a4c7ba467c976dab51853 : SUCCESS in 48m 30s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1394963677a040e1be62806d27cd9c55 : SUCCESS in 50m 22s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f43eee1949f0441b8856aea9f1e17b69 : SUCCESS in 45m 49s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4fdb8618ac9a43b69694697d83acf341 : SUCCESS in 47m 57s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e2d257e2a5964358b307931744cb4dec : SUCCESS in 1h 03m 54s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/91c314728786499eb19d2199039f24af : SUCCESS in 1h 09m 26s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9265986bc5164c38a6f665ef1dcdd387 : SUCCESS in 1h 01m 09s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7a704f65fa47468bab802b8610c3f719 : SUCCESS in 54m 12s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1340d54819e047df83a3ad469248da80 : SUCCESS in 56m 54s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/826d851848c2440e88a81355e16583a4 : SUCCESS in 22m 09s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/58b5dd864e024027bdee3cbe5b4ecbea : SUCCESS in 14m 46s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/535f4b92ae864dfd8996027527ca0e87 : SUCCESS in 15m 25s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0aa0bc28337b49909bc33b86de393248 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":1},{"id":"fa11b920859d7c1bc3895cd2211ba0b511d8f471","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-10-23 07:47:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"410f817cfe3254ce16c26e1078f212558a52aa06","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-10-23 09:34:02.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7f46b2ca58dd21bcbccc4985ec8e01222163fed4","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-10-26 16:10:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f4d690720804d9618bf60a0db0c85e335e211194","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 16:11:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"35c0e0734d42877f0e475f39dbc89961f29a3f88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 17:41:47.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2009a1499df24720a8c541cbbef2cb9a\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f0c7fae1b1024a9f8e2c4664354805bf : SUCCESS in 2m 58s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52252f3610844323940e36be6104a987 : SUCCESS in 2m 57s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/82fff96ed7954ffab98dcf168b791076 : SUCCESS in 17m 05s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d62be275b84b48d2b8d010199b5ca8a8 : SUCCESS in 13m 36s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/d929764cf4444f23b1a64b209823d4b6 : SUCCESS in 16m 59s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9603a4dd2f5648c099d11879704cdbe3 : SUCCESS in 1h 09m 41s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/64398eba155d4e13be006c0f9ab5eafe : SUCCESS in 1h 22m 06s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3f68cb354b7b4bb6a0e62399ff78e3dd : POST_FAILURE in 1h 08m 17s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6032249abaf34dcc97b98edd1897fb4e : SUCCESS in 55m 57s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3cf18e635867488f8c30520b5cff1996 : SUCCESS in 59m 26s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/36b05df7da6a45518d6f883f4084f9b4 : SUCCESS in 21m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c384d49514284b6e8dc9ce345e789312 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":1},{"id":"731e47fa3b89e75a3fd2438082f1bf215154067a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-26 23:28:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9f5bea4fe4d8eee2cadd428357a97ce87c5df69a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 01:04:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c32d7de9f4a479796fdb4fcd7a5be22\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/354a0a56cc1047cf9b0cda8762d98019 : SUCCESS in 2m 27s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bffa98e094f749b089725000ddedaf87 : SUCCESS in 2m 25s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/77988e7b503944d9b6eefda881a4cab1 : SUCCESS in 5m 32s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/970ce433b2e84afab3614ca76a19775d : SUCCESS in 11m 37s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67735b39ca2f4a02be5e93b61efd04da : SUCCESS in 11m 19s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/9987d728fe87460cb367a070bad0b90a : SUCCESS in 11m 52s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be060272af9049b8bf3ff0299f2f1185 : SUCCESS in 11m 16s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ed756abba0c042a8a4b297458cfd0e42 : SUCCESS in 37m 59s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/12fe8f75a396454e8abcca0bcaa09daa : SUCCESS in 1h 02m 57s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e77e33437ddf4b68b99178a2abd27bf2 : SUCCESS in 38m 59s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39e62e8784d746d884d62c60f20cf8b8 : SUCCESS in 43m 33s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f76dd7ab79ba46748e5fdcd0099eb97d : SUCCESS in 41m 33s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7a210b2e67214ebcafda08966ed8bce6 : SUCCESS in 48m 03s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a8e9989790024f969fe4cce67b81b428 : SUCCESS in 1h 28m 20s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ade11eee28c41e5a62fff6865f385bd : SUCCESS in 1h 03m 53s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/01a1b3b1ab934d84ba493fa781b183e6 : SUCCESS in 43m 44s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0b9b730b54e247c38b3a28133e34baea : SUCCESS in 55m 28s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/dcd7bea85bc740af927eed8fbbb6c6f0 : SUCCESS in 16m 10s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac5fc25dc6dc43669e374f9095df9e79 : SUCCESS in 16m 08s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/527323b0e76748cbb736ea7ae653ce81 : SUCCESS in 11m 59s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae872dbf9dc243a0b6316a116666c61d : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":1},{"id":"f443503745ec9d3bd11a41fdd87ba7e2d8f0f428","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 01:04:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a8450cb89293d463f34f1999b950d6981680b911","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 02:41:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a94f952a2cc54fd7b6826d11d526b7c9\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc06636b86a04454b43c09c638a10f57 : SUCCESS in 2m 36s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd0360349ec54f8593e2a0c64ed9d420 : SUCCESS in 2m 37s\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/7b0b565799b74b6384da4d0116b67954 : SUCCESS in 14m 51s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6db7d767052d4955bb390fff0b8b8779 : SUCCESS in 11m 53s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/e016dd16865745bb941c593cb147a6cf : SUCCESS in 15m 04s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b4860baee0a24e36ab76a74ae4babd43 : SUCCESS in 51m 53s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fd6dc7971ce24581973341886aee83cf : SUCCESS in 1h 30m 07s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/44b6b884dc844f6f9376cd2f0699d19c : SUCCESS in 1h 14m 56s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ccb28e70f2b4148a02f9c66023fea1e : SUCCESS in 1h 02m 42s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/07c482560a7f405e89eb2e09a7c82f77 : SUCCESS in 52m 08s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/1daf04d154944e00a013b0874b4a5e8c : SUCCESS in 23m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce9c7ef13bf849d99908403dcf1c34e4 : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":1},{"id":"4f4b8e1462eb4247a317dee606b55637fa897d9c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 02:41:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d18e815d753d168bb2635cbc49c47f91211cbcfb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 02:43:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f1c0888af8947c29651c459e8f199f5\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7964b29b8a234d638f72c46dd53e7bbb : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"98ea1311fbedab83c054a3498a59803f50a48ff9","revisions":{"98ea1311fbedab83c054a3498a59803f50a48ff9":{"kind":"REWORK","_number":1,"created":"2023-10-23 02:42:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/899011/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/11/899011/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/11/899011/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/11/899011/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/11/899011/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/11/899011/1"}}},"commit":{"parents":[{"commit":"5078eb55e13f4942e76de085da95fe5e96eaf299","subject":"Merge \"ovs: Accept hash for bridge_mapping\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/5078eb55e13f4942e76de085da95fe5e96eaf299"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-23 02:40:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-23 02:40:13.000000000","tz":540},"subject":"Deprecate support for networking-mlnx again","message":"Deprecate support for networking-mlnx again\n\nThe networking-mlnx project[1] looks unmaintained again. The project\ndoes not provide releases since 2023.1.\n\n[1] https://opendev.org/x/networking-mlnx\n\nChange-Id: Ifba4c6d2061aab72294d4cee46fadfae0bf30dc8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/98ea1311fbedab83c054a3498a59803f50a48ff9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/98ea1311fbedab83c054a3498a59803f50a48ff9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
