)]}'
{"id":"openstack%2Fpuppet-neutron~900201","triplet_id":"openstack%2Fpuppet-neutron~master~If2bfbd71d0aa6ed5742516422e9fa1dc06b9304a","project":"openstack/puppet-neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-11-07 21:02:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If2bfbd71d0aa6ed5742516422e9fa1dc06b9304a","subject":"Remove deprecated neutron::enabled","status":"MERGED","created":"2023-11-06 14:50:34.000000000","updated":"2023-11-07 21:03:31.000000000","submitted":"2023-11-07 21:02:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900201","meta_rev_id":"918e734811543749f4b4f2cb8c0233a279f864f5","_number":900201,"virtual_id_number":900201,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-07 21:02:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2023-11-06 16:03:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2023-11-07 19:32:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2023-11-07 19:32:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-06 15:59:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-06 16:03:47.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-11-07 19:32:41.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"c1c9bd22b7ead3e31e241015073f7e97145daa76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-06 14:50:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ea5eb9b2d3bacf897f7667bcebc0f20efc2607f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 15:59:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6f29f839a8d46c3a32b167342914e9c\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f5376a0eefb7431d88ed9bd519b16331 : SUCCESS in 2m 25s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd0d34d7f9844ed1880aa2c644d078ca : SUCCESS in 2m 23s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/3b91afd38e7f422b91a78e22161bf6d4 : SUCCESS in 4m 57s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/57588569d2b547b79990d6c65c045d90 : SUCCESS in 13m 17s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3aa9acca3ca417a85f7b55ecfe00d6e : SUCCESS in 11m 05s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/23325d0c382c4769adbfd0b5d887756a : SUCCESS in 13m 14s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/908a1811037b47148765876209015939 : SUCCESS in 12m 19s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/51e47664d7fc4525acc9838b761261fe : SUCCESS in 49m 51s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f0403dcfc534693ade8879c0a37b1a5 : SUCCESS in 58m 44s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ba7f9176df44403804a6879f7b360ff : SUCCESS in 37m 51s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b32c5efc73b4b41889eb7e44cd3e1d8 : SUCCESS in 44m 22s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3cdb5d4b1acb4ad482536596dc371f02 : SUCCESS in 35m 25s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/32eedd9720ec4085ab5bf3fc196d64f2 : SUCCESS in 1h 03m 45s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/14cd8c1089e64ef1a0d9d54d3e9183a6 : SUCCESS in 1h 05m 28s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d4fcc01dcb0c4ff794587087f1af9d7e : SUCCESS in 1h 03m 33s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/14fd6bd9eb6b496caec1a4acab762cb3 : SUCCESS in 47m 57s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3f61682c9ef548fea406d36c16bc914d : SUCCESS in 44m 14s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/6577a18e6f8d4cc8b82418eef7d8fb97 : SUCCESS in 19m 33s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f48565b1e834b178eb7dd9dc6d99c19 : SUCCESS in 12m 59s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b1b100e6eab54504a93a56cccf6f8dfe : SUCCESS in 12m 36s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f7ff8d6eabb4e46b9ac7e7743a0a6e4 : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":1},{"id":"2ea3fa937657e9ffd5d97fb08de8abb766a55f62","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-11-06 16:03:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9bda64daec6d6606834fb46cca9149b25ef17c8a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-11-07 19:32:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2f1f403aee09773f6a3456cd300e59263e8a604e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 19:33:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d35aeaab0656ab8a718bf1b8e6402a78d8f9e39f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 21:02:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20d7c9f01b414165814425bd2730b99e\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/89cf1131f535440f896f169883e8dc81 : SUCCESS in 2m 40s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/502e2800fda14698b188062eb7a0120f : SUCCESS in 2m 52s\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/ba9bb10790f245f3874b9c03eab8fb7f : SUCCESS in 16m 50s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04d7fa05a9d0474eb6e872cd0464c175 : SUCCESS in 10m 37s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/62c4fd7a64644e0d9613186716b2b3ba : SUCCESS in 15m 15s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/22a226961e3d4e40865be959f69d2ce3 : SUCCESS in 59m 39s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b5687d4325604d7da0fd8e4603263e92 : SUCCESS in 1h 22m 13s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ba8c814fe1a4114af23db460790e72e : SUCCESS in 1h 04m 23s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/32939c76bc2446fe87703107ec9adaca : SUCCESS in 1h 00m 03s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/14e781e8a63e4a7d9a70ca522f167340 : SUCCESS in 58m 37s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/a869c8855c4740c6adf84ed48c5577e5 : SUCCESS in 22m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26956d6143f14913a73c7caa48413584 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":1},{"id":"2a63b884e38499e114585dcb7b4d84297a5878e3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 21:02:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"918e734811543749f4b4f2cb8c0233a279f864f5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 21:03:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6c5b8cc4a5f4fdc8bf8e8a43f6da89b\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e6103c688a8487fab0b7d33b2135c68 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b1e18bf6ff8734c0cffa58ebcc6b13d8c0c7d47c","revisions":{"b1e18bf6ff8734c0cffa58ebcc6b13d8c0c7d47c":{"kind":"REWORK","_number":1,"created":"2023-11-06 14:50:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/01/900201/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/01/900201/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/01/900201/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/01/900201/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/01/900201/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/01/900201/1"}}},"commit":{"parents":[{"commit":"5078eb55e13f4942e76de085da95fe5e96eaf299","subject":"Merge \"ovs: Accept hash for bridge_mapping\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/5078eb55e13f4942e76de085da95fe5e96eaf299"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-06 14:49:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-06 14:50:23.000000000","tz":540},"subject":"Remove deprecated neutron::enabled","message":"Remove deprecated neutron::enabled\n\nThis parameter was deprecated during 2023.1 cycle[1].\n\n[1] 7a5be0030d31f62c31387a3756c9e7b795d00ac7\n\nChange-Id: If2bfbd71d0aa6ed5742516422e9fa1dc06b9304a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b1e18bf6ff8734c0cffa58ebcc6b13d8c0c7d47c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b1e18bf6ff8734c0cffa58ebcc6b13d8c0c7d47c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
