)]}'
{"id":"openstack%2Fpuppet-neutron~900704","triplet_id":"openstack%2Fpuppet-neutron~master~I778a10d8375b841e3cf9dcb3fb09ec10fe669c79","project":"openstack/puppet-neutron","branch":"master","topic":"vpnaas","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-11-15 17:38:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I778a10d8375b841e3cf9dcb3fb09ec10fe669c79","subject":"Add class to manage VPNaaS service plugin","status":"MERGED","created":"2023-11-11 15:53:01.000000000","updated":"2023-11-15 17:39:26.000000000","submitted":"2023-11-15 17:38:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900704-vpnaas","meta_rev_id":"ce065b38f67c4b3d3e4e31a5eb29848b90664a7d","_number":900704,"virtual_id_number":900704,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-15 17:38:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2023-11-13 09:03:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2023-11-15 15:40:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2023-11-15 15:40:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-11 17:33:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-13 09:03:31.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-11-15 15:40:12.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"b2d54c041fdd060029dbb2b15392b8c8cfe6855d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-11 15:53:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c3999b1a2d15431a4ceee721dc94f78c30313609","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-11 15:57:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"98074bb5dc60ad450f63bc799018ecb876ec73e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-11 16:01:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"14787896b0229f13ef77f4bceeeba68159e9f13a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-11 16:08:50.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"ab6c98ea610ffdd28ce030d2d9d7de4c124e847d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-11 16:10:25.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"eec5c6f61818edf99f533c4503504343aa6ea8e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-11 16:29:02.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"54280196ef8da2666fb22552d4daaca5fe235899","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-11 17:33:27.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/126ff51fc78049f1afed2c2ac9f19b9b\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/11687618a312482fac43d17c38cf715d : SUCCESS in 2m 47s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e9e952a02e84b7aa333dd6ab5d9908a : SUCCESS in 2m 35s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/bab89ee79fcf40d8a722d90702e71075 : SUCCESS in 6m 24s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/0125571324b846e99ec42dac2e85af53 : FAILURE in 13m 38s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f72427e7c704411add1639d2099c463 : FAILURE in 9m 54s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/2db45821ce1f4b98825e8f2c336fd7e9 : FAILURE in 14m 38s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3cf0cd62ee8a4598b3a05c8d45cdcfe1 : FAILURE in 9m 40s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/663b822a957d4edb86c627bb4082ef27 : SUCCESS in 51m 57s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d51a795a01f34a5f9b1d980ed745b3f6 : SUCCESS in 44m 00s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e140d6f766904862a20976de90df21c0 : SUCCESS in 40m 50s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d6288a786c6c4da885a64af6a2cca33d : SUCCESS in 40m 02s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b8b0fc748164a50b62be5ba541d3652 : SUCCESS in 37m 04s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c286542610c040ff9a2ca9428a6a82fb : SUCCESS in 58m 43s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d2ebd35abbc41a4a34787828ad957fa : SUCCESS in 56m 37s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f2b0f7cbc7a8466f8d5e9ab7f908d3e0 : SUCCESS in 51m 27s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0742c1e3e8fb4a4d9cb63c7df0749f4e : SUCCESS in 50m 25s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bb980ac03f034089bd0fce74c5954f38 : SUCCESS in 46m 32s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/618d4b3776c247bab1a8e08aeaf23481 : SUCCESS in 20m 10s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f5a439f4bed428db2c742f088686b04 : SUCCESS in 11m 37s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/96786e7646d541f28cbd454d34564d84 : SUCCESS in 15m 19s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fff8dc2ac52c428bbac4430663ae2d2f : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":6},{"id":"bcbf3d263f90da5681a108febf0c690d9a0a53f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-11 17:39:32.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"31688631ba47de2412d8bc7c0dda2ea2055c2204","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-11 17:41:44.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"4e96f1b85b7e6ba2d3297a372db007e3fc1f16f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-11 18:52:06.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"1dc14d5e0439e0fbd076b3f88b82cfeb38370142","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-11 19:32:16.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"fc248aba7056e10b3fd43c54f52ad552e7ec9815","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-11 20:49:14.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bf9e11ea95274ad2a888c9fef9d61498\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/32578d65185a454b92190eb450507b65 : SUCCESS in 2m 49s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/854adc8df1fd46ae98a3303ae092ba51 : FAILURE in 3m 11s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/e0c6bc4cf7304c58a5402c5078d3f61a : FAILURE in 4m 20s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/9f6f75e0343b4c9d98ae9ca451b1ad50 : FAILURE in 8m 41s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7781cf05fd094965ab5fbca0f9cc6b00 : FAILURE in 5m 53s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/1eb3479570da4bafb5bf676a7bd1c215 : FAILURE in 8m 55s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/19640bdf935e4f75b7fe5bfa4ec5cb0a : FAILURE in 5m 34s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9087f3776cd473eba10141e7e012a80 : SUCCESS in 47m 14s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2663fd32a69c426da5c49e43e6ec04bb : SUCCESS in 56m 53s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e709d140a1444013bd552753571d77b8 : SUCCESS in 37m 02s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e3b8a799a1014c27856179cf143cbc93 : SUCCESS in 43m 55s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29be71e4388a48aebbc5e2080dc11735 : SUCCESS in 34m 29s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/07e6783e5ec14c70adf04d57e2d5134a : SUCCESS in 45m 02s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ae97b7b1082144158e59025eb787cc7d : SUCCESS in 59m 32s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b61a493a19a84e8b981ae8e124e613e9 : SUCCESS in 1h 09m 27s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e180136f4a134b0bb9d71dda3e55046c : SUCCESS in 45m 40s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c1ae51e42a2a475791e930096264299a : SUCCESS in 54m 55s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/97580bf646fa4624bc38c32a50b8490c : SUCCESS in 20m 31s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aebf0dfbccfb492c87aa14b7e067bbaf : SUCCESS in 12m 12s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b558bb9252d41d2a83cf820087dd451 : SUCCESS in 11m 11s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f94202819e942f982f9f00dcdd20ace : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":10},{"id":"6760f89e3815eea21fd62c91ce38a5cc0b6860b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-12 05:19:17.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"32067d13be4eb844ee2d90ea26285a5359ae162e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-12 05:57:20.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"5120e7bc1be1dfd8625b65507167a5c80f209ae3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-12 06:58:46.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/df57842b4de0486587d52a48e5b1f5d9\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52d6a163f2134c0eacca7804dcc6342b : SUCCESS in 2m 46s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8bd4544ad523422386f25d4939a16c8b : SUCCESS in 2m 27s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/85f22196b9d84bdcb77cce94577510d7 : SUCCESS in 6m 25s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/f6144d2747344996b33f66227e9d8058 : FAILURE in 10m 39s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bdceb8700d184537b3dc6ccfc5dc9123 : FAILURE in 8m 33s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/491d54b75b284283b541245a18335e26 : FAILURE in 13m 23s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5356d851a0f440bd8a824348729fb6ca : FAILURE in 8m 15s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2372e4c2b95e4d70849e65a00ecbc0a9 : SUCCESS in 46m 34s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e92176e37694afc84208faac4195aed : SUCCESS in 42m 47s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a4c694c6c784c429b04d1f47c0a7f6a : SUCCESS in 38m 27s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82eb5be805de4c5cbf8242ec7b2458fd : FAILURE in 46m 40s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f64a5b137c845f681eeb76495aa5266 : SUCCESS in 43m 23s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f2e12e3860874dd5bb67096497b385a9 : SUCCESS in 56m 24s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c337419db9e84987a6d800a4ad3e0345 : SUCCESS in 55m 49s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fb6e76e5596847daa4fda6ee646da197 : SUCCESS in 47m 50s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/43114af6695440a8aaaa36adf44f0495 : SUCCESS in 46m 31s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c7a6430da3e4538b81287c12df34ac0 : SUCCESS in 50m 46s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/80e4f868ce814a3a9104ffd7276cb40c : SUCCESS in 20m 38s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eb37db9f91714e3282933d7b6aff13b5 : SUCCESS in 15m 48s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/af9ee2856d6f4cea9b80681782ae9e76 : SUCCESS in 14m 27s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d228f8ba6e04f988ee5b44bcefd9a52 : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":12},{"id":"46ab8c51f2f911d786f01905ecf74b7dbb75c321","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-12 10:38:08.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"9f02309971147f3278b721619ef1c056b1469020","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-12 11:51:20.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8d11f309842e4373b6055dded98633a2\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e187d72fbdda435f8169040c6510079d : SUCCESS in 2m 44s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fdf09ac7b27f488396f3dc714e362709 : SUCCESS in 2m 35s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/f854db2fe65b4c97a3615e15db32a7eb : SUCCESS in 5m 39s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/0d51a4f6db45471fab11a5e1ca4726c3 : FAILURE in 14m 02s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/06e9c47471874beba4a4d399e79de7f7 : FAILURE in 8m 21s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/c884b8f2af5048bf95b77b50816c6642 : FAILURE in 11m 49s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b83c5dc997184482b7e54214ed9a74c3 : FAILURE in 10m 08s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe57bd37f5d74a39b18e9a910ebae8e9 : SUCCESS in 46m 59s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a127d737f874264bcb4bcc608580915 : SUCCESS in 1h 04m 02s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff1b8b9d243c423fb7ce12a0335a656d : SUCCESS in 36m 15s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5908a61844844744b6e852d562434f18 : SUCCESS in 44m 47s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29b2d20ad37c45809d60f7f6ccaeaead : SUCCESS in 34m 12s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/47adcc6c4ad14cc9851ecb21d6bce4e0 : SUCCESS in 1h 00m 07s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/35c2e873afc242abae63125c5377c872 : SUCCESS in 1h 07m 18s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c163ae06dce541ce9c87c78b0e561357 : SUCCESS in 47m 32s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5e4bf478d21b430a9e4c60fe94e7c704 : SUCCESS in 48m 52s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a8f2dc0ba92a45db97327445703e6128 : SUCCESS in 41m 18s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/9988eb1af2ca4077871a65c0ad566469 : SUCCESS in 15m 21s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7431a2ce478f4665b210583eee659e15 : SUCCESS in 13m 45s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7998defae39f42f8b94462cb22a9b2e6 : SUCCESS in 13m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08240673ed0f42b0b229ada8f035904f : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":13},{"id":"b4d20b564a513d8b0de2a40c9c8579214533522f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-12 12:41:55.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"06648e3e72e5001006471feabc9587df9a223ede","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-12 12:49:29.000000000","message":"Uploaded patch set 15: Commit message was updated.","accounts_in_message":[],"_revision_number":15},{"id":"ffeaf4db2b3d7e616dac4daa9dea898e8f41055a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-12 14:00:47.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23277b3fbf7444d891e907f71f9426a7\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7b0dbfcb01f4458ad8ed09da8caa34c : SUCCESS in 2m 44s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b81b9006b9b54693b357813a748e3bab : SUCCESS in 3m 00s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/d72eeb3660ad460691cf00d7141ef87f : SUCCESS in 5m 51s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/bb0d2e94114545ebbdb10e028483752f : SUCCESS in 13m 36s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2923c8bfbb0b4bfe97f4f8c877eaecbb : SUCCESS in 10m 03s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/fd78e386b898477b82e0f4ff539847b0 : SUCCESS in 14m 12s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc23e211330e487283594215c5e5a067 : SUCCESS in 9m 37s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d4bdfc8b4d24b95a4c5dbce5e798c3b : SUCCESS in 36m 04s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/221c3aa86cc24ed293901edf0746ae54 : SUCCESS in 1h 02m 52s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a34233378456487792193fb469ebaae7 : SUCCESS in 45m 42s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5057cc1ed3be49b4a1034117be34111a : SUCCESS in 42m 45s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/85d731d3e2d64727b12fc05fdf986758 : SUCCESS in 33m 15s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/045a92a6cfc544349391076d64314551 : SUCCESS in 44m 13s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1b4f862a2ba440e1bf0c210a252724fe : SUCCESS in 57m 19s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0b9af73173b7441c913b672f54329c84 : SUCCESS in 1h 02m 59s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/afff77648ce745c5a70092c0144866f1 : SUCCESS in 47m 25s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dd09764c08fb40faab54c7d4cc3ed9d5 : SUCCESS in 42m 50s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/b885210c5b824231a876a4179d5882a2 : SUCCESS in 20m 05s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/208a68e7653141a58c12800923d0c230 : SUCCESS in 12m 02s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf47a5b43e844288a1acb0fe5d62b9f5 : SUCCESS in 12m 27s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbcfed5f76454b34b476141adef0e59f : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":15},{"id":"54007036de762bd459786a9347da5dbb94de0b21","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-11-13 09:03:31.000000000","message":"Patch Set 15: Code-Review+2","accounts_in_message":[],"_revision_number":15},{"id":"6ef6d09525f6682d96e5ef7194af27135e84c8e4","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-11-15 15:40:12.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":15},{"id":"e4cb1d5c466deae456de739d356c403cffef0a66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 15:41:11.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"28b65365748bdb0c67cffb35cbaa5023fb9d7463","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 17:38:24.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29a06a74663b4358ba8675381c8175ae\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d3c43e9675d34669baacb02c652b51f1 : SUCCESS in 3m 35s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29b8aa9ff0d5400fa9d7dc0f2271dad2 : SUCCESS in 3m 03s\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/41c24c56686147a39b8c818b748b29b6 : SUCCESS in 15m 59s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2e2246d4387a46e69568c14684cccd0b : SUCCESS in 14m 07s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/caf3e560dc674974a005d01cbb347b7d : SUCCESS in 17m 33s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc5995603c234fa383e2888e7efe38e5 : SUCCESS in 1h 14m 41s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f0609a52b2b34ba9a389afc9baf6a505 : SUCCESS in 1h 20m 27s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fa6f6c6e14754e5c99cd7687807e18ca : SUCCESS in 1h 10m 41s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c42ee46230c5487e96dbe87b9988c8ca : SUCCESS in 1h 01m 53s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6441c086a17e4edca0369bbfac855035 : SUCCESS in 58m 50s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/96fa00b6d70b4a27889801127e7442a2 : SUCCESS in 21m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5dee1725e26a4723b5beab969b6ba694 : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":15},{"id":"50fbfc0d8b38c18d1ab812df2a7ea3a564c9d2e3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 17:38:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":15},{"id":"ce065b38f67c4b3d3e4e31a5eb29848b90664a7d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 17:39:26.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb18c529f3bf468ea3becf0e555410b5\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0617acf381bd4c5e9fc1d4f0eb38123a : SUCCESS in 37s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"ba6d1045a78e6c8a63ce8b1009981eae62656d75","revisions":{"4eb61f0db5297529e435ed18800a090bcf935103":{"kind":"REWORK","_number":1,"created":"2023-11-11 15:53:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/04/900704/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/04/900704/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/1"}}},"commit":{"parents":[{"commit":"c69ceb7faf60b2cae181c263c7b5cb57dcf9ed72","subject":"Merge \"neutron_network: Remove l3_ext network for NSX plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c69ceb7faf60b2cae181c263c7b5cb57dcf9ed72"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-11 15:52:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-11 15:52:25.000000000","tz":540},"subject":"wip: Add class to manage VPNaaS service plugin","message":"wip: Add class to manage VPNaaS service plugin\n\nChange-Id: I778a10d8375b841e3cf9dcb3fb09ec10fe669c79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/4eb61f0db5297529e435ed18800a090bcf935103"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/4eb61f0db5297529e435ed18800a090bcf935103"}]},"branch":"refs/heads/master"},"40f046d9b3fc2f5a0b6403f1bb560f3121d6505d":{"kind":"REWORK","_number":2,"created":"2023-11-11 15:57:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/04/900704/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/04/900704/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/2"}}},"commit":{"parents":[{"commit":"c69ceb7faf60b2cae181c263c7b5cb57dcf9ed72","subject":"Merge \"neutron_network: Remove l3_ext network for NSX plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c69ceb7faf60b2cae181c263c7b5cb57dcf9ed72"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-11 15:52:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-11 15:56:53.000000000","tz":540},"subject":"Add class to manage VPNaaS service plugin","message":"Add class to manage VPNaaS service plugin\n\nTo use neutron VPNaaS, we have to configure VPNaaS service plugin as\nwell as VPNaaS agent. This introduces the class to manage the service\nplugin which has been missing.\n\nChange-Id: I778a10d8375b841e3cf9dcb3fb09ec10fe669c79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/40f046d9b3fc2f5a0b6403f1bb560f3121d6505d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/40f046d9b3fc2f5a0b6403f1bb560f3121d6505d"}]},"branch":"refs/heads/master"},"537b6b6a5b31c162a158bcd8978a8424b84566a8":{"kind":"REWORK","_number":3,"created":"2023-11-11 16:01:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/04/900704/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/04/900704/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/3"}}},"commit":{"parents":[{"commit":"c69ceb7faf60b2cae181c263c7b5cb57dcf9ed72","subject":"Merge \"neutron_network: Remove l3_ext network for NSX plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c69ceb7faf60b2cae181c263c7b5cb57dcf9ed72"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-11 15:52:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-11 16:01:15.000000000","tz":540},"subject":"Add class to manage VPNaaS service plugin","message":"Add class to manage VPNaaS service plugin\n\nTo use neutron VPNaaS, we have to configure VPNaaS service plugin as\nwell as VPNaaS agent. This introduces the class to manage the service\nplugin which has been missing.\n\nChange-Id: I778a10d8375b841e3cf9dcb3fb09ec10fe669c79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/537b6b6a5b31c162a158bcd8978a8424b84566a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/537b6b6a5b31c162a158bcd8978a8424b84566a8"}]},"branch":"refs/heads/master"},"bbc7bbbdfc0f21056a095336b473408529191d60":{"kind":"REWORK","_number":4,"created":"2023-11-11 16:08:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/04/900704/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/04/900704/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/4"}}},"commit":{"parents":[{"commit":"c69ceb7faf60b2cae181c263c7b5cb57dcf9ed72","subject":"Merge \"neutron_network: Remove l3_ext network for NSX plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c69ceb7faf60b2cae181c263c7b5cb57dcf9ed72"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-11 15:52:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-11 16:08:50.000000000","tz":0},"subject":"Add class to manage VPNaaS service plugin","message":"Add class to manage VPNaaS service plugin\n\nTo use neutron VPNaaS, we have to configure VPNaaS service plugin as\nwell as VPNaaS agent. This introduces the class to manage the service\nplugin which has been missing.\n\nChange-Id: I778a10d8375b841e3cf9dcb3fb09ec10fe669c79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/bbc7bbbdfc0f21056a095336b473408529191d60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/bbc7bbbdfc0f21056a095336b473408529191d60"}]},"branch":"refs/heads/master"},"8e5edffc99021723b471275489e4af2bb3d948e0":{"kind":"REWORK","_number":5,"created":"2023-11-11 16:10:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/04/900704/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/04/900704/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/5"}}},"commit":{"parents":[{"commit":"c69ceb7faf60b2cae181c263c7b5cb57dcf9ed72","subject":"Merge \"neutron_network: Remove l3_ext network for NSX plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c69ceb7faf60b2cae181c263c7b5cb57dcf9ed72"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-11 15:52:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-11 16:10:07.000000000","tz":540},"subject":"Add class to manage VPNaaS service plugin","message":"Add class to manage VPNaaS service plugin\n\nTo use neutron VPNaaS, we have to configure VPNaaS service plugin as\nwell as VPNaaS agent. This introduces the class to manage the service\nplugin which has been missing.\n\nChange-Id: I778a10d8375b841e3cf9dcb3fb09ec10fe669c79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/8e5edffc99021723b471275489e4af2bb3d948e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/8e5edffc99021723b471275489e4af2bb3d948e0"}]},"branch":"refs/heads/master"},"6ab717d7cc8093d6a192f136d64f0e8bfb292777":{"kind":"REWORK","_number":6,"created":"2023-11-11 16:29:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/04/900704/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/04/900704/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/6"}}},"commit":{"parents":[{"commit":"c69ceb7faf60b2cae181c263c7b5cb57dcf9ed72","subject":"Merge \"neutron_network: Remove l3_ext network for NSX plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c69ceb7faf60b2cae181c263c7b5cb57dcf9ed72"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-11 15:52:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-11 16:28:53.000000000","tz":540},"subject":"Add class to manage VPNaaS service plugin","message":"Add class to manage VPNaaS service plugin\n\nTo use neutron VPNaaS, we have to configure VPNaaS service plugin as\nwell as VPNaaS agent. This introduces the class to manage the service\nplugin which has been missing.\n\nChange-Id: I778a10d8375b841e3cf9dcb3fb09ec10fe669c79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6ab717d7cc8093d6a192f136d64f0e8bfb292777"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6ab717d7cc8093d6a192f136d64f0e8bfb292777"}]},"branch":"refs/heads/master"},"8909090a1fa1692177ea1758bf97556b784a4ee5":{"kind":"REWORK","_number":7,"created":"2023-11-11 17:39:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/04/900704/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/04/900704/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/7"}}},"commit":{"parents":[{"commit":"c69ceb7faf60b2cae181c263c7b5cb57dcf9ed72","subject":"Merge \"neutron_network: Remove l3_ext network for NSX plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c69ceb7faf60b2cae181c263c7b5cb57dcf9ed72"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-11 15:52:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-11 17:39:24.000000000","tz":540},"subject":"Add class to manage VPNaaS service plugin","message":"Add class to manage VPNaaS service plugin\n\nTo use neutron VPNaaS, we have to configure VPNaaS service plugin as\nwell as VPNaaS agent. This introduces the class to manage the service\nplugin which has been missing.\n\nChange-Id: I778a10d8375b841e3cf9dcb3fb09ec10fe669c79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/8909090a1fa1692177ea1758bf97556b784a4ee5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/8909090a1fa1692177ea1758bf97556b784a4ee5"}]},"branch":"refs/heads/master"},"b2f246528db0ca0e1039960e0801f124a1e120e5":{"kind":"REWORK","_number":8,"created":"2023-11-11 17:41:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/04/900704/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/04/900704/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/8"}}},"commit":{"parents":[{"commit":"c69ceb7faf60b2cae181c263c7b5cb57dcf9ed72","subject":"Merge \"neutron_network: Remove l3_ext network for NSX plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c69ceb7faf60b2cae181c263c7b5cb57dcf9ed72"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-11 15:52:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-11 17:41:35.000000000","tz":540},"subject":"Add class to manage VPNaaS service plugin","message":"Add class to manage VPNaaS service plugin\n\nTo use neutron VPNaaS, we have to configure VPNaaS service plugin as\nwell as VPNaaS agent. This introduces the class to manage the service\nplugin which has been missing.\n\nChange-Id: I778a10d8375b841e3cf9dcb3fb09ec10fe669c79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b2f246528db0ca0e1039960e0801f124a1e120e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b2f246528db0ca0e1039960e0801f124a1e120e5"}]},"branch":"refs/heads/master"},"9c2d1fbafee3546e96349c5a504a9bbdac5eefbc":{"kind":"REWORK","_number":9,"created":"2023-11-11 18:52:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/04/900704/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/04/900704/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/9"}}},"commit":{"parents":[{"commit":"c69ceb7faf60b2cae181c263c7b5cb57dcf9ed72","subject":"Merge \"neutron_network: Remove l3_ext network for NSX plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c69ceb7faf60b2cae181c263c7b5cb57dcf9ed72"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-11 15:52:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-11 18:51:57.000000000","tz":540},"subject":"Add class to manage VPNaaS service plugin","message":"Add class to manage VPNaaS service plugin\n\nTo use neutron VPNaaS, we have to configure VPNaaS service plugin as\nwell as VPNaaS agent. This introduces the class to manage the service\nplugin which has been missing.\n\nChange-Id: I778a10d8375b841e3cf9dcb3fb09ec10fe669c79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/9c2d1fbafee3546e96349c5a504a9bbdac5eefbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/9c2d1fbafee3546e96349c5a504a9bbdac5eefbc"}]},"branch":"refs/heads/master"},"486ac6d5faebefaf6d334650a16735d9ab3ebdc1":{"kind":"REWORK","_number":10,"created":"2023-11-11 19:32:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/04/900704/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/04/900704/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/10"}}},"commit":{"parents":[{"commit":"c69ceb7faf60b2cae181c263c7b5cb57dcf9ed72","subject":"Merge \"neutron_network: Remove l3_ext network for NSX plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c69ceb7faf60b2cae181c263c7b5cb57dcf9ed72"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-11 15:52:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-11 19:31:19.000000000","tz":540},"subject":"Add class to manage VPNaaS service plugin","message":"Add class to manage VPNaaS service plugin\n\nTo use neutron VPNaaS, we have to configure VPNaaS service plugin as\nwell as VPNaaS agent. This introduces the class to manage the service\nplugin which has been missing.\n\nChange-Id: I778a10d8375b841e3cf9dcb3fb09ec10fe669c79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/486ac6d5faebefaf6d334650a16735d9ab3ebdc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/486ac6d5faebefaf6d334650a16735d9ab3ebdc1"}]},"branch":"refs/heads/master"},"587e4927eabf6dd42bd1bf8b0a77f18046e1490d":{"kind":"REWORK","_number":11,"created":"2023-11-12 05:19:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/04/900704/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/04/900704/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/11"}}},"commit":{"parents":[{"commit":"c69ceb7faf60b2cae181c263c7b5cb57dcf9ed72","subject":"Merge \"neutron_network: Remove l3_ext network for NSX plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c69ceb7faf60b2cae181c263c7b5cb57dcf9ed72"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-11 15:52:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-12 05:19:08.000000000","tz":540},"subject":"Add class to manage VPNaaS service plugin","message":"Add class to manage VPNaaS service plugin\n\nTo use neutron VPNaaS, we have to configure VPNaaS service plugin as\nwell as VPNaaS agent. This introduces the class to manage the service\nplugin which has been missing.\n\nChange-Id: I778a10d8375b841e3cf9dcb3fb09ec10fe669c79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/587e4927eabf6dd42bd1bf8b0a77f18046e1490d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/587e4927eabf6dd42bd1bf8b0a77f18046e1490d"}]},"branch":"refs/heads/master"},"f8bc09e1af079b1e4bb69edefb5aae1c44b0dde1":{"kind":"REWORK","_number":12,"created":"2023-11-12 05:57:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/04/900704/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/04/900704/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/12"}}},"commit":{"parents":[{"commit":"c69ceb7faf60b2cae181c263c7b5cb57dcf9ed72","subject":"Merge \"neutron_network: Remove l3_ext network for NSX plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c69ceb7faf60b2cae181c263c7b5cb57dcf9ed72"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-11 15:52:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-12 05:57:11.000000000","tz":540},"subject":"Add class to manage VPNaaS service plugin","message":"Add class to manage VPNaaS service plugin\n\nTo use neutron VPNaaS, we have to configure VPNaaS service plugin as\nwell as VPNaaS agent. This introduces the class to manage the service\nplugin which has been missing.\n\nChange-Id: I778a10d8375b841e3cf9dcb3fb09ec10fe669c79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/f8bc09e1af079b1e4bb69edefb5aae1c44b0dde1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/f8bc09e1af079b1e4bb69edefb5aae1c44b0dde1"}]},"branch":"refs/heads/master"},"bfa8b9fe28add7692ffda92373eb9613367a42d7":{"kind":"REWORK","_number":13,"created":"2023-11-12 10:38:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/04/900704/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/04/900704/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/13"}}},"commit":{"parents":[{"commit":"c69ceb7faf60b2cae181c263c7b5cb57dcf9ed72","subject":"Merge \"neutron_network: Remove l3_ext network for NSX plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c69ceb7faf60b2cae181c263c7b5cb57dcf9ed72"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-11 15:52:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-12 10:37:05.000000000","tz":540},"subject":"Add class to manage VPNaaS service plugin","message":"Add class to manage VPNaaS service plugin\n\nTo use neutron VPNaaS, we have to configure VPNaaS service plugin as\nwell as VPNaaS agent. This introduces the class to manage the service\nplugin which has been missing.\n\nAlso, python3-neutron-vpnaas package in CentOS does not provide any\nconfig files required to set up agents, and we should install\nthe openstack-neutron-vpnaas package instead.\n\nChange-Id: I778a10d8375b841e3cf9dcb3fb09ec10fe669c79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/bfa8b9fe28add7692ffda92373eb9613367a42d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/bfa8b9fe28add7692ffda92373eb9613367a42d7"}]},"branch":"refs/heads/master"},"7059883db34c3e40e558b04ecea47e476fd2b2d1":{"kind":"REWORK","_number":14,"created":"2023-11-12 12:41:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/04/900704/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/04/900704/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/14"}}},"commit":{"parents":[{"commit":"c69ceb7faf60b2cae181c263c7b5cb57dcf9ed72","subject":"Merge \"neutron_network: Remove l3_ext network for NSX plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c69ceb7faf60b2cae181c263c7b5cb57dcf9ed72"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-11 15:52:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-12 12:40:45.000000000","tz":540},"subject":"Add class to manage VPNaaS service plugin","message":"Add class to manage VPNaaS service plugin\n\nTo use neutron VPNaaS, we have to configure VPNaaS service plugin as\nwell as VPNaaS agent. This introduces the class to manage the service\nplugin in the separate neutron_vpnaas.conf file.\n\nChange-Id: I778a10d8375b841e3cf9dcb3fb09ec10fe669c79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/7059883db34c3e40e558b04ecea47e476fd2b2d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/7059883db34c3e40e558b04ecea47e476fd2b2d1"}]},"branch":"refs/heads/master"},"ba6d1045a78e6c8a63ce8b1009981eae62656d75":{"kind":"NO_CODE_CHANGE","_number":15,"created":"2023-11-12 12:49:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/04/900704/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/04/900704/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/04/900704/15"}}},"commit":{"parents":[{"commit":"c69ceb7faf60b2cae181c263c7b5cb57dcf9ed72","subject":"Merge \"neutron_network: Remove l3_ext network for NSX plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c69ceb7faf60b2cae181c263c7b5cb57dcf9ed72"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-11 15:52:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-12 12:44:07.000000000","tz":540},"subject":"Add class to manage VPNaaS service plugin","message":"Add class to manage VPNaaS service plugin\n\nThe neutron-vpnnas package provides the separate config file for\nvpnaas service plugin (neutron_vpnaas.conf). This change introduces\nthe new class to leverage that separate file.\n\nChange-Id: I778a10d8375b841e3cf9dcb3fb09ec10fe669c79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/ba6d1045a78e6c8a63ce8b1009981eae62656d75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/ba6d1045a78e6c8a63ce8b1009981eae62656d75"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
