)]}'
{"id":"openstack%2Fpuppet-neutron~906016","triplet_id":"openstack%2Fpuppet-neutron~master~I8b4dfd6cd02d61cb4ef07e117af39c44420e93a4","project":"openstack/puppet-neutron","branch":"master","topic":"taas","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-01-22 17:08:23.000000000","reason":"Change was submitted"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2024-01-21 07:17:08.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_9816\u003e by clicking the attention icon","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I8b4dfd6cd02d61cb4ef07e117af39c44420e93a4","subject":"Add support for Taas (Tap-as-a-Service)","status":"MERGED","created":"2024-01-18 02:20:34.000000000","updated":"2024-01-22 17:09:23.000000000","submitted":"2024-01-22 17:08:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906016-taas","meta_rev_id":"5f77f07ac2dd929485b91929aa9f90f515f7b004","_number":906016,"virtual_id_number":906016,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-22 17:08:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2024-01-22 15:06:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2024-01-22 15:45:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2024-01-22 15:45:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-18 04:38:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-20 22:24:41.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-01-22 15:45:18.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"2b915cf4b4202dad600b64e25cf0f518979cd1e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-18 02:20:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"731352b5f7b344e6dcd30376f5c7577bbfafc263","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-18 02:29:08.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"f0a72afb6aa39e8aefe317e0d9e1471832b7793b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-18 02:33:56.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"93f550d7ce34c06351ac615eedf30e351ddc0a0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 04:38:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/63f19fc3d5e54c0ea3de0609fb1a1536\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a8623bd0ac3472391edd979e80f0dd8 : SUCCESS in 2m 51s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1cef809bbe2e4eecbfe6eb4c9c39fe80 : SUCCESS in 2m 56s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a7ed652c9b4492397232792f0d02077 : SUCCESS in 6m 57s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/173b7697317d4dcdb2f2e3be4aac2128 : SUCCESS in 15m 48s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ec494849d024743bcfd632a326ce202 : SUCCESS in 11m 59s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/ba061022bcbe46f086635a24602663dd : SUCCESS in 17m 31s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b6566a7d70d42e28153ad11fb238ccf : SUCCESS in 12m 53s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9705913aeb6d4f6e85c545dc7c6e8cad : SUCCESS in 43m 20s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e3f70c01c83b4a648f6c345f22d40bec : FAILURE in 56m 29s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1f9cd1f111834bc3a4fbcdcc888dd8bd : SUCCESS in 49m 01s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/268d91d7c8ea400288c80ceebd975f72 : SUCCESS in 47m 36s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9f40ad0ead0444da3972c096168923d : SUCCESS in 38m 21s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a2e0b0c9a4b14624b79a9d59c3b4c281 : SUCCESS in 1h 05m 15s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/aba895079a41493fb63e659b30a5cf10 : SUCCESS in 1h 19m 46s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/31da3dc162bd4bcda698b85f8e9f9a2e : SUCCESS in 1h 06m 58s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5dc60df1a6c64803b5221a38115b0498 : FAILURE in 1h 02m 37s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/30145d3dbe6445c0b09f304f6ca15de9 : SUCCESS in 58m 03s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/5eeab0a470b64a0b9303fa67b8624e61 : SUCCESS in 19m 18s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/51b0c05194454366985d023f2a220545 : SUCCESS in 19m 38s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3298d216b68c4f1cabd35c15e387a8ea : SUCCESS in 13m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a0bb8f0b708c842269713e7a44d29c714910fa5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-18 08:39:58.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"eb02c8f701a76e2a0fa9c392c6cc1d21b022c769","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 10:10:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e288c924ee2b40289f5ea67f7575b3e4\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac010f2fbf28419e855a611e0df731c6 : SUCCESS in 2m 54s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67938fde48874cbca7d2bbeeb86843fb : SUCCESS in 2m 28s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/f9db4da9aaa64f049a6b7ad0bd8c3a42 : SUCCESS in 5m 44s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/5cd15f6f39e24823ba7e0c76d66ff613 : SUCCESS in 16m 35s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fedaf4c77ac6485390464f324a7d6fb7 : SUCCESS in 10m 35s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/583ba43db30d47c8adef9e96f9beeeec : SUCCESS in 13m 05s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ed6534370b284dcf94f41fda81925d5a : SUCCESS in 11m 37s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8ff61122d5c04dc2b1b04b0933cf89f7 : SUCCESS in 46m 24s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aca4ea8633c1498a8ddf96a728fe2791 : FAILURE in 54m 23s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b5df454e540e47e4a77e94a440d1b71d : SUCCESS in 52m 21s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9674684b9c814dcca0cbecab682d3c88 : SUCCESS in 46m 20s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6193e3c88394446ebe6ac3de3ffa48e3 : SUCCESS in 44m 44s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/51c21454044641c3930d4cce72fd4ae0 : SUCCESS in 59m 32s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a7c4c0cd47b440eb8f4357d48e0d349d : SUCCESS in 1h 25m 01s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/30dc55e8bfa3420fba76ce2637d83b9c : SUCCESS in 1h 08m 08s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/66c3a1b3f944476f98c00f3546ea7de9 : SUCCESS in 1h 01m 51s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/623b2ac1a26c4719af8c41837b1ebb39 : SUCCESS in 57m 51s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/4d7f09ea745e4728819dc9e40ccb493f : SUCCESS in 20m 28s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52b1f08fa1564baf88affd56d8ffaa1a : SUCCESS in 14m 33s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c9b57dc35364854a6dc8e4ca64b20a0 : SUCCESS in 15m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"52e631c24399421ed9547d7760ce4cc04d66bb46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-18 12:37:03.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"5fb8ffa72c770ed39f6e049bc3f27f355ed6f549","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-18 13:40:46.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2ea2615412a593414f619b487fd7f55bdef66500","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-18 13:45:01.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"9feca3c7a9eaa9a073320218ee975e7f3f9315b0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-18 14:16:12.000000000","message":"Patch Set 7: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"8a7fb666c7b2338250a6d203ce7116ccff746c0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 15:16:11.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2f84e39b5684e4183f575b0aaf82120\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1246a699405041d0b1856b7ef385d72d : SUCCESS in 2m 59s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/14d80988cd72413289694aaf4aca3c61 : SUCCESS in 2m 32s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/9a4c223682ab46b7aae2034a3a0658d2 : SUCCESS in 5m 20s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/933dd73e27c9427f8664c1a69ea77aa3 : SUCCESS in 13m 54s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/20d34c48c6014df48679cd45842fcbb8 : SUCCESS in 12m 35s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/03b6363ce14c4307823c2be76e1f42f7 : SUCCESS in 13m 09s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c8523ba470a46f1a59e107d1a0d5318 : SUCCESS in 12m 50s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e84411a209d446988385f736a850a87b : SUCCESS in 48m 06s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0eedf35fd881456c801208ab6047de1b : SUCCESS in 57m 34s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2684e23129b4443e9910ea4161dbef91 : SUCCESS in 52m 54s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6345a13d3f3d4c5c93e8257584dc4232 : SUCCESS in 51m 16s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df6b4e586b504038b03a83859a4164d2 : SUCCESS in 45m 28s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c7f57a54a4fc42fb86b7f4acbded6ce3 : SUCCESS in 1h 06m 40s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c88517a63bc640aca15efbda749e40b8 : SUCCESS in 1h 23m 16s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9279e599c848497caf4f2da369e8a44f : SUCCESS in 1h 08m 30s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c6aa3575c8c6460381790f34a1e8b4bd : SUCCESS in 52m 16s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7f1f50870dcc426a83f619a6af210cdd : SUCCESS in 1h 02m 45s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/2141824ede4a4676acab99dfe48440d2 : SUCCESS in 23m 03s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe4dbe102f654b47b44d4c3d9b4d4a7c : SUCCESS in 15m 52s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8461e248e9ba4076840c7e0bbe227f97 : SUCCESS in 14m 38s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3473efb36a674f10be2813393d09de4f : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":7},{"id":"b02419b0ce25f9c1b65c6098b3aafc5317c82600","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-18 16:02:24.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":8},{"id":"37e2295c5df33284265a3c6cbe051048cd80f380","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 17:30:52.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9bbd3b2640f24a8b99abc782f61bc0d1\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6117928e233649ebbeb1cbae1d16f548 : SUCCESS in 3m 03s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/648e6a7ebfed4e099a8c08d8558f81e5 : SUCCESS in 2m 38s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/3e3c0dbbe7ab4126a74f6a76192239cb : SUCCESS in 5m 47s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/d857d36aedb24ba29bd67d9289b58414 : SUCCESS in 15m 38s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c602d7d80084f19b15c2d7d68d65cf2 : SUCCESS in 12m 45s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/120dfa2211c54b548a180961cfa5434b : SUCCESS in 16m 20s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07f552eb33074abea2693654a7990b0d : SUCCESS in 13m 01s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ef3e572680c4e4e8dfc87b66fb922c5 : SUCCESS in 44m 53s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/18aa69e69e8841a5957d173318a8145f : SUCCESS in 1h 10m 10s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/503e9290e9884b7da7b554e35ac7ded6 : SUCCESS in 49m 00s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc9d1bedbddc4888bd7925bd29d20c0c : SUCCESS in 49m 50s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/01e9dcd9e26542b89f746d38385f46a7 : SUCCESS in 42m 49s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5e418b1803054253bb97ad24ae1e1cb0 : SUCCESS in 1h 00m 36s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d7a0df5ab5e34f5ca1a593d4d4c2e3c4 : SUCCESS in 1h 21m 15s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2253e3447c11491f9fad5e6726a7977f : FAILURE in 1h 02m 17s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/92c64dda8c6d49da9323e2300cca5fc8 : SUCCESS in 1h 06m 24s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/27dcb545a0b8436c9f1021953bb2f0cb : SUCCESS in 1h 02m 11s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/f62bdfdfa1154701a05c84d85bd3378a : SUCCESS in 17m 01s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/347be486e6c04c298eac93ccdc3a9156 : SUCCESS in 14m 21s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c93b2283a68c4d33a20b8607b13a21ba : SUCCESS in 14m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa77d2318f334bd09f3f58cdf0528b45 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":8},{"id":"3cc8b2283aa31c1f2fc61a9044693c32559068ee","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-19 01:08:13.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f52391349904c5677bdd67469f5fa07009863679","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-19 01:08:18.000000000","message":"Patch Set 8: Workflow-1","accounts_in_message":[],"_revision_number":8},{"id":"e0e9205b5874a1e8b59d1bc955701d201885b02e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 02:39:33.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fa13344ed704c5e8b69e4d487425a29\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b1ba2119e6046908357b6488add4dde : SUCCESS in 2m 35s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bcb2d4bfbdd545759832c6cd88448730 : SUCCESS in 2m 34s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/9171458f65524436a87108488f9dc840 : SUCCESS in 5m 36s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/b2ccb939869c4d96b8f3b4b5f292be85 : SUCCESS in 13m 18s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d6d688932724457581748edacc4ff29d : SUCCESS in 12m 12s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/9fcbb247204843079a528346085933c9 : SUCCESS in 13m 20s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7672f1bec25a45ce90c0a1001d2dd3b8 : SUCCESS in 11m 40s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/837ca189ece142a782ca9e21f4ef9fe0 : SUCCESS in 48m 58s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/399c4b22f3cb4651ad5e3012a605ce11 : SUCCESS in 1h 02m 51s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/124c697b8deb490c82128928489d0d4c : SUCCESS in 54m 33s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0697b10c0fef40e1b557248cae527b37 : SUCCESS in 43m 26s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/53679f146c80493b89083b811b3444bc : FAILURE in 47m 53s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4dee4d0de8e248d3a03163055c685b81 : SUCCESS in 1h 00m 19s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4e2ad42cbffe4a1989d0876600dc8be9 : SUCCESS in 1h 25m 09s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e1cf42d9cb3e479e8cb4382a4275354e : SUCCESS in 1h 00m 56s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4e584f0c06fa47ae8431a6794daa0884 : SUCCESS in 58m 04s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e8b196e10d484558bf0d2dd50e9e3d38 : SUCCESS in 58m 05s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/6f1ed96bd34048f99b99512bea4cf2e1 : SUCCESS in 18m 38s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a592118501c64b6aacb6fd8d6d5659b5 : SUCCESS in 14m 04s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f91a28a02a04418b98ca8d87f29a6aa2 : SUCCESS in 14m 49s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d282212e5d734584a80fa49059adf380 : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":8},{"id":"6697c49f30f4a4a063dd652f743e288290ec5189","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-19 14:50:22.000000000","message":"Patch Set 8: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"37df9f8bb846ee1b2696df01cd5ca1971b1948ed","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-01-20 22:24:41.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"e424c980c1b4a7a09e9b899dc106a4ba14f7bddc","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-01-20 22:26:22.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"0c393b03a0528744b1b78dc3f1d91b5c9d7d3277","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-21 06:13:46.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"054534f0bbd9b6ae8594acf136709cf6f8ddd39d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-21 06:14:40.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"2f53c1de273dc0a67d5282b5183f22256c22274a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-21 06:58:41.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"5451a5dd3f026ab4db69be774d5d531e7b71fb3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-21 08:32:34.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6736c9142c514e7f8eaf9267b2b60194\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7283d5bc0c174b6abc197ef664ec790b : SUCCESS in 2m 54s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05ef1feaa64a40d287d279f11f89e04a : SUCCESS in 3m 16s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/21ae558780934219a6a1c6a14c545d47 : SUCCESS in 6m 05s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/146cf9dfb03d438982325f8586289273 : SUCCESS in 15m 53s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de51bdd20c63497aac39259f1c1fce1e : SUCCESS in 8m 18s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/1645968391844dfca2c2fb9d97f4bf8d : SUCCESS in 15m 53s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e8595f2fca1f411cbbdf2943122b39dd : SUCCESS in 11m 55s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bd68400bcd0140c88e5db48b9661ea65 : SUCCESS in 36m 44s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/009bb70378a6442188e45ec6351560f3 : SUCCESS in 55m 29s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/06dbd758009d4364a4cd18d02eee46b5 : SUCCESS in 54m 18s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2e4d5d36697940cb8e3aace511e1e26e : SUCCESS in 42m 41s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/44a0210c84c14941a100b1b40c2191d1 : SUCCESS in 38m 44s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5cac1a90f91e42659ff1484c948a2d7e : SUCCESS in 46m 18s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/214d0c2382bd4d41a8b6232b83945835 : SUCCESS in 1h 17m 48s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/681086a87a9a4e5f8e3eca25047996e3 : SUCCESS in 1h 08m 14s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e76f5e494ac6444c8ff1d71af87647c5 : SUCCESS in 40m 52s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8f37729ea3e140c58a92663847ead6f8 : SUCCESS in 50m 25s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/aefccc2dc61246c6b86032b6ca07636b : SUCCESS in 21m 38s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e287e9ab9a9b4053bf6180e0fe50c1a3 : SUCCESS in 14m 19s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab53edb2ca0e4956b74371de962f1f1a : SUCCESS in 10m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae0caf82b7434a8ab9e32a7cf68a7872 : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":9},{"id":"b5e3e5decc9642b976e2d37a73ecb91a9d676b90","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-01-22 15:06:30.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"e722db85e535922320d6ac4c82164673620d55b3","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-01-22 15:45:18.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"45f8d530083728a4899c1d73806e6778918dd3aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 15:46:31.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"ed9f803d4a9ee36ba38656d7efa48d5be9069955","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 17:08:23.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/821bcdb2ea164d1e9597cf38e27e79e2\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/15ef8ac8608d47d4b6ede530062441db : SUCCESS in 2m 54s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2520cd8269f46daa13aa4e17923818c : SUCCESS in 2m 48s\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/bb41763c33a443daa6b898e584768746 : SUCCESS in 16m 31s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a193f8c1ac924c32ad96825abd075a95 : SUCCESS in 14m 04s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/16533647ea9249079ccafd9efc1533ce : SUCCESS in 1h 10m 25s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4823ae624fd040739ffc934a402446ee : SUCCESS in 1h 17m 31s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a1f325f865c048a9a25d040196e0444b : SUCCESS in 1h 09m 47s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cddb899ae9c34eaebba252c6117317ea : SUCCESS in 53m 14s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f89891f26eac41e393da6ef7a4ec5453 : SUCCESS in 1h 00m 53s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/700a8e9bdbdf49cd92f28fc75f937729 : SUCCESS in 22m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aed7663364b1474a8e5e676dbe7a1d0e : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":9},{"id":"4ce66441fae206dacd5e5ad480b7b6574bd4ab78","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 17:08:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"5f77f07ac2dd929485b91929aa9f90f515f7b004","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 17:09:23.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09a3375c292144818a8cb7a9e3d76bd3\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e84f6bd3218c443b8ef00f5303749810 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"be83863a17c2b27af87f1ba409690d36c0ed23cd","revisions":{"60c5a37e738c0e5dd82b57cbb4fc92e4f47eb9e9":{"kind":"REWORK","_number":1,"created":"2024-01-18 02:20:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/16/906016/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/16/906016/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/16/906016/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/16/906016/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/16/906016/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/16/906016/1"}}},"commit":{"parents":[{"commit":"6dd346b71932c98b1299f09b093222a5aaa59cfc","subject":"healthcheck: Expose allowed_source_ranges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6dd346b71932c98b1299f09b093222a5aaa59cfc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-18 02:14:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-18 02:20:07.000000000","tz":540},"subject":"wip: Add support for Taas (Tap as a Service)","message":"wip: Add support for Taas (Tap as a Service)\n\nChange-Id: I8b4dfd6cd02d61cb4ef07e117af39c44420e93a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/60c5a37e738c0e5dd82b57cbb4fc92e4f47eb9e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/60c5a37e738c0e5dd82b57cbb4fc92e4f47eb9e9"}]},"branch":"refs/heads/master"},"f5fb21ecdfec092f273545287b25e19aa6dc22dd":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-01-18 02:29:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/16/906016/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/16/906016/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/16/906016/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/16/906016/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/16/906016/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/16/906016/2"}}},"commit":{"parents":[{"commit":"6773794242d507c6af8ad527346a22a198deccfc","subject":"Revert changes unintentionally included","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6773794242d507c6af8ad527346a22a198deccfc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-18 02:14:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-18 02:29:08.000000000","tz":0},"subject":"wip: Add support for Taas (Tap as a Service)","message":"wip: Add support for Taas (Tap as a Service)\n\nChange-Id: I8b4dfd6cd02d61cb4ef07e117af39c44420e93a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/f5fb21ecdfec092f273545287b25e19aa6dc22dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/f5fb21ecdfec092f273545287b25e19aa6dc22dd"}]},"branch":"refs/heads/master","description":"Rebase"},"2b9ad2c222726db9473fccd377ca98808444574c":{"kind":"REWORK","_number":3,"created":"2024-01-18 02:33:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/16/906016/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/16/906016/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/16/906016/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/16/906016/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/16/906016/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/16/906016/3"}}},"commit":{"parents":[{"commit":"6773794242d507c6af8ad527346a22a198deccfc","subject":"Revert changes unintentionally included","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6773794242d507c6af8ad527346a22a198deccfc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-18 02:14:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-18 02:33:56.000000000","tz":0},"subject":"wip: Add support for Taas (Tap as a Service)","message":"wip: Add support for Taas (Tap as a Service)\n\nChange-Id: I8b4dfd6cd02d61cb4ef07e117af39c44420e93a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/2b9ad2c222726db9473fccd377ca98808444574c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/2b9ad2c222726db9473fccd377ca98808444574c"}]},"branch":"refs/heads/master"},"1122732f50cfc8084a898b26d426fb94d7387f55":{"kind":"REWORK","_number":4,"created":"2024-01-18 08:39:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/16/906016/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/16/906016/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/16/906016/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/16/906016/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/16/906016/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/16/906016/4"}}},"commit":{"parents":[{"commit":"6773794242d507c6af8ad527346a22a198deccfc","subject":"Revert changes unintentionally included","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6773794242d507c6af8ad527346a22a198deccfc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-18 02:14:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-18 08:39:39.000000000","tz":540},"subject":"wip: Add support for Taas (Tap as a Service)","message":"wip: Add support for Taas (Tap as a Service)\n\nChange-Id: I8b4dfd6cd02d61cb4ef07e117af39c44420e93a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/1122732f50cfc8084a898b26d426fb94d7387f55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/1122732f50cfc8084a898b26d426fb94d7387f55"}]},"branch":"refs/heads/master"},"0be881e017c3bfedb0978389c051f672da81cd24":{"kind":"REWORK","_number":5,"created":"2024-01-18 12:37:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/16/906016/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/16/906016/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/16/906016/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/16/906016/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/16/906016/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/16/906016/5"}}},"commit":{"parents":[{"commit":"6773794242d507c6af8ad527346a22a198deccfc","subject":"Revert changes unintentionally included","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6773794242d507c6af8ad527346a22a198deccfc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-18 02:14:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-18 12:36:43.000000000","tz":540},"subject":"wip: Add support for Taas (Tap as a Service)","message":"wip: Add support for Taas (Tap as a Service)\n\nChange-Id: I8b4dfd6cd02d61cb4ef07e117af39c44420e93a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/0be881e017c3bfedb0978389c051f672da81cd24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/0be881e017c3bfedb0978389c051f672da81cd24"}]},"branch":"refs/heads/master"},"0ca227b9ebb3d56e242c25f7e05306e5e9a740f2":{"kind":"REWORK","_number":6,"created":"2024-01-18 13:40:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/16/906016/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/16/906016/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/16/906016/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/16/906016/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/16/906016/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/16/906016/6"}}},"commit":{"parents":[{"commit":"6773794242d507c6af8ad527346a22a198deccfc","subject":"Revert changes unintentionally included","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6773794242d507c6af8ad527346a22a198deccfc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-18 02:14:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-18 13:39:33.000000000","tz":540},"subject":"Add support for Taas (Tap-as-a-Service)","message":"Add support for Taas (Tap-as-a-Service)\n\nThis introduces support for the Tap-as-a-Service plugin[1].\n\n[1] https://opendev.org/openstack/tap-as-a-service\n\nChange-Id: I8b4dfd6cd02d61cb4ef07e117af39c44420e93a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/0ca227b9ebb3d56e242c25f7e05306e5e9a740f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/0ca227b9ebb3d56e242c25f7e05306e5e9a740f2"}]},"branch":"refs/heads/master"},"c3faec46790e185878e0a4c82d6f60305d75fa3d":{"kind":"REWORK","_number":7,"created":"2024-01-18 13:45:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/16/906016/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/16/906016/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/16/906016/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/16/906016/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/16/906016/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/16/906016/7"}}},"commit":{"parents":[{"commit":"d984412b86bc99995f83aa14cd800e6b8569c73e","subject":"Fix incomplete acceptance tests of neutron_vpnaas_service_config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/d984412b86bc99995f83aa14cd800e6b8569c73e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-18 02:14:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-18 13:44:22.000000000","tz":540},"subject":"Add support for Taas (Tap-as-a-Service)","message":"Add support for Taas (Tap-as-a-Service)\n\nThis introduces support for the Tap-as-a-Service plugin[1].\n\n[1] https://opendev.org/openstack/tap-as-a-service\n\nChange-Id: I8b4dfd6cd02d61cb4ef07e117af39c44420e93a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c3faec46790e185878e0a4c82d6f60305d75fa3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c3faec46790e185878e0a4c82d6f60305d75fa3d"}]},"branch":"refs/heads/master"},"b8da591c5fe5dd1f355688f3b009ef027c60ae93":{"kind":"REWORK","_number":8,"created":"2024-01-18 16:02:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/16/906016/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/16/906016/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/16/906016/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/16/906016/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/16/906016/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/16/906016/8"}}},"commit":{"parents":[{"commit":"d984412b86bc99995f83aa14cd800e6b8569c73e","subject":"Fix incomplete acceptance tests of neutron_vpnaas_service_config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/d984412b86bc99995f83aa14cd800e6b8569c73e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-18 02:14:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-18 16:02:09.000000000","tz":540},"subject":"Add support for Taas (Tap-as-a-Service)","message":"Add support for Taas (Tap-as-a-Service)\n\nThis introduces support for the Tap-as-a-Service plugin[1].\n\n[1] https://opendev.org/openstack/tap-as-a-service\n\nChange-Id: I8b4dfd6cd02d61cb4ef07e117af39c44420e93a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b8da591c5fe5dd1f355688f3b009ef027c60ae93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b8da591c5fe5dd1f355688f3b009ef027c60ae93"}]},"branch":"refs/heads/master"},"be83863a17c2b27af87f1ba409690d36c0ed23cd":{"kind":"REWORK","_number":9,"created":"2024-01-21 06:13:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/16/906016/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/16/906016/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/16/906016/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/16/906016/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/16/906016/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/16/906016/9"}}},"commit":{"parents":[{"commit":"d984412b86bc99995f83aa14cd800e6b8569c73e","subject":"Fix incomplete acceptance tests of neutron_vpnaas_service_config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/d984412b86bc99995f83aa14cd800e6b8569c73e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-18 02:14:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-21 06:13:45.000000000","tz":0},"subject":"Add support for Taas (Tap-as-a-Service)","message":"Add support for Taas (Tap-as-a-Service)\n\nThis introduces support for the Tap-as-a-Service plugin[1].\n\n[1] https://opendev.org/openstack/tap-as-a-service\n\nChange-Id: I8b4dfd6cd02d61cb4ef07e117af39c44420e93a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/be83863a17c2b27af87f1ba409690d36c0ed23cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/be83863a17c2b27af87f1ba409690d36c0ed23cd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
