)]}'
{"id":"openstack%2Fpuppet-neutron~908793","triplet_id":"openstack%2Fpuppet-neutron~master~I6f136315f781d581e808906a85ef54740e8ba7a8","project":"openstack/puppet-neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-02-17 05:53:09.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I6f136315f781d581e808906a85ef54740e8ba7a8","subject":"ovn: Support options related to OVN\u0027s native DHCP","status":"MERGED","created":"2024-02-12 14:26:14.000000000","updated":"2024-02-17 08:20:10.000000000","submitted":"2024-02-17 08:19:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"908793","meta_rev_id":"3003662e5e9632ac25810b00aef6aad4b55ff93c","_number":908793,"virtual_id_number":908793,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-17 08:19:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2024-02-14 23:28:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2024-02-15 20:20:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2024-02-15 20:20:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-12 15:56:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-14 23:28:46.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2024-02-15 20:20:41.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"5a74c55a2d4ba5336cc2379f3d998afa2b20d831","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-12 14:26:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"229b9c9860e98bc884dfdfe4422099630a9f33f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 15:56:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/05bf8e0f0e2b47dfbaa63a66a40af9f5\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6eef434cb654b5d81a5c63a5c8ef0f0 : SUCCESS in 2m 54s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/48f22339ad444b6db6f9da2cef389043 : SUCCESS in 3m 37s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/6d86cca818eb4450b83e44f8deed6c76 : SUCCESS in 6m 01s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/da9cc3ef3465460190a4c87ebd676d14 : FAILURE in 16m 47s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65fa5fafae11482ab44b1b1581387962 : FAILURE in 12m 18s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/7759f79f94694e35bac47fdce1a652af : FAILURE in 16m 49s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d3d8c072337c4e48af6670aba9505dc1 : FAILURE in 11m 05s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/44bd5c65eba54db1b966ca35f6ec2c09 : SUCCESS in 54m 10s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/566bab049ed74082941cf3f4738c9718 : SUCCESS in 1h 05m 53s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/270c611808d74878ab98c4280f24b533 : SUCCESS in 53m 52s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc9219a67d5a4a7abd0a9ca4ef72562b : SUCCESS in 50m 36s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/694e697c61ac40ab933958eff715d821 : SUCCESS in 43m 31s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/47dbd21da217400aa474416c721abc25 : SUCCESS in 1h 06m 08s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/36c0eb1f1be947d6bfbb31fa317f8627 : SUCCESS in 1h 18m 30s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c8e2bb71687247ff9b816c8a3d62465d : SUCCESS in 59m 36s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5cff0b6548fd493684aabd4b032cd16b : SUCCESS in 57m 16s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bdc575e6abc14dbca0cd862397cf4e5e : SUCCESS in 1h 01m 33s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/7a6dc3aa2df3409cb78167195b83268e : SUCCESS in 19m 30s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6d551288ade14f95b85e9abdce83a0eb : SUCCESS in 16m 15s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/338bee357c9248ad820eabcf8e1a6825 : SUCCESS in 15m 45s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/afea4d3b449e49a7b9d56e6fd6dbba70 : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":1},{"id":"e9225d4096bc23efb10a1c942544ab51e05a85ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-14 04:07:29.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d4e4a2e4c327555cd8b2d54ecc6a6ce99b606dd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 05:28:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb645912ad85498d85b4df11cd07260c\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/995c6793cb8e4372b020675412f5f662 : SUCCESS in 2m 34s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b98b5e7f6b89488cbb662dd3033e1926 : SUCCESS in 2m 28s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/2cd906be1e24497fb60bed1e189474a1 : SUCCESS in 5m 31s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/18980febf6ef44a5a1ecaab4abf0e153 : FAILURE in 12m 15s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b99686e8112343c18ebf7f512a49411c : FAILURE in 12m 24s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/3ddfaf99852b4622a8784cd00f6595c3 : FAILURE in 16m 32s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/308590e769d941ca8a3eda078d389820 : FAILURE in 10m 11s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ecae5f67c754769b5fa3eef3226e780 : SUCCESS in 47m 59s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/51b2c6a59a7c4dc3a4f63652b94389b5 : SUCCESS in 49m 21s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b912b6d0c304881b2ce90e066931563 : SUCCESS in 53m 51s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1541c28c2d2f46038046cf48efa7fd28 : SUCCESS in 54m 51s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d20db3a77bd14d9ba64409825bfcea00 : FAILURE in 44m 17s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2cf1539433fa40fda4df43b7a457bd78 : SUCCESS in 57m 21s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6a034e5ff7284e0bb60b4a4c85b57fd0 : SUCCESS in 1h 13m 29s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c6e5e06dda094afe91a38e2dff00a456 : SUCCESS in 55m 50s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6717f5a24c8747f9b0620910b1b894c3 : SUCCESS in 1h 00m 02s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ca87aee2f664dba97edd77c67ce6e68 : SUCCESS in 1h 02m 18s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/0d484e94ee92471cafc29aeb8af4e56c : SUCCESS in 16m 31s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e6c224c56a314e70bb87ba7e4620f002 : SUCCESS in 13m 33s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab4b4bfd4d8145768ba8b18e85f53673 : SUCCESS in 12m 32s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e142e065f804e5d9888b61caf76edc4 : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":2},{"id":"69692386f38a7a83447a2869c43ba894419c9470","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-14 06:20:27.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"c6e6d95e8432c39970eba19045a73d899e1f995f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 07:44:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b73a93e58ff14b148368d3e45e35c960\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/08800433beef475ebd613bfec50b149a : SUCCESS in 3m 17s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/36f70c6de7344515ab2c2d61ba3f3f7e : SUCCESS in 3m 22s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/7140cc0538ce41d689331b6a3f899e00 : SUCCESS in 5m 04s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/b95a21977cf648b19650b05a353fd086 : SUCCESS in 14m 42s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93e64fd986254f098df24d41caa0efa1 : SUCCESS in 10m 06s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/71f96a3329e4481e91daddc730d9ee10 : SUCCESS in 15m 06s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05eff1fbc7ee448d8080d6c6d84927d9 : SUCCESS in 12m 30s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09d5aace3d0e4e13993ff26515400df6 : SUCCESS in 46m 04s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/778010d64c8f43faaf3469cf81c82e53 : SUCCESS in 1h 02m 21s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/caa2ecad68ed46b6863b63f71b83206d : SUCCESS in 48m 34s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4fc6d6c37d4942a488ca82977c8aeae8 : SUCCESS in 54m 28s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/74e98185fd9f44ebb3ba38d3b8a99a8f : FAILURE in 43m 24s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b197bec5f7af4daebf03357923446456 : SUCCESS in 1h 00m 48s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/317dac8d127d4781920017caf6e2c09d : SUCCESS in 1h 16m 53s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8f56a3c9a74542589181da48f199fe7a : SUCCESS in 51m 25s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c3cc99d81a254870983774325a59445a : SUCCESS in 53m 45s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2778c018f402434ca7a24c967d250831 : SUCCESS in 47m 33s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/c8e7422842984fb1a080538eda9bd4c0 : SUCCESS in 24m 44s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e23a60ede6ab46c7af98c7196b3f4436 : SUCCESS in 14m 02s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e79f4e5d0b0d4c788e7301358e49d2c6 : SUCCESS in 15m 08s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4aa3348c28f341ae9362230d11dba21e : SUCCESS in 2m 36s","accounts_in_message":[],"_revision_number":3},{"id":"d44c13617799a816d5f03fb8ca797bbae8716998","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-02-14 23:28:46.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"beeaba935e8f418098f0335fcbb6fd1bac40c2b7","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-02-15 20:20:41.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c9d391d768d0a9107877a876a4a0fbd7d432f1a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 20:21:31.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"151b1dba94b123d0d159854392de2f4d4026d2d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 21:30:50.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f7644101b3744ff69c9925559ad7e16e\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/684e4677b80a410988d6b7de2a9f1e0b : SUCCESS in 3m 14s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/37455180cb4948f78cf04c1c84d863da : SUCCESS in 2m 21s\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a29ea0d06d041aca976263912fb9761 : SUCCESS in 15m 23s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bedd5374e8734a9d987c9298a8345e39 : SUCCESS in 12m 25s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0ec80488cb4d44369589a8167f169f1b : FAILURE in 57m 36s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8e8bb798feee46819aba1435af923756 : FAILURE in 57m 54s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ec07b846f0714c8a96444e9dba73c07f : FAILURE in 1h 04m 59s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/921019b0e3ba4890ae51473a30ebeb97 : FAILURE in 56m 46s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2a4e3df890114fd1a90b36bb3c325863 : FAILURE in 52m 53s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/648957808b014e379f6ddb3f6045c1a3 : FAILURE in 32m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00df31b15b7840f596f903f3f3d1327b : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":3},{"id":"07e52f670c613e65f667613206b3b99ec14aa456","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-16 13:26:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"240e7a2535e66f75ddb3c3fd143272f6c59f8c3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 14:39:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fdc4f50296d4fa8a9012a77d895c28f\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78902f6943524b88b87c4d831bbea2ed : SUCCESS in 2m 50s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/370e029db65448f0a515bdfd06cd0241 : SUCCESS in 2m 35s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/09834c24371340daa076971e994605c5 : SUCCESS in 6m 45s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd995c142fb24966aea687a57ada79cc : SUCCESS in 16m 46s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55e61e05078e468c8e9fc68971c1eb2b : SUCCESS in 11m 10s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/9b63c064cb034ee3adaaa6fed6e3fc0a : SUCCESS in 16m 37s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/44116ccffd03442281c528c0f8909f90 : SUCCESS in 10m 16s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bac9125fceb445fc9404f600c44a9dda : SUCCESS in 49m 49s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38cbabe9d8ba4d628ffc81c4eeb25254 : SUCCESS in 58m 34s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/866f0252fb214dfbb2f3d14ed7a575b7 : FAILURE in 57m 58s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/89f38fcc1493468d83aea4afa3d957fb : SUCCESS in 45m 16s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59bca848679f4730b9775a147c9033c0 : SUCCESS in 44m 06s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a1bdea2b55de4785a1e3be8350172d8d : SUCCESS in 1h 01m 43s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e521899c934f4e4c810ec63b3ee4e572 : SUCCESS in 1h 05m 52s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3e0d11771ba84eb08193b0fb04de558b : SUCCESS in 1h 01m 12s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/de17cb255223471f9279b15988c7a34b : SUCCESS in 52m 55s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a461ba8d22ef4bce8a6dca709e645903 : SUCCESS in 57m 56s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/5d16e4c60dad454fab182723c1201d37 : SUCCESS in 24m 15s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b0429d527b2d46f3ab28704010c3eaa8 : SUCCESS in 14m 29s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/75825e0bf4d549b6b3e75de3bb87e810 : SUCCESS in 15m 17s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e114855683714768ba6973bd9296f153 : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":3},{"id":"ac38fcc81d6ab60322c52e4c3e85fdee9a9b3990","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 14:40:06.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9901f2ac41ad345179c8a7b471e6fd484df11b1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 16:02:49.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c74e6b14197f41faaf470f2d27f05657\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/22145691c2be40bb827bf51fef44b82f : SUCCESS in 2m 21s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b0c1a9511d1348afa07c9854d108b82d : SUCCESS in 2m 21s\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/d115022108854af3a02dff943d0de302 : SUCCESS in 13m 06s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2eeaa62e4fe41cd91fda822cf86197c : SUCCESS in 12m 29s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2d228cbb6c424b159b2d8c70a42f669c : SUCCESS in 54m 31s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/907c7f8fd9914c6289fcc636c83f7103 : SUCCESS in 1h 19m 18s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fd058ffd020641a3ba750e134110d561 : SUCCESS in 1h 01m 39s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/665f15db8f6846ca9ddd68081afedf14 : SUCCESS in 1h 03m 16s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7f89a1e1decc45f6b07dadd27aaa83f6 : FAILURE in 1h 08m 46s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/5dc875d54834433dae74089805b6b1eb : SUCCESS in 18m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e6f065f322e4e41bab25eee9ffda915 : SUCCESS in 2m 24s","accounts_in_message":[],"_revision_number":3},{"id":"78433c6750ce383219e976064057da72e1e200b5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-16 22:05:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"eb58f7fdc7fa6692aecdc12583e5ce2745750abc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 23:09:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcbe0b1afa0049f79fa68bc5c4ac5ad0\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eddbfd4470f9424bb504794930a170b8 : SUCCESS in 2m 45s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/873ec0be8857498b8875ab07947a1fca : SUCCESS in 2m 40s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/76edc3a0a1c34ffab6300abf085c0be5 : SUCCESS in 5m 32s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/0eb3be4c16cb4f29ad88e3671e5149e8 : SUCCESS in 18m 11s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52da4449733b4a76a656632d1fcb8f7a : SUCCESS in 12m 53s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/2a648a802f5e43479916b3da707dfc6c : SUCCESS in 17m 14s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe37278f17c04219bc989b72c4f4739e : SUCCESS in 9m 53s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/15831fe95c5d4279b62e51eb80b53817 : SUCCESS in 45m 03s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/48fbffdbb73f48b286c2b4955d809fe3 : SUCCESS in 52m 02s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a093d982fa94056aa098c890736fd9e : SUCCESS in 54m 27s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d05b2aa427344ef88ad3c95fb6d34749 : FAILURE in 50m 05s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d42e04abdaf40be832a2113d0e4146a : SUCCESS in 34m 05s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bfd4a399fbed4463ad80c6fe9e343b5d : SUCCESS in 55m 17s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/baf3e6b526c84d5ba5c681c1d1384aa6 : SUCCESS in 56m 26s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d56a00bab5254d3bac1c6407a49483e6 : SUCCESS in 49m 30s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0c6c1627fc064b2aaa37b54972f91821 : SUCCESS in 58m 32s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1be22fad39004cbf9dc29cb72c48ec8c : SUCCESS in 54m 01s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/97d9b7ee15b149d69e1e887f3cb46672 : SUCCESS in 18m 47s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29cbf642d7004736b90ac06dd7489387 : SUCCESS in 14m 28s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dcbf6e601ee14dfca9489468c2b2d62b : SUCCESS in 15m 10s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a6b50d4bbfa4c69b81537a8805da175 : SUCCESS in 2m 27s","accounts_in_message":[],"_revision_number":3},{"id":"d42e0d079bbb90e4c847388454283609152ca3cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 23:10:23.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3d46554a3815205830aa534ca0a0a9029180b9d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-17 00:35:16.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/08d4e213c3e043fead2b4100be18e9b8\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f5d6fac2bad543bda03a09c86222a1d9 : SUCCESS in 2m 25s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6228ddf471414b8280070a52ea68e5ff : SUCCESS in 2m 40s\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/62c2846e43a3441e928c87673b5dfe26 : SUCCESS in 14m 42s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/50e058794e244745aa28cd19e4fb3e9c : SUCCESS in 11m 18s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b3b1a3e99c024dd38f8fbc0353445bc0 : SUCCESS in 1h 02m 40s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9edf206d60094cdd82e0cc9330ff748e : SUCCESS in 1h 17m 28s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6cf129dccbbd4427917f5503fd6b1180 : SUCCESS in 1h 19m 41s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/02f3ab91484849599f5fef8061693e64 : FAILURE in 1h 10m 42s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dfae8cc0890a44aeb8dc32e7da210246 : SUCCESS in 1h 00m 53s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/cfb63bd67ea3402695b15ddcfbf19b8f : SUCCESS in 19m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8482acd1af424babb1f99ed0862850c5 : SUCCESS in 2m 20s","accounts_in_message":[],"_revision_number":3},{"id":"09828f7aab71a10f742d7b258ea3114718260db1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-17 02:30:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"977a6aedf2c13bfe1b9f8a726a6d50d1aca4be54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-17 04:17:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f8d7b3b7fe14d7d8e594fda8cf7a058\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c0101591a2f24101a16e892ef064daca : SUCCESS in 3m 05s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a973470bc35f45428a6caee90a270afa : SUCCESS in 4m 03s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/faf8b4086ce54be8bd066589885fb438 : SUCCESS in 8m 50s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/fa9a15e331064c258a6daf5485826f1e : SUCCESS in 16m 39s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab0b1774faf54af98d3a8853991a027e : SUCCESS in 11m 55s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/7c8a3be7868a40f3adbffff9411b0185 : SUCCESS in 14m 39s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c2b9a3634264023b94be63dc1260e04 : SUCCESS in 13m 58s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bef5c0c2e3a04422a62fed56bb3195f0 : SUCCESS in 54m 42s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/58e2d87afd7c435f8de341a6b3f61728 : SUCCESS in 1h 06m 35s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e38aeeb586e94fea80bd2ded6b74fea0 : SUCCESS in 54m 48s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1dcb950852bb4b2e950e3bcef27021c2 : SUCCESS in 46m 30s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24db79f3a7cc479bbe149152cbf2e662 : FAILURE in 49m 29s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd8be9a86d6c4ef1932defb461dccede : SUCCESS in 1h 06m 52s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/20fced4f70324e7c9f5d6a77ef329cf5 : SUCCESS in 1h 22m 23s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/37c16a5769cf45899efd447b8e5df26c : SUCCESS in 1h 02m 13s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/215835fe529e478f8d282aec0ddbc3b4 : SUCCESS in 1h 00m 23s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d0e403f38f214b0780500b27a786b4a7 : SUCCESS in 54m 02s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/bb499701f4734677aa329f9c14e76488 : SUCCESS in 22m 56s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4667fd64364f4063982aafdd6ab8f03e : SUCCESS in 13m 40s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30417e003868444eb5ff63ff4402c5a8 : SUCCESS in 13m 32s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7fe3c33b16c4570bbeeb421d56171de : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":3},{"id":"80aa05ad310eaee77cf7bf25923f150726ec99df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-17 04:18:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"57ac116319379d76906366f7618b3c16d52f98e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-17 05:41:25.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d9d8f18d0bc4f0888a9e78ce5f98234\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a652c6a06f747d38ce9dbfff5061dfa : SUCCESS in 3m 14s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/28a22e9e13fc4c95830bcf8bf4e86670 : SUCCESS in 2m 14s\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/77aaa9081dd743689abc9c58f6935fdc : SUCCESS in 13m 35s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00278ccb0a5e42368e82c7b8efb76c2b : SUCCESS in 12m 23s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/836baad37eea41b89b63845006d9d056 : SUCCESS in 45m 28s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4616929015804c9f98c798d1940c2778 : SUCCESS in 1h 09m 54s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fb545c4fbcc04bde911d81427645e67b : SUCCESS in 49m 50s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/af2c33c75d6f490d9cbf1fed96930d7d : SUCCESS in 52m 31s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0f2b412a417942c3bca98fde1d89d0f0 : FAILURE in 59m 33s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/0107eeb645d94c2d9a1e99de7c1ee8a5 : SUCCESS in 22m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08d1e4ddf5454a0f9f20a4b537ed4c85 : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":3},{"id":"14b6b3aae493f7c62259f4bcd721fc317a98a594","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-17 05:53:09.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8337c5b821dead379bc20d9045ded42ed917c9c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-17 07:13:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4cf7aeb01ed4d4b981da2d12925d3fb\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/671e32a023f94b92b369f1e955ec7c6c : SUCCESS in 2m 52s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/42c1bd9929f845728e0f142ba21c62a8 : SUCCESS in 2m 16s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/042d3886d2384343b127b38705ea54ba : SUCCESS in 5m 43s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/9c16d3f9cf184da6bbaec4b3ef917cfe : SUCCESS in 10m 42s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b34b9e3579ac4ca4a7a4e5dd8f37f807 : SUCCESS in 11m 43s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/3c82ad8c5dd944f782370fa2f589e0f8 : SUCCESS in 15m 48s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d8981496d58d44a58ad72fbd2127f3f6 : SUCCESS in 12m 10s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c51edd64dcbf42f2a824e9381bcbf62c : SUCCESS in 47m 29s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4d5c9b70c7c4af6b177681bf4eee297 : SUCCESS in 50m 21s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69e05b95b0ae4ed580fd73473dc46c2b : SUCCESS in 45m 31s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3edb0ff715944a62a8f43ffc81333f9e : SUCCESS in 35m 21s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/36257b7c21ff4a88b18609c60d5b2081 : SUCCESS in 37m 21s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6075463abe3b4bdb994b4a4a07d280b3 : SUCCESS in 46m 01s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/25a9656a34c34b01acb8c696a50995f1 : SUCCESS in 56m 00s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8ea521f2a95141d1bc01b1476d3e3c96 : SUCCESS in 1h 14m 12s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d1d6456d1f43410a94b284693b6a37af : SUCCESS in 48m 03s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/883f35e86a984ed4b8be1d6af041ea6e : SUCCESS in 44m 20s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/8ef93efd2d5745829677b156ab7dc0a4 : SUCCESS in 15m 46s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b668c5a978a243bdbb169edbc3ba1618 : SUCCESS in 13m 19s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99e3bb5470584effa49b1073d1788ffa : SUCCESS in 13m 45s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98dccf01ab9947ad83b32b2849195e01 : SUCCESS in 2m 22s","accounts_in_message":[],"_revision_number":3},{"id":"bcf38483faf16d468854bbfba9d3fdf68ecbb98f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-17 07:14:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"59cc224533ce31f7c8431a2ae37142f75a741ec8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-17 08:19:14.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/466e43f9bcab46bca2a39aa0ed2e893a\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b6bc58adee543adb06c3f1d7f671f19 : SUCCESS in 2m 14s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c84fb5301884e4f9d8b9ec15f8bc008 : SUCCESS in 2m 12s\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/7df7d86d6fb5416ba4ffa493c5db0a7d : SUCCESS in 11m 56s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0829881be66743b2a33cdfdbe07f2187 : SUCCESS in 8m 47s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0dfbad8b12654387ae4d0d71d0acbb33 : SUCCESS in 59m 08s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d6671f155395439bb93153c303be6c92 : SUCCESS in 1h 04m 26s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5995721877eb45579bdc5908da5ce06b : SUCCESS in 50m 08s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f355a5066ed54b22b963d8115533107a : SUCCESS in 45m 01s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/05928ca2713248669c2528bd63d83fde : SUCCESS in 55m 12s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc547172eeaa43b8be09770d07f97951 : SUCCESS in 19m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9424d1e88174b09806ac14b613a2ff9 : SUCCESS in 2m 29s","accounts_in_message":[],"_revision_number":3},{"id":"99c080dbbab57d7811b6a790d6f6825283520a5e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-17 08:19:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"3003662e5e9632ac25810b00aef6aad4b55ff93c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-17 08:20:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85d5817b5bf34ea8832fa0db175ebaf6\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff1497f34d3f4b4281ea567d57189c93 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4c8188efc04bb10e6f5d933b22a498cddbb5f539","revisions":{"b3c5ee9a8660a243135559d31e71edf54ad428e5":{"kind":"REWORK","_number":1,"created":"2024-02-12 14:26:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/93/908793/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/93/908793/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/93/908793/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/93/908793/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/93/908793/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/93/908793/1"}}},"commit":{"parents":[{"commit":"68cceacff2d4c2ca7d945ef82c1b96021c8280c0","subject":"Merge \"Provide config class interface for rootwrap.conf\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/68cceacff2d4c2ca7d945ef82c1b96021c8280c0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-12 14:25:45.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-12 14:25:45.000000000","tz":540},"subject":"ovn: Support options related to OVN\u0027s native DHCP","message":"ovn: Support options related to OVN\u0027s native DHCP\n\nChange-Id: I6f136315f781d581e808906a85ef54740e8ba7a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b3c5ee9a8660a243135559d31e71edf54ad428e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b3c5ee9a8660a243135559d31e71edf54ad428e5"}]},"branch":"refs/heads/master"},"e2a64b008267615df85e7b7efe3ee7b657c31081":{"kind":"REWORK","_number":2,"created":"2024-02-14 04:07:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/93/908793/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/93/908793/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/93/908793/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/93/908793/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/93/908793/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/93/908793/2"}}},"commit":{"parents":[{"commit":"68cceacff2d4c2ca7d945ef82c1b96021c8280c0","subject":"Merge \"Provide config class interface for rootwrap.conf\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/68cceacff2d4c2ca7d945ef82c1b96021c8280c0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-12 14:25:45.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-14 04:07:29.000000000","tz":0},"subject":"ovn: Support options related to OVN\u0027s native DHCP","message":"ovn: Support options related to OVN\u0027s native DHCP\n\nChange-Id: I6f136315f781d581e808906a85ef54740e8ba7a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/e2a64b008267615df85e7b7efe3ee7b657c31081"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/e2a64b008267615df85e7b7efe3ee7b657c31081"}]},"branch":"refs/heads/master"},"4c8188efc04bb10e6f5d933b22a498cddbb5f539":{"kind":"REWORK","_number":3,"created":"2024-02-14 06:20:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/93/908793/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/93/908793/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/93/908793/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/93/908793/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/93/908793/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/93/908793/3"}}},"commit":{"parents":[{"commit":"68cceacff2d4c2ca7d945ef82c1b96021c8280c0","subject":"Merge \"Provide config class interface for rootwrap.conf\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/68cceacff2d4c2ca7d945ef82c1b96021c8280c0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-12 14:25:45.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-14 06:20:27.000000000","tz":0},"subject":"ovn: Support options related to OVN\u0027s native DHCP","message":"ovn: Support options related to OVN\u0027s native DHCP\n\nChange-Id: I6f136315f781d581e808906a85ef54740e8ba7a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/4c8188efc04bb10e6f5d933b22a498cddbb5f539"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/4c8188efc04bb10e6f5d933b22a498cddbb5f539"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
