)]}'
{"id":"openstack%2Fpuppet-neutron~910112","triplet_id":"openstack%2Fpuppet-neutron~master~I71c1a83e150e3ca0ec6cb17d0b8a4440383ee477","project":"openstack/puppet-neutron","branch":"master","topic":"hc-ignore_proxied_requests","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-02-26 01:40:46.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I71c1a83e150e3ca0ec6cb17d0b8a4440383ee477","subject":"healthcheck: Expose ignore_proxied_requests parameter","status":"MERGED","created":"2024-02-25 16:27:52.000000000","updated":"2024-02-26 03:10:10.000000000","submitted":"2024-02-26 03:08:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"910112-hc-ignore_proxied_requests","meta_rev_id":"d3791aa512498d4db10d91daa3bf50adc3f30298","_number":910112,"virtual_id_number":910112,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-26 03:08:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-02-26 01:40:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2024-02-25 20:16:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-02-26 01:40:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-25 17:47:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-25 20:16:09.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"4965c956af02566917ff97fe5b39f9491f1d7fcd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-25 16:27:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c20877dceb786fd8498dafb16b414b1cb6111cf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-25 17:47:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3d49ea784e844a0bc4b59e6a6603fc5\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6d96d484ae44fb0a67487da74adc94d : SUCCESS in 2m 32s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e1761a94917a40598081ddf2af571180 : SUCCESS in 2m 24s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/a13c3abccac447dcba9ee500d6f7679b : SUCCESS in 5m 11s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/adc045c0c1c84c5985f680b323cd3c20 : SUCCESS in 14m 06s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b64868e216b042358c4cf10a6f266b8c : SUCCESS in 11m 24s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/66e4e21b71c8476b835c5bf433da3159 : SUCCESS in 16m 34s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0d98a9b701b84dea924f0ecc47d06297 : SUCCESS in 11m 06s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/01a551c0332b4acab6a7e001265a3dd0 : SUCCESS in 50m 52s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3223553ef31e4251a66b313ccc1308a5 : SUCCESS in 59m 18s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c3508555b71d45ada0cf626cecbcce4d : SUCCESS in 45m 13s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/64509ad8cc204c19855ef2c71de2740e : SUCCESS in 38m 31s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4613b4e675dc4005a675e324d18c981a : SUCCESS in 41m 13s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/50840ef1b1134036bb0139595623de34 : SUCCESS in 57m 56s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6cc28aecf0274182a37bb0c770799548 : SUCCESS in 1h 14m 35s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4eb41cd2008e4099a75578a665d32300 : SUCCESS in 52m 44s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d11daf92f6f4491d9365c7e11fde3ef1 : SUCCESS in 50m 40s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/58a5b7ac33de4865a17ea32770c88d4e : SUCCESS in 55m 33s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/71723b9173da43dd94fce98a291ea145 : SUCCESS in 19m 07s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99b870c941e6494a96cc0b332c207715 : SUCCESS in 14m 16s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e95d8223a3a4bfa8f67cdb6b543dfcd : SUCCESS in 14m 13s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/420150b652fd40279111180a53fa471f : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":1},{"id":"336eea11f71869968eb3bb015e2c045973da0456","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-02-25 20:16:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"872ba4f87fd156bce4a9e0635e2deb221797cda5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-26 01:40:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"77523e0d214d54b2f0f90f3cad2cef6679666cc1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 01:41:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"95cdcc0d693b592d25db89a9e4dfae0ee985e05c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 03:08:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/962f7cf33575424c8a1556b69730e6fb\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69f4b3163d8c4162945ca8fd391451de : SUCCESS in 2m 28s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/001f3c1f60624d50adbde3f7f6598497 : SUCCESS in 2m 32s\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/0d66f515d2604c3fa388b3dd7c4de091 : SUCCESS in 15m 08s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a9d03231640495abe57b7fd2af96470 : SUCCESS in 9m 12s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9bf8c9da09fa4c08a354e43a3bcfeadc : SUCCESS in 54m 37s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d22fa4b0c4034fc8a4840ad0c243a4a2 : SUCCESS in 1h 22m 23s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5b0ffdebb1de4f908b161917dc53ada2 : SUCCESS in 52m 14s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4d0ca41edc194ee08f915936134667be : SUCCESS in 1h 02m 52s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd452aff6e81424db4a745e3da682b59 : SUCCESS in 58m 09s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/f7c4fd6df30445448cca280ae4e6ca73 : SUCCESS in 22m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f098ec01d2e4204a4079831907e25f9 : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":1},{"id":"510b47b0ed1fffbdab3412add90007976ae887ee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 03:08:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d3791aa512498d4db10d91daa3bf50adc3f30298","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 03:10:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0925dc304dc4c3a881179cf38b753ad\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14f24063c75c4a68b06690a8499ea3d8 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2a5c79cf6b53336120f007759d65c784d4939c87","revisions":{"2a5c79cf6b53336120f007759d65c784d4939c87":{"kind":"REWORK","_number":1,"created":"2024-02-25 16:27:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/12/910112/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/12/910112/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/12/910112/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/12/910112/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/12/910112/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/12/910112/1"}}},"commit":{"parents":[{"commit":"68cceacff2d4c2ca7d945ef82c1b96021c8280c0","subject":"Merge \"Provide config class interface for rootwrap.conf\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/68cceacff2d4c2ca7d945ef82c1b96021c8280c0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-25 16:27:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-25 16:27:40.000000000","tz":540},"subject":"healthcheck: Expose ignore_proxied_requests parameter","message":"healthcheck: Expose ignore_proxied_requests parameter\n\nDepends-on: https://review.opendev.org/909807\nChange-Id: I71c1a83e150e3ca0ec6cb17d0b8a4440383ee477\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/2a5c79cf6b53336120f007759d65c784d4939c87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/2a5c79cf6b53336120f007759d65c784d4939c87"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
