)]}'
{"id":"openstack%2Fpuppet-neutron~910684","triplet_id":"openstack%2Fpuppet-neutron~master~Ib722982e06a15396d09d9ff380abe4f24b0280da","project":"openstack/puppet-neutron","branch":"master","topic":"memcache-sasl","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-03-05 04:05:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib722982e06a15396d09d9ff380abe4f24b0280da","subject":"cache: Support options for SASL mechanism in memcached","status":"MERGED","created":"2024-03-01 01:39:49.000000000","updated":"2024-03-05 04:06:52.000000000","submitted":"2024-03-05 04:05:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"910684-memcache-sasl","meta_rev_id":"3a4d1c4ef37c00b5b41c969f4bf600a7f35e3613","_number":910684,"virtual_id_number":910684,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-05 04:05:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2024-03-01 15:10:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2024-03-04 10:32:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-03-05 01:57:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2024-03-04 10:32:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-01 03:18:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-01 15:10:17.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2024-03-04 10:32:28.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-03-05 01:57:13.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"7d15ce6bd8b8ea110c793131061906063950453a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-01 01:39:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d6e3e80dcbd4eeb3a0137dd85289a6b11e0d81a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-01 01:41:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9a6688872090998520fd1c8eca061ffb33982252","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 03:18:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff52b057dc6541dfa76a0e3a0f4f4f29\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9ab08e4236a4f6bba22c38743d8356c : SUCCESS in 2m 18s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b968ca1dcb5446eaf677b1b2ddaf7af : SUCCESS in 2m 21s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/2652d570801b499fa778a5b4ebd3be36 : SUCCESS in 4m 29s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/b7fdfec01e5a40dba7ce6101db19f8eb : SUCCESS in 16m 10s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fb88198b0e15464682382523d00dd6d4 : SUCCESS in 12m 54s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/de2e80871a0c4a2f96d37008eca0dfb3 : SUCCESS in 16m 32s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47e9655e16204b46af618d09d347e693 : SUCCESS in 13m 02s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05a6ba9a5ddf4d8a9a0add6b09822385 : FAILURE in 7m 15s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/64f70c728bb947b98b4edb7ab2186098 : FAILURE in 5m 46s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a1c78c13eff94b61899c99cee6b03f95 : FAILURE in 6m 39s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a78391355b34df4b41d36c5ec5066bd : FAILURE in 6m 06s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2a6b53b21f8a4b3dbbc05a5203465d33 : FAILURE in 6m 30s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d44412b95ccd437b9bf62b629ef70dcf : SUCCESS in 55m 14s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/038dad549525449c9f1dddb70bc8a946 : SUCCESS in 1h 30m 57s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d02d4362162242eeb24f85b4742c8cca : SUCCESS in 1h 09m 07s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b3c401cbda8240ddb19f59c78f58d42b : SUCCESS in 1h 04m 12s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/983700ff5cd54980b29163f9b2af9027 : SUCCESS in 1h 08m 44s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/e8ec3b6945084f6d937ab59a3566b826 : SUCCESS in 19m 18s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/387ab49202c94450951f911ff0462cb0 : SUCCESS in 11m 43s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f0fd705fdc7d43238438ac60bbec2a60 : FAILURE in 6m 07s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a234136846c642799ba5b88d42450499 : SUCCESS in 2m 29s","accounts_in_message":[],"_revision_number":2},{"id":"e9ad8814a9e6e3ba595c052624e18160b9abca5c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-03-01 15:10:17.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"69e1b42e218d00cd5244299e1a21692d096444c5","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-03-04 10:32:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"18683ffdf5ca270436056cf5a8e08c3e0a2763d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 10:33:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1056782a861526489ffab75e95b3eb29072d6628","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 11:51:12.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eee7101bef3c4a5d982022ea9b66820f\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66e9808a6fde4d418f6d9006b944f353 : SUCCESS in 3m 27s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/23e12c0615734e598f60a965379b0f69 : SUCCESS in 2m 42s\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/4de592574e4748bd94c3eb702c19258a : SUCCESS in 13m 28s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4fe52eacc6454c5aaa41c2540ce8a0a0 : SUCCESS in 8m 54s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1cd23a888bcb45758037b2fe459ad01e : SUCCESS in 1h 02m 32s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/214b3669ca0e4a82b8d9076e6c8fff9a : SUCCESS in 1h 00m 13s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6a474c258ebd4bed9c2538795986752b : SUCCESS in 44m 57s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3e34ad6fba284d7286c7d1cf7c12dc11 : SUCCESS in 57m 17s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8bf23f33ecfa40199b33202355f59240 : FAILURE in 54m 42s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/5742c105d8334c00a02177c53289f710 : SUCCESS in 17m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37814ebc40724c5fb6d9ede4110b9467 : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":2},{"id":"e0e1c6d23c24a459b082518882527e5780296fb6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-04 13:25:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b48d90b895e89c6590ac18bae1591d638bcb1770","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 14:54:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e8e7ad924c24daf80305f6422e454fe\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6676982e4b75452c87838d6bb0690770 : SUCCESS in 2m 27s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/502b0085795446ec88a55ee4491bd355 : SUCCESS in 2m 52s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/611bfc6835414f1c9671c211ae2ad019 : SUCCESS in 5m 42s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/5f9026c23fff456a97e07b30e44f3073 : SUCCESS in 18m 13s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/698c7f9c51674204a98a1b7452117ca0 : SUCCESS in 12m 41s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/d39eb96f68a142a183577241981dd8bc : SUCCESS in 14m 09s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a34d8bf4aba34ec1bee56627b9f5156f : SUCCESS in 12m 55s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3bb906ed4b64cf18c0b1746a53ad8e5 : FAILURE in 7m 11s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/228d7e70502b4366ac0ad5f79af68d87 : FAILURE in 6m 46s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/468cba71fb5847bd9c6c4da51266b58f : FAILURE in 7m 28s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c54a585a2bfb435a81be317757e2454c : FAILURE in 8m 07s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba16a916bccf4243bda9af3c3ecb8210 : FAILURE in 5m 14s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6f4a8ece453c45439ca2ecc76e8971a9 : SUCCESS in 1h 01m 32s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0e9fe051ce3f4e70b2016a3007600f69 : SUCCESS in 1h 22m 49s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/03e02b50e15d45d0a1797ff0660709c9 : SUCCESS in 1h 05m 44s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0df02526c06a48f78e6e340fd7c2c21e : SUCCESS in 52m 50s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b7b48b6345e348d7adf4e1a4060565d1 : FAILURE in 1h 15m 35s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/4eada9fdc1b14498afe0f9e5f06bba3f : SUCCESS in 22m 12s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/149a1623265f4916aa24ed8c785f9abd : SUCCESS in 11m 51s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2cb38e8d32a1493d8765060f1205e230 : FAILURE in 6m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e408cf65e50b4147b7c8c42d3f8b537b : SUCCESS in 2m 30s","accounts_in_message":[],"_revision_number":2},{"id":"ccca737304c00dfd8900e04bde70db6e0e16fb54","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-05 01:00:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"41b0c0cc6e6bc880b32571ca88a939cb59fc299e","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2024-03-05 01:57:13.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"43f6799a0f0e088a7cc85ce5b6027553cc8c6d05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 02:37:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ba533e6d9d4406580fb6ab2858ac38f\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/27d4f9f1d45a47c6892de12749a0bb05 : SUCCESS in 3m 24s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/90e8736b6f22442e82bbf8030d8e63c7 : SUCCESS in 2m 49s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/7dd2b37901e44a15a37432cce858afcc : SUCCESS in 5m 28s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/33f7001cc3154f20ba5e47f9118c57e1 : SUCCESS in 15m 23s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/62ad58166b5344f6ab9377519f50a95f : SUCCESS in 11m 29s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/1fa121fdee2d44cca4c178d85d72be87 : SUCCESS in 12m 48s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/89d519a6b6c6442a958c4f9903e52105 : SUCCESS in 12m 40s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/781fa784ef474096afe3fa9ba33d56cd : FAILURE in 6m 26s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/faa972559bab4018a470d75aef71e28c : FAILURE in 5m 17s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ea256969488450285bd27d7d3877db0 : FAILURE in 6m 24s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7ff2d7cc0bd42a098862604e84d993a : FAILURE in 6m 42s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9e3e8211d594a4e82ec40ee676fb666 : FAILURE in 7m 10s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2beced6b2b354992b6c3291cb3c63484 : SUCCESS in 1h 01m 36s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6319785160474524a78e5d57f9886350 : SUCCESS in 1h 27m 25s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0b25d3812c9c4e21ae5cbd4a9805be9c : SUCCESS in 1h 04m 36s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a8f615a32c104b5ba11ba76f029f16ea : SUCCESS in 52m 52s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/93847ae0ef80442c90c7c891209c9b4f : SUCCESS in 50m 17s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/c057df54f33146dc8bacac918d9b1718 : SUCCESS in 15m 54s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ccb703ef32b5474297d11078bd361a08 : SUCCESS in 12m 25s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c3799fc1fe73452c829705275ab1fa2d : FAILURE in 5m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/664d490008a547e28f8e90794e7c7afa : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":2},{"id":"0e601c919f22d87f3d7ec16d01132ad2089adbcf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 02:38:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a64a4e8394eed18aa16ba43266e6ffc6352d693e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 04:05:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56a2e44eb76c41d0927169f54814eef9\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00ad28edaf994519a9458e5df3ea49e3 : SUCCESS in 3m 33s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/800c62e3239b4d50850c43264aa21270 : SUCCESS in 4m 04s\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/7517f7656b004b4ca521fddce92ed2c8 : SUCCESS in 18m 02s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/81be0115cc364af18d59e1b8b9208662 : SUCCESS in 14m 55s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c74a05daf3e4d788d178676ca2eb36c : SUCCESS in 1h 01m 18s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc0ee9a2507049a483e64351e3fdb80e : SUCCESS in 1h 05m 27s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c1624545ec354e60962ffb571b9c792d : SUCCESS in 1h 15m 04s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f7126cad47b540ccb0c87558c6bc5330 : SUCCESS in 1h 02m 19s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/adff6afe421a4164b575369da578447f : SUCCESS in 59m 15s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/2c31ea8324ef434fbb190c818d439249 : SUCCESS in 22m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6736a483a5f94f3092ef3b60ee35d057 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":2},{"id":"66e6a37061e8ef6fefe52d3f519ed05c6f7413ca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 04:05:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3a4d1c4ef37c00b5b41c969f4bf600a7f35e3613","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 04:06:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb69880b64e14640bad9174ef2751a83\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7bdaddbb5aab43589b9354d59951c412 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3832c80d64883212761ff32d22e9df2fc727ac96","revisions":{"baea65f8a0dc7cf0056e9e538f94fca8658adea0":{"kind":"REWORK","_number":1,"created":"2024-03-01 01:39:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/84/910684/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/84/910684/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/84/910684/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/84/910684/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/84/910684/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/84/910684/1"}}},"commit":{"parents":[{"commit":"2a5c79cf6b53336120f007759d65c784d4939c87","subject":"healthcheck: Expose ignore_proxied_requests parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/2a5c79cf6b53336120f007759d65c784d4939c87"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-01 01:39:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-01 01:39:17.000000000","tz":540},"subject":"cache: Support options for SASL mechanism in memcached","message":"cache: Support options for SASL mechanism in memcached\n\nDepends-on: https://review.opendev.org/910122\nChange-Id: Ib722982e06a15396d09d9ff380abe4f24b0280da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/baea65f8a0dc7cf0056e9e538f94fca8658adea0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/baea65f8a0dc7cf0056e9e538f94fca8658adea0"}]},"branch":"refs/heads/master"},"3832c80d64883212761ff32d22e9df2fc727ac96":{"kind":"REWORK","_number":2,"created":"2024-03-01 01:41:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/84/910684/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/84/910684/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/84/910684/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/84/910684/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/84/910684/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/84/910684/2"}}},"commit":{"parents":[{"commit":"2a5c79cf6b53336120f007759d65c784d4939c87","subject":"healthcheck: Expose ignore_proxied_requests parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/2a5c79cf6b53336120f007759d65c784d4939c87"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-01 01:39:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-01 01:41:10.000000000","tz":540},"subject":"cache: Support options for SASL mechanism in memcached","message":"cache: Support options for SASL mechanism in memcached\n\nDepends-on: https://review.opendev.org/910122\nChange-Id: Ib722982e06a15396d09d9ff380abe4f24b0280da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3832c80d64883212761ff32d22e9df2fc727ac96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3832c80d64883212761ff32d22e9df2fc727ac96"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
