)]}'
{"id":"openstack%2Fpuppet-neutron~916354","triplet_id":"openstack%2Fpuppet-neutron~master~I819fa33182729902e5f723dd0ca0db8a1be7d33a","project":"openstack/puppet-neutron","branch":"master","topic":"plugin-cleanup","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-04-25 15:18:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I819fa33182729902e5f723dd0ca0db8a1be7d33a","subject":"Remove support for contrail plugin","status":"MERGED","created":"2024-04-19 01:56:44.000000000","updated":"2024-04-25 15:19:35.000000000","submitted":"2024-04-25 15:18:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916354-plugin-cleanup","meta_rev_id":"5646508f90cc1b1ee0a69aff6205cd2bfa506914","_number":916354,"virtual_id_number":916354,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-25 15:18:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2024-04-19 06:30:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2024-04-25 13:56:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2024-04-25 13:56:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-19 02:53:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-19 06:30:11.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-04-25 13:56:58.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"7964e6fb1d4e2070e97135530925f86b7ca36061","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-19 01:56:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc956ded0d4cb6081c62b0ddf0136413eeb0fc2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 02:53:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/59f87268bca143fc827d23f824deb4bb\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dfb1d33c0bb84846ac3d546e8aa02f54 : SUCCESS in 2m 28s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f28af2e80d54cbaad845f4f5c79198b : SUCCESS in 5m 07s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/7e23db2288464fddbe7b5a572ba98795 : SUCCESS in 7m 07s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/4e5836306a8b4df28135acdc879eac4d : FAILURE in 14m 05s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29ac8735514f45d095a44ebf3b46d536 : FAILURE in 10m 03s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/a48ca5cd61f14c52b0b71c2f2522ea74 : FAILURE in 13m 32s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6fc6d1be36194541b4a2a30e7fb85e77 : FAILURE in 10m 36s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bfb6c908b7644e3ab2cfa8c790733787 : FAILURE in 5m 43s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/73e20cd556de4df292eded8ae647f9e1 : FAILURE in 6m 13s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e201c4a1b364404b6729f59773654f3 : FAILURE in 5m 59s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b0d90a0242a4e14a27060720a04bc77 : FAILURE in 6m 03s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ccb1b9952ac46589ca6ad4e4062c917 : FAILURE in 6m 00s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/efa97683277242e29a4b5794c2ab2b7f : FAILURE in 10m 56s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e22c0757b15d414d9ae3117a9400e888 : FAILURE in 13m 39s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/88952c7e5a7b482497c455aaf8075573 : FAILURE in 10m 22s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/51d8b62e1db74e00b2bed5b5cbdd99f4 : FAILURE in 13m 06s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/05aab25bf20643028544b74e2b9d1351 : FAILURE in 12m 38s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/ec51c7da22f24f2f8b5717361407b042 : FAILURE in 9m 26s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/35953e38c37f4487b997b17ae4eecbbb : FAILURE in 5m 42s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e989dccf36d64affb6dd1f9d3dc363a8 : FAILURE in 8m 28s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b359644a726541f3aeb6e8b1439f4d64 : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":1},{"id":"3cd485b1ee4f6e13053ec6fb7594803226e4da51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-19 03:39:26.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e2b8ec5f2eb01736034f0054c42445782009a091","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-19 04:59:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ab8b261ea39acb296fb7d5e772b887240cf8f086","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 06:14:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b0f6100dedb452fa23ebc04d8d0b263\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c4894b63ec2041358036a88e2688d0a7 : SUCCESS in 2m 51s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1605672cbd2d458893c0b2fc7c6a8708 : SUCCESS in 3m 07s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/3987b267366845348efea0fe87c9238c : SUCCESS in 5m 09s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/7149b8f8370844578800d3a9f22ae3ad : SUCCESS in 14m 44s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a1f2bf82909d42f28555353e5b4e8423 : SUCCESS in 11m 13s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/c3dbc240a0e44ba784dd452548dfaedf : SUCCESS in 12m 20s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b0f611aa15b428d859be75f1a29c4ad : SUCCESS in 10m 38s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7da19a4380a41f0bb56a3e8745bb4ac : SUCCESS in 44m 10s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da581a056f834600b9633bcd208ace65 : FAILURE in 56m 55s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ac5e3ff535644578108ec393a1067b1 : FAILURE in 29m 06s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/761de97b57654af889031b281e7f967c : SUCCESS in 42m 00s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c365fefcee1949098917cb76599ece18 : FAILURE in 47m 16s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3e4451cc83fb464cbdcf0d40e4f74a93 : SUCCESS in 54m 20s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3bbb205272c749498c48f1a5b1d13004 : SUCCESS in 1h 11m 52s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/14fbc83237614a86b8a90c5118be029c : SUCCESS in 1h 02m 33s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7b3c98e61ce54322be8e80b5262079de : SUCCESS in 59m 32s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/868a99356e32430f8f5540b972e406cd : SUCCESS in 1h 08m 23s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/8792c7140071490193b64a9dcec0ac73 : SUCCESS in 20m 02s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/077f6e80b2c340ccaa9d92453b977652 : SUCCESS in 15m 19s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b29264caa16b422c990309246eaecf96 : SUCCESS in 13m 22s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/331708c72636479192eaa6e4e27ad60d : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":3},{"id":"4ad63997cc3ee7c4f64311c3774688f48adb1189","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-04-19 06:30:11.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"63a1ccc4dc920f20e4a4e49f74fa67772541026f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-04-25 13:56:58.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"78fbd8abc6cecf5d85d081c11b6a754d76ed18f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 13:57:52.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e823b0a3bb33f277401dcdf6887b73035453ff93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 15:18:24.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cba9c242b9c475ab577c39ae90d2edd\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/92d71afd16d74a45910b4d176cd19dc2 : SUCCESS in 3m 16s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/053da439de824bc48a871733c9a9cf11 : SUCCESS in 3m 13s\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/917a8478d5464972b638da4f1dc6da0b : SUCCESS in 17m 01s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/070e80bb7fbd4c91a5f6749db7c6f122 : SUCCESS in 12m 51s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b32d781dc7444233b4ecdec60b43e685 : SUCCESS in 57m 36s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/49dc1cdab7a44820aa482f35d4466cb4 : SUCCESS in 1h 12m 45s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/02f0b33b29894b7fae025c3157d3a0ee : SUCCESS in 53m 55s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/79ddee196d144285bb685055fe99f115 : SUCCESS in 59m 13s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7ef763145dcc4954873e8052c0d62c2b : SUCCESS in 53m 55s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/95bf51f2707242c699a8fdd2e817cc61 : SUCCESS in 18m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ed4713062c54ee3b858ed589091d302 : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":3},{"id":"6c52e15cbdcba47f32125441d65090891bf51ed4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 15:18:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"5646508f90cc1b1ee0a69aff6205cd2bfa506914","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 15:19:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bcc1d16a32643999afea67ed18b9def\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d943afe3d10a4f7d810a364023902bc1 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c6c6fb87a588810aa63682ccf85a7d8fea52c5ee","revisions":{"041fa66b7cd66ca07a8c959a9edc1b6bc03c9c62":{"kind":"REWORK","_number":1,"created":"2024-04-19 01:56:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/54/916354/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/54/916354/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/54/916354/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/54/916354/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/54/916354/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/54/916354/1"}}},"commit":{"parents":[{"commit":"12a2b5fd6f21c6f39d7c11d705ab64c790ab96ed","subject":"Remove support for Cisco VTS mechanism driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/12a2b5fd6f21c6f39d7c11d705ab64c790ab96ed"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 01:55:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 01:55:05.000000000","tz":540},"subject":"Remove support for contrail plugin","message":"Remove support for contrail plugin\n\n... which was deprecated during the previous cycle[1].\n\n[1] 0877c62e924250ee0b9088d7eaf4c1d38892b8ac\n\nChange-Id: I819fa33182729902e5f723dd0ca0db8a1be7d33a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/041fa66b7cd66ca07a8c959a9edc1b6bc03c9c62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/041fa66b7cd66ca07a8c959a9edc1b6bc03c9c62"}]},"branch":"refs/heads/master"},"7d10d87b7a382ab567628494c51f4884ffd035e8":{"kind":"REWORK","_number":2,"created":"2024-04-19 03:39:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/54/916354/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/54/916354/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/54/916354/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/54/916354/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/54/916354/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/54/916354/2"}}},"commit":{"parents":[{"commit":"12a2b5fd6f21c6f39d7c11d705ab64c790ab96ed","subject":"Remove support for Cisco VTS mechanism driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/12a2b5fd6f21c6f39d7c11d705ab64c790ab96ed"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 01:55:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 03:36:55.000000000","tz":540},"subject":"Remove support for contrail plugin","message":"Remove support for contrail plugin\n\n... which was deprecated during the previous cycle[1].\n\n[1] 0877c62e924250ee0b9088d7eaf4c1d38892b8ac\n\nChange-Id: I819fa33182729902e5f723dd0ca0db8a1be7d33a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/7d10d87b7a382ab567628494c51f4884ffd035e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/7d10d87b7a382ab567628494c51f4884ffd035e8"}]},"branch":"refs/heads/master"},"c6c6fb87a588810aa63682ccf85a7d8fea52c5ee":{"kind":"REWORK","_number":3,"created":"2024-04-19 04:59:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/54/916354/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/54/916354/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/54/916354/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/54/916354/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/54/916354/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/54/916354/3"}}},"commit":{"parents":[{"commit":"12a2b5fd6f21c6f39d7c11d705ab64c790ab96ed","subject":"Remove support for Cisco VTS mechanism driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/12a2b5fd6f21c6f39d7c11d705ab64c790ab96ed"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 01:55:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 04:58:57.000000000","tz":540},"subject":"Remove support for contrail plugin","message":"Remove support for contrail plugin\n\n... which was deprecated during the previous cycle[1].\n\n[1] 0877c62e924250ee0b9088d7eaf4c1d38892b8ac\n\nChange-Id: I819fa33182729902e5f723dd0ca0db8a1be7d33a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c6c6fb87a588810aa63682ccf85a7d8fea52c5ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c6c6fb87a588810aa63682ccf85a7d8fea52c5ee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
