)]}'
{"id":"openstack%2Fpuppet-neutron~916355","triplet_id":"openstack%2Fpuppet-neutron~master~I0271bf41035858b154782f20d0cbc9f5df537d51","project":"openstack/puppet-neutron","branch":"master","topic":"plugin-cleanup","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-04-25 15:18:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0271bf41035858b154782f20d0cbc9f5df537d51","subject":"config: Remove deprecated plugin_nuage_config","status":"MERGED","created":"2024-04-19 01:56:44.000000000","updated":"2024-04-25 15:18:26.000000000","submitted":"2024-04-25 15:18:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916355-plugin-cleanup","meta_rev_id":"f241e558b93e8eb180ec930cbe3550db6799244f","_number":916355,"virtual_id_number":916355,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-25 15:18:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2024-04-19 06:30:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2024-04-25 13:57:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2024-04-25 13:57:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-19 02:49:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-19 06:30:24.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-04-25 13:57:02.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"2ad105e6c60af184312e942ca4522c782d22bb71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-19 01:56:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d03c2eb92edbfa86ce3ba271020d1fce593acb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 02:49:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/da0811f56d8744fbaa2f17ef922b27fe\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ebbc7cc1032d47098e0ef05997291826 : SUCCESS in 2m 29s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/576f0623e90b49a09bb76ac3ec16ab9b : SUCCESS in 3m 19s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/0ae4436f1fc144c38d0fac72aed465cc : SUCCESS in 5m 44s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc1e4f87403e45ac8b96341d8abcd737 : FAILURE in 14m 54s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9512b7299de4965ab5355e24a12f16b : FAILURE in 10m 00s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/581e2ea0f6174732853e32f24ff030c8 : FAILURE in 15m 49s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/40e126cc6c5d430096df5389d838228a : FAILURE in 10m 23s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ffdd047987f7420f9728176f78d12203 : FAILURE in 7m 49s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e3812c4dfc424f40861eb019ab10b63d : FAILURE in 5m 42s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e9d4f2f5afb4d78a721c4baf6a74c5a : FAILURE in 6m 24s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0743a2ff28ff4db2bb82ea3335d8893c : FAILURE in 7m 57s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c730e30327b24c96aced7815ccb3ddc5 : FAILURE in 7m 50s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/497e277986b24f9aa44d2b5002d7e4b3 : FAILURE in 11m 50s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6dc175e0450440c29987d83d51e15153 : FAILURE in 14m 02s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fc1e6710409f45a5a20de19712929634 : FAILURE in 9m 57s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f8d4b62bd57f4d83a55f107c67cca6fa : FAILURE in 11m 56s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/878d8c826e144690a40c87445129e639 : FAILURE in 12m 03s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/6cd70af5747d4346bffb22b27497ef1c : FAILURE in 10m 34s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca8a00e65c4742efaa3de3e81413223b : FAILURE in 6m 10s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/093a3347b0ae4c6e9592c2e8475c690e : FAILURE in 5m 15s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/895afb4da5424ac6af0f60160033f4cf : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":1},{"id":"f783ce5a9bc12589356304a36f6ea31786e77ad7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-19 03:39:26.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"797547801e124e71e4690cc50e032e25d4977ace","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-19 04:59:11.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"be49eba8ef7ed7134e764cbf3163b512ebb8d899","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 06:08:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/777084afa447459090bf595be13a4754\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ba9f32c577f47019480079164e5c8ed : SUCCESS in 3m 19s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/96a2c8709ed04fa4966044bfaf1ab8f1 : SUCCESS in 3m 25s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/07522f5105534d90bb756dd43ac23df5 : SUCCESS in 5m 14s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/0fbf099b15e74f009c99c1c7086cb009 : SUCCESS in 13m 09s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3eed4936e13546eb9ead00eea933bc5b : SUCCESS in 11m 10s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/d030c9e8abf14fa4a9de8fd73f0b1fab : SUCCESS in 15m 41s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b56f6f8483954acba3ad37919f6afa69 : SUCCESS in 9m 23s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f9981a3ad9c405e931fd55027703aaf : SUCCESS in 43m 43s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30dd4b7ec62244cfab0bb941f66bcb9e : FAILURE in 52m 35s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cc6900bf66ad47d0a3f8d2037ba50581 : FAILURE in 28m 23s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/44d6216dae7942b6bc1e1d7a94ccd435 : SUCCESS in 46m 30s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b4c9d296c0a4de7af82cc394f8aad83 : SUCCESS in 43m 30s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/143754b22d564394b5bf5756350a203f : SUCCESS in 57m 47s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d3aecf8eb3654d88a430a6d130d99938 : SUCCESS in 1h 08m 05s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c23c3b5419b0410193f0c60fc4379cb4 : SUCCESS in 54m 47s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d9986412584e4a6ca7d5eac6e3fb9711 : SUCCESS in 1h 03m 32s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7be9278c6af441f790572d8dc205847e : SUCCESS in 56m 42s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/f63ca6fbff034135b0e52dc71ae6109f : SUCCESS in 19m 17s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae54e71611924e17a9e81b5265205e16 : SUCCESS in 12m 43s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f77ef5184224ce0b4e2753edbbea7c3 : SUCCESS in 14m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c03e5b6170144ddc8c034ecc390eca74 : SUCCESS in 2m 25s","accounts_in_message":[],"_revision_number":3},{"id":"527f8e2dd3ffe303e1c09765c979cdf5225e60fa","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-04-19 06:30:24.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"5c67a8c999b2c0c8a9338b0d4cb34ad38f8b34ed","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-04-25 13:57:02.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6a7488f546960ec03260d45c5d720e50fc409633","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 13:57:57.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"15da21bff7a3a1fbc9ba9616f79e008370ce5843","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 15:18:26.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/878a99719c32480497c94ac61b3e224a\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e1cb05d88c94ccbb7b87af9348fd024 : SUCCESS in 2m 50s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/073606e2b17645deb44a2728d283e482 : SUCCESS in 2m 49s\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/51a8c6287e3e4d99ad5d210df0e082a0 : SUCCESS in 11m 32s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc3dc660e29b47a6981ae982c8e3b25b : SUCCESS in 9m 45s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e8e08f6b5f0443419ee69a4e6cc0312e : SUCCESS in 52m 17s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/397b153b9dff4f21bb9e708fc20d4ac0 : SUCCESS in 1h 13m 02s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0bd914e2ac964c4b8f7122c96aa3fbf9 : SUCCESS in 52m 07s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5756ac8af6fc4baf99a6db2abb76ff0a : SUCCESS in 57m 14s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3baa0733d4754f269c609c20a5e888f1 : SUCCESS in 50m 13s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/9cd0f38b166042bc974d8623b0f3c462 : SUCCESS in 20m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a60b2b10b444d5b89537d2a8bbd9328 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":3},{"id":"f241e558b93e8eb180ec930cbe3550db6799244f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 15:18:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"21474f02baed251b5dd821c3a510f1898298f3cb","revisions":{"9d3d7c6fffc69b6d0f1b7e99da85bd6377ff59f1":{"kind":"REWORK","_number":1,"created":"2024-04-19 01:56:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/916355/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/55/916355/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/55/916355/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/55/916355/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/55/916355/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/55/916355/1"}}},"commit":{"parents":[{"commit":"041fa66b7cd66ca07a8c959a9edc1b6bc03c9c62","subject":"Remove support for contrail plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/041fa66b7cd66ca07a8c959a9edc1b6bc03c9c62"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 01:56:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 01:56:03.000000000","tz":540},"subject":"config: Remove deprecated plugin_nuage_config","message":"config: Remove deprecated plugin_nuage_config\n\nThis was deprecated and has had no effect. stable/2024.1 was released\nso now we can remove the parameter.\n\nChange-Id: I0271bf41035858b154782f20d0cbc9f5df537d51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/9d3d7c6fffc69b6d0f1b7e99da85bd6377ff59f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/9d3d7c6fffc69b6d0f1b7e99da85bd6377ff59f1"}]},"branch":"refs/heads/master"},"3cf39ef5d44db9873a4fd060b2c407787ecfa4dc":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-04-19 03:39:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/916355/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/55/916355/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/55/916355/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/55/916355/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/55/916355/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/55/916355/2"}}},"commit":{"parents":[{"commit":"7d10d87b7a382ab567628494c51f4884ffd035e8","subject":"Remove support for contrail plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/7d10d87b7a382ab567628494c51f4884ffd035e8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 01:56:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 03:36:58.000000000","tz":540},"subject":"config: Remove deprecated plugin_nuage_config","message":"config: Remove deprecated plugin_nuage_config\n\nThis was deprecated and has had no effect. stable/2024.1 was released\nso now we can remove the parameter.\n\nChange-Id: I0271bf41035858b154782f20d0cbc9f5df537d51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3cf39ef5d44db9873a4fd060b2c407787ecfa4dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3cf39ef5d44db9873a4fd060b2c407787ecfa4dc"}]},"branch":"refs/heads/master"},"21474f02baed251b5dd821c3a510f1898298f3cb":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-04-19 04:59:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/916355/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/55/916355/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/55/916355/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/55/916355/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/55/916355/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/55/916355/3"}}},"commit":{"parents":[{"commit":"c6c6fb87a588810aa63682ccf85a7d8fea52c5ee","subject":"Remove support for contrail plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c6c6fb87a588810aa63682ccf85a7d8fea52c5ee"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 01:56:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 04:59:01.000000000","tz":540},"subject":"config: Remove deprecated plugin_nuage_config","message":"config: Remove deprecated plugin_nuage_config\n\nThis was deprecated and has had no effect. stable/2024.1 was released\nso now we can remove the parameter.\n\nChange-Id: I0271bf41035858b154782f20d0cbc9f5df537d51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/21474f02baed251b5dd821c3a510f1898298f3cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/21474f02baed251b5dd821c3a510f1898298f3cb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
