)]}'
{"id":"openstack%2Fpuppet-neutron~930105","triplet_id":"openstack%2Fpuppet-neutron~master~I31f0b2bab7d1d7ba58a79662598ee18f756c0600","project":"openstack/puppet-neutron","branch":"master","hashtags":[],"change_id":"I31f0b2bab7d1d7ba58a79662598ee18f756c0600","subject":"Revert \"Use consistent path pattern for lock_path\"","status":"MERGED","created":"2024-09-21 05:54:20.000000000","updated":"2024-09-21 08:23:42.000000000","submitted":"2024-09-21 08:23:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930105","meta_rev_id":"769e9defc1434ad77a829b780af610a828afb4f4","_number":930105,"virtual_id_number":930105,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-21 08:23:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-09-21 05:54:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-09-21 05:54:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-21 06:59:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"76fa0a7c13177d354d8622d0805874e4376a8961","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-21 05:54:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b7846c2d863e61eac5798271d395d7a5c843f29","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-21 05:54:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"64b4038695f25c5d83e36e8143e5eca65825d646","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-21 06:59:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a19d8466769649b398d7a74113d5705f\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1fd9ab0fdd3f42e587edd60f2ed3a4bb : SUCCESS in 1m 58s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f110d5ad06c344fa87c10933ee6d1df1 : SUCCESS in 2m 37s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/b4a9b1c1fef74e58b83409a5e5d336fb : SUCCESS in 4m 42s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/6a91323b859c4b46bf75d097254d0483 : SUCCESS in 9m 41s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b39c518044c4151a4c72b9b104d8735 : SUCCESS in 5m 31s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/380402686a434b84ad5e6575c6619317 : SUCCESS in 9m 51s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9611c1116e7d4f3a85b1460642367624 : SUCCESS in 6m 38s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3078f141fcee450580b22b30fce25bce : SUCCESS in 28m 20s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4399135e5ec4442fb0d1da2cfe261e04 : SUCCESS in 46m 29s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/645dcc010b0b47aa80387ab5a58e6804 : SUCCESS in 25m 48s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c64902bc0924ba29c7635f4582469b3 : SUCCESS in 38m 25s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67f31522fb734b90a3778f01e1a2410b : SUCCESS in 32m 57s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5021d553b6a74fc6abd1fde8c872d814 : SUCCESS in 59m 17s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/73dc2124e6214821ab38d6da79122ef9 : SUCCESS in 46m 09s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cf74704cb1b44a21a9d97ddd5933293b : SUCCESS in 47m 21s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6d041fc1458749c8b5c69761cac08eba : SUCCESS in 32m 47s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a8f04558d10c476fbd216ec5fb670860 : SUCCESS in 52m 48s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/bbf0a01b6d074c9fb55d6396289e5c29 : SUCCESS in 15m 03s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24feb5855625497d8a8e4bda00628d41 : SUCCESS in 7m 53s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f5134575748442ba9c3571241c3bb095 : SUCCESS in 14m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"10fb7a16710c5e6a84f705aa76ccce724dcfe698","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-21 07:00:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7a2130f576e99d9228484b496da9e90f8e266c2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-21 08:23:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14b583ace18e43b0bb636d91760dfd29\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3034c07c2acf47fbaae76efc8e781d58 : SUCCESS in 3m 19s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7cd814a82f3643d0971150c10d0e517d : SUCCESS in 2m 45s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/604c8e6099ae4bc5a5f0c91976ca9ee4 : SUCCESS in 9m 00s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d4bba9bad37a450999351f5cdd68f95d : SUCCESS in 6m 50s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/43778bd85c9c4a38b2905eeb5b456605 : SUCCESS in 1h 03m 30s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2758f39c7b7640289d3c5331e70c70a8 : SUCCESS in 1h 18m 47s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/91252433f3424d3bb63ee7738293db57 : SUCCESS in 57m 09s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/69abe0cb97f2482aad4bf0513bae5ed6 : SUCCESS in 58m 31s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fe37cb277b2e46fa8c0c2a08ce6c834c : SUCCESS in 59m 41s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/f128e7a3d409413284f52d6a9496bee9 : SUCCESS in 20m 19s","accounts_in_message":[],"_revision_number":1},{"id":"769e9defc1434ad77a829b780af610a828afb4f4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-21 08:23:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f3e88e70916f8e017e8be0e63f678fb1bd2d3d25","revisions":{"f3e88e70916f8e017e8be0e63f678fb1bd2d3d25":{"kind":"REWORK","_number":1,"created":"2024-09-21 05:54:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/930105/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/05/930105/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/05/930105/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/05/930105/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/05/930105/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/05/930105/1"}}},"commit":{"parents":[{"commit":"3a974aa69100e37171043bb41a34d2c7dbbe95b1","subject":"Merge \"taas: Remove OVN NB DB options from ovn agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3a974aa69100e37171043bb41a34d2c7dbbe95b1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-21 05:45:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-21 05:54:10.000000000","tz":540},"subject":"Revert \"Use consistent path pattern for lock_path\"","message":"Revert \"Use consistent path pattern for lock_path\"\n\nThis reverts commit 1ad35d2790f17f293463131d137f963475733613.\n\nReason for revert:\n$state_path/lock is the path used in neutron installed using RDO\npackages by default. Also usage of /var/lock/neutron breaks ml2-ovn,\nbecause neutron can\u0027t access to the directory and get Permission\nError.\n\nChange-Id: I31f0b2bab7d1d7ba58a79662598ee18f756c0600\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/f3e88e70916f8e017e8be0e63f678fb1bd2d3d25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/f3e88e70916f8e017e8be0e63f678fb1bd2d3d25"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
