)]}'
{"id":"openstack%2Fpuppet-neutron~940969","triplet_id":"openstack%2Fpuppet-neutron~master~I202fbe275232d43eaad8939faa4d560f343e82f9","project":"openstack/puppet-neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-02-12 21:24:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I202fbe275232d43eaad8939faa4d560f343e82f9","subject":"Add support for [ovn] live_migration_activation_strategy","status":"MERGED","created":"2025-02-07 13:41:03.000000000","updated":"2025-02-12 21:25:31.000000000","submitted":"2025-02-12 21:24:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"940969","meta_rev_id":"f63169641636a55ddbf8e689a4a95ef981a2a100","_number":940969,"virtual_id_number":940969,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-12 21:24:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2025-02-12 18:00:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2025-02-12 19:49:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2025-02-12 19:49:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-07 15:23:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-12 18:00:22.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2025-02-12 19:49:44.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"9290a432808ffd0e36864280e30dae6836d1724d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-07 13:41:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df0933a1198ca9857fdc293dc8d753345191ddba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 15:23:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e5332d63b06543b1a68f606858e8f99f\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b455cba2d8e4fe0b87c89bda8d18195 : SUCCESS in 2m 30s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0d31566ec6644b3c8423dbff84b6b8be : SUCCESS in 2m 24s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/5c18a3b836064df396b408572d8bed98 : SUCCESS in 5m 42s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/4e63cdab084646f1acc87ce3d041dddc : SUCCESS in 12m 30s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f48c5023af0442aa59d59c622d30bfe : SUCCESS in 9m 49s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/c6ca0e93a7d142bf9600ca66b999d859 : SUCCESS in 12m 31s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b30066a97ff04d528885123d4b3bfe78 : SUCCESS in 8m 45s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7062bdd348d4dd2947dc2037547c4c0 : SUCCESS in 48m 17s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f32553095f9a47a487d6c7623e05024a : SUCCESS in 57m 57s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97704db9786e44d3ab7aa4b04daec28a : FAILURE in 17m 03s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dde21a7e8dff4e64b90786eff56405bf : SUCCESS in 40m 52s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/864fd6a848914191bf082baaec754cef : SUCCESS in 24m 17s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/16a9525418f4435890e3e2be147d1f59 : SUCCESS in 1h 10m 58s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c6cc8971e29946aaaa14be6c8364423f : FAILURE in 1h 14m 30s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e5683d8672f245dcaf6b128f50a32afb : FAILURE in 1h 36m 33s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/60e896b4ddaa4d0caf4132c97eb7f667 : SUCCESS in 1h 00m 11s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/918b9829ba5544a791429bd7477f2349 : SUCCESS in 1h 22m 08s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/570592549a3343c1b8a6a59e8d1cec2a : SUCCESS in 27m 36s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/92625001a2cd46ffa328bb82081f9792 : SUCCESS in 16m 03s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ead09bc3da8f4d92b467ae63f3b0311c : SUCCESS in 11m 59s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26d4d9bf6bb94f16ad78855dccc6f564 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":1},{"id":"936a47339062021c2f7116b84186533c2d48be6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-12 16:39:44.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"5d3c0e8bad4595f8f720bc9777259f645f6aa4c2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-02-12 18:00:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ba7e5c95eafa4a8b86a1e788e2b7778a4eb74b3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 18:24:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13b95b1043784100a58d20a673b985e0\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7d044866f3b4d229c10e1010d33db85 : SUCCESS in 2m 36s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/874ce0a1baf545239a8f1fbb7c5f18d4 : SUCCESS in 2m 26s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/77a88eb46753441e9f447e504557ab33 : SUCCESS in 5m 21s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/042eb28dd52e4256b73fd6a74f0b88b1 : SUCCESS in 12m 24s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6c0b3fc758c4b21a1656d09c4362830 : SUCCESS in 10m 22s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/ae1e80025a6d4393907565d63772b785 : SUCCESS in 14m 19s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a853a3ad9f4647749ce23e0e531b81f0 : SUCCESS in 8m 50s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b28b4a68b3954720a351f8395e35181e : SUCCESS in 46m 29s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e615cead1cd84660a64dfe843cef494a : SUCCESS in 59m 13s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ea6a246a70d40419c1bcf4947592b1e : FAILURE in 29m 04s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c546f7825ca4023b979fcdb1e4169eb : SUCCESS in 40m 34s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3a71203938d423ead2b04d5383a74a8 : SUCCESS in 41m 11s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/529668bd9fb54574afb97eef1b6ea3f3 : SUCCESS in 1h 16m 37s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5fbe7d8f996449d5b8e74e78d5361a9e : SUCCESS in 1h 30m 35s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b4c33cdb0d7b4634bea2e4c770013b41 : SUCCESS in 1h 16m 30s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b82f605f0e7e488fb351107ea34853bd : SUCCESS in 1h 13m 24s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/818d5cb4b8f34ae994e2ada87ed70ac4 : SUCCESS in 1h 05m 35s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/eb9bb4776fbc465eb0b555e6642f70f6 : SUCCESS in 1h 26m 59s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/362e9801a5c841f6b7c921cb3a6c4aba : SUCCESS in 1h 35m 50s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f0a196a878344481976280f07de679d3 : SUCCESS in 53m 42s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6f4c549cf6c249dda3c99c7b4530a3a4 : SUCCESS in 1h 06m 00s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cbb865aadf314d6c9c08e8d87cfe099a : SUCCESS in 52m 01s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/54283c78ad8f4de8947a3af1a2a5ae81 : SUCCESS in 36m 38s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/249526dcdfa941a4a03a673db8d35530 : SUCCESS in 12m 28s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/123f2ade640e47778dde8195d5a2d406 : TIMED_OUT in 1h 31m 02s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f35ad46a26b43f1baa29f6e5bca16b4 : SUCCESS in 14m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ad794c95d99451b91ffb20d05ce9db4 : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":2},{"id":"44c583439b0fd268fc43314eefee2c10ca2b58c6","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-02-12 19:49:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bd6b3597208056432bee453471410fc72d027ccd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 19:50:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"652394075970307d20666fa178d3ca72c8de7f9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 21:24:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f191d771937f4fcebbcb0319ca44d7ff\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b735be5e946148a1ab08108f5d6c68ae : SUCCESS in 2m 35s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ff35b2ad7ab48d2a76de9f60983cd36 : SUCCESS in 2m 32s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/57692f3f999043a185086c8356bd29c0 : SUCCESS in 7m 22s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3caafc8b105d4d8dad38fc56d33b393c : SUCCESS in 9m 23s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/9790af53fea04c4f8e06049793c00389 : SUCCESS in 7m 45s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c464ae2b80224f899600bc6595fab454 : SUCCESS in 1h 20m 09s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/322e590d87904da2be9ad81c1b77cb63 : SUCCESS in 53m 18s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/164b29a47f3a427d887fbce47b5f36dd : SUCCESS in 1h 04m 25s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0007382b39094de4aef8bad361bf4c9f : SUCCESS in 1h 10m 37s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/11ced65b8a7b449a8015a792f34b3279 : SUCCESS in 42m 32s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/59cbd32d283d4f1ba8979297dc13e1a9 : SUCCESS in 29m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c887d7a6091f475e84b554983052af72 : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":2},{"id":"15d55e66f648ada75afaa92580dab4316cfb38ae","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 21:24:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f63169641636a55ddbf8e689a4a95ef981a2a100","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 21:25:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da25cc8bc0fa43d7a87c91cc754e5614\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8885385c3ddd44da9ab628d931785679 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7d299d93bc71412a4443e8c26769e2306a361e89","revisions":{"2edaee00df6a83dd0206a98b45dbd95c9cae38b0":{"kind":"REWORK","_number":1,"created":"2025-02-07 13:41:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/69/940969/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/69/940969/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/69/940969/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/69/940969/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/69/940969/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/69/940969/1"}}},"commit":{"parents":[{"commit":"14b6e6197999e8c092a771ca4875821fface60fb","subject":"Merge \"Fix outdated README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/14b6e6197999e8c092a771ca4875821fface60fb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-07 13:40:23.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-07 13:40:52.000000000","tz":540},"subject":"Add support for [ovn] live_migration_activation_strategy","message":"Add support for [ovn] live_migration_activation_strategy\n\nDepends-on: https://review.opendev.org/938106\nChange-Id: I202fbe275232d43eaad8939faa4d560f343e82f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/2edaee00df6a83dd0206a98b45dbd95c9cae38b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/2edaee00df6a83dd0206a98b45dbd95c9cae38b0"}]},"branch":"refs/heads/master"},"7d299d93bc71412a4443e8c26769e2306a361e89":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-02-12 16:39:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/69/940969/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/69/940969/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/69/940969/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/69/940969/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/69/940969/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/69/940969/2"}}},"commit":{"parents":[{"commit":"2657b151777d51acd6616da48d3e47a4a342a0a9","subject":"Merge \"Fix list of mechanism drivers in parameter description\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/2657b151777d51acd6616da48d3e47a4a342a0a9"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-07 13:40:23.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-12 16:39:44.000000000","tz":0},"subject":"Add support for [ovn] live_migration_activation_strategy","message":"Add support for [ovn] live_migration_activation_strategy\n\nDepends-on: https://review.opendev.org/938106\nChange-Id: I202fbe275232d43eaad8939faa4d560f343e82f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/7d299d93bc71412a4443e8c26769e2306a361e89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/7d299d93bc71412a4443e8c26769e2306a361e89"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
